)]}'
{"id":"openstack%2Fpuppet-swift~826206","triplet_id":"openstack%2Fpuppet-swift~master~I1923a5aed12a503f450c7d2e4a7784e4746fb46c","project":"openstack/puppet-swift","branch":"master","topic":"system-scope-params","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-26 15:08:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1923a5aed12a503f450c7d2e4a7784e4746fb46c","subject":"Accept system scope credentials for Keystone API request","status":"MERGED","created":"2022-01-25 00:58:38.000000000","updated":"2022-01-26 15:10:24.000000000","submitted":"2022-01-26 15:08:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826206-system-scope-params","meta_rev_id":"0d29d88fd29427d3e769feb88fd88db3ef0315cd","_number":826206,"virtual_id_number":826206,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-26 15:08:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-01-25 16:03:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-01-26 13:41:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-01-26 13:41:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-25 03:30:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-25 16:03:08.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-01-26 13:41:09.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"749fd85c97271cb6dabcb9c412b2b5bfaa524c23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-25 00:58:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e50c5f1edba417b1e7e2579225ee788ad5724c56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-25 01:30:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"319c2601aadead4154fb3d58495c1b529190d16c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-25 01:53:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f2934af01456f16fc1d4d60c6857a09a80e07c82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 03:30:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4956c6232fb84ffab02fdcfd8bb7420a : FAILURE in 3m 02s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/058d7dfd08634945a32ee14d8e71e723 : FAILURE in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98ce9037f0db41da940586081b2a66ea : FAILURE in 3m 59s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae648d84574c41b59b9176ef5f82498f : FAILURE in 16m 14s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a1229e9618694524965a67804c67ffb0 : FAILURE in 14m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61fc4ade557d47a39753444af29eb61b : FAILURE in 12m 22s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f323b5eb1f740738257376cd207ed50 : FAILURE in 15m 23s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/374d7ad242f745e79420f704a7e7f5a8 : FAILURE in 11m 22s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3f9f3df873c40d68a3aadee9b5c9d77 : FAILURE in 36m 33s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a51bb16aa1b2480a999192ab5c8fa3d3 : SUCCESS in 1h 31m 31s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/11b5625e3f8e4edca74c7bda82568b14 : SUCCESS in 50m 29s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c50a0a5890248c499b318d556414ccd : FAILURE in 25m 46s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/68597e66af3f48019d0cbdc50b1de21e : SUCCESS in 25m 58s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b9bed3f6d044c24bd8724e88aab7394 : FAILURE in 25m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c85c6c11332346a4bc6934682ad73d37 : FAILURE in 20m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/621dfcb2e9ee49ab9eae2e03b6dd701c : SUCCESS in 4m 37s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/597edc453e6c438382f02c010dcda878 : FAILURE in 53m 46s","accounts_in_message":[],"_revision_number":3},{"id":"564a71e17f4e2469436c3dac89939fa92dcd5314","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-25 04:32:03.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"863a42b0fd3287d7c0bfbcf5b52b2996ee493f83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-25 04:34:06.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"595ccadeec003968463637d8fab3c2e904882d28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 05:55:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a974e61ae0b54278a6fba0c7240ec383 : SUCCESS in 2m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b80bda429b0d459d941d9fce50650ae3 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ab4b887d9944dba9f6f7c0c8412964a : SUCCESS in 3m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/242e2cd5ca494cbea5325053023b7261 : FAILURE in 12m 15s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec219230aee24cb392415eeea80e2ad5 : FAILURE in 12m 17s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/613ccb3b358645a6a6b7ee8beba2511a : FAILURE in 15m 07s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca79e0945e7e4af3a4334f8d4783ee29 : FAILURE in 10m 37s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0ae5b57297f47bab0dceed044f6a69b : FAILURE in 10m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ca17e561ac442feae91a875d7086c8a : FAILURE in 35m 06s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f9f8107625c542f6b28763528bc826c5 : FAILURE in 1h 07m 19s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0c451eca6cc4977852c0e15bee21585 : SUCCESS in 49m 44s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41301a88814043aebd3a61086fdc00d8 : FAILURE in 18m 30s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/abb107c60edd447db9993740ce2bb530 : SUCCESS in 28m 27s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87001506b24e4547830ba1126bd4b865 : FAILURE in 18m 45s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22bd30bf45af4e6bb22760b4384adae7 : FAILURE in 23m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cbcabd1ca924ca09413404714c4ab98 : SUCCESS in 3m 17s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1efe64e7e4b142768c40332ca6185bfa : SUCCESS in 1h 18m 43s","accounts_in_message":[],"_revision_number":5},{"id":"530d721fe7563875a8206dd022af067ddda6c42b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-25 06:20:50.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"571aed3ebfe3dd11e9aedcb36bb354da6a869112","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 07:46:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/519aa5d2cfa749b4a455e90443940fff : SUCCESS in 4m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56d316d9032f4dcf8c2c3926e8d66f70 : SUCCESS in 3m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6671609058d4f46bf4d6f9b1ce51828 : SUCCESS in 3m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/687cbad578b44a77a6ef6cbea1205b45 : SUCCESS in 15m 41s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/cca13f84916e43caa87e945b75b15524 : SUCCESS in 16m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d1e3008a84549febd4e589563fa9c70 : SUCCESS in 15m 21s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d28556fa0ce442e88d4e90c335f023f : SUCCESS in 14m 06s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca1f7a8f13254da4b62c4a09aa14cb84 : SUCCESS in 15m 42s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2a407588d06445aa05fefe5cac5b474 : FAILURE in 33m 17s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/48d5547a5d894ea183c3f8da172fdf1e : SUCCESS in 1h 23m 07s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8fdc26b1e35a4bb29ecc502883e225a6 : SUCCESS in 51m 06s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2d32c04c7694f9a94021087c122cb95 : FAILURE in 24m 40s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c6f38e0b5e847fe8a39338a2a1b65cc : SUCCESS in 28m 43s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b45e2c0dc9b2493089a1fdd923eb583e : FAILURE in 26m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/122a554a50624db39bf5eed654dac469 : FAILURE in 20m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cefeace824e479e8d30d0c1ff3d09a2 : SUCCESS in 5m 08s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5ea560e5aab6467893224ca34bab62ee : SUCCESS in 1h 22m 17s","accounts_in_message":[],"_revision_number":6},{"id":"e45929032f49696eb85a611e410541340c2fd2ba","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-25 16:03:08.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"14ed1b4378f39821639709591934487f2173cc1e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-01-26 13:41:09.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"fb5fcf573411309a9bdb288ff78ba86e2f188ae0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 13:42:53.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1efa00d6317c374cb079302a83a8b43c35016501","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 15:08:13.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04e0aee69d22408cbef6ef5df75dd84b : SUCCESS in 4m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f32fe279a722479bbbb9648968ca5b2d : SUCCESS in 4m 25s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0985fe4ddc6c48ffac3e80d05b515d0b : SUCCESS in 3m 59s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/03a86ea401cb4dad93876f0c0d2f5985 : SUCCESS in 16m 10s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/31e361e18a1e4259b48e510800769d19 : SUCCESS in 16m 31s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05f3f534966543ffb93431b5cd7bec01 : SUCCESS in 14m 45s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93ecd97c33f24aaa918efa3686d8d287 : SUCCESS in 15m 58s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4d1c4d7150843b1bc07f2027e5f2743 : SUCCESS in 1h 15m 27s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/d58c367199624a59840a94bee0ab8098 : SUCCESS in 29m 25s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc1083ef66a24097ab3c2a232e705df6 : FAILURE in 25m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c52342571bb845689e86a4708e115ccd : SUCCESS in 5m 34s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0c952887e57e41bc962d1ecf9464dac7 : SUCCESS in 1h 24m 24s","accounts_in_message":[],"_revision_number":6},{"id":"001f4158fd08f463eb7edcb98dfcaba5e9c35275","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 15:08:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"0d29d88fd29427d3e769feb88fd88db3ef0315cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 15:10:24.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd18f74df975453984a1d0783024b4d4 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"2f2183e9d634c4d46834472ba221969ca4a88052","revisions":{"7c243c95daffaa4de7a1bafeea48b47db52c7ba8":{"kind":"REWORK","_number":1,"created":"2022-01-25 00:58:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/826206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/06/826206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/1"}}},"commit":{"parents":[{"commit":"95c974c27d6e7cc92c8372e238ec9a5c60a6e4e9","subject":"Remove redundant dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/95c974c27d6e7cc92c8372e238ec9a5c60a6e4e9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-25 00:58:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-25 00:58:19.000000000","tz":540},"subject":"Accept system scope credentials for Keystone API request","message":"Accept system scope credentials for Keystone API request\n\nThis change is the first step to support secure RBAC and allows usage\nof system scope credentials for Keystone API request.\n\nThis change covers the following three items.\n - assignment of system scope roles to system user\n - credential parameters for authtoken middleware\n - credential parameters for s3token middleware\n\nDepends-on: https://review.opendev.org/804325\nChange-Id: I1923a5aed12a503f450c7d2e4a7784e4746fb46c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7c243c95daffaa4de7a1bafeea48b47db52c7ba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7c243c95daffaa4de7a1bafeea48b47db52c7ba8"}]},"branch":"refs/heads/master"},"dbf9da8358647df88a9ff527bdc0c39c496bc594":{"kind":"REWORK","_number":2,"created":"2022-01-25 01:30:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/826206/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/06/826206/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/2"}}},"commit":{"parents":[{"commit":"95c974c27d6e7cc92c8372e238ec9a5c60a6e4e9","subject":"Remove redundant dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/95c974c27d6e7cc92c8372e238ec9a5c60a6e4e9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-25 00:58:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-25 01:30:08.000000000","tz":540},"subject":"Accept system scope credentials for Keystone API request","message":"Accept system scope credentials for Keystone API request\n\nThis change is the first step to support secure RBAC and allows usage\nof system scope credentials for Keystone API request.\n\nThis change covers the following three items.\n - assignment of system scope roles to system user\n - credential parameters for authtoken middleware\n - credential parameters for s3token middleware\n - credential parameters for ceilometer middleware\n\nDepends-on: https://review.opendev.org/804325\nChange-Id: I1923a5aed12a503f450c7d2e4a7784e4746fb46c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/dbf9da8358647df88a9ff527bdc0c39c496bc594"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/dbf9da8358647df88a9ff527bdc0c39c496bc594"}]},"branch":"refs/heads/master"},"0788efc85619f12cb7112fcfc0edd2e306011710":{"kind":"REWORK","_number":3,"created":"2022-01-25 01:53:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/826206/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/06/826206/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/3"}}},"commit":{"parents":[{"commit":"95c974c27d6e7cc92c8372e238ec9a5c60a6e4e9","subject":"Remove redundant dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/95c974c27d6e7cc92c8372e238ec9a5c60a6e4e9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-25 00:58:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-25 01:53:49.000000000","tz":540},"subject":"Accept system scope credentials for Keystone API request","message":"Accept system scope credentials for Keystone API request\n\nThis change is the first step to support secure RBAC and allows usage\nof system scope credentials for Keystone API request.\n\nThis change covers the following three items.\n - assignment of system scope roles to system user\n - credential parameters for authtoken middleware\n - credential parameters for s3token middleware\n - credential parameters for ceilometer middleware\n\nDepends-on: https://review.opendev.org/804325\nChange-Id: I1923a5aed12a503f450c7d2e4a7784e4746fb46c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0788efc85619f12cb7112fcfc0edd2e306011710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0788efc85619f12cb7112fcfc0edd2e306011710"}]},"branch":"refs/heads/master"},"68e88f9962f0a6f2be35f74b878535a795518761":{"kind":"REWORK","_number":4,"created":"2022-01-25 04:32:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/826206/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/06/826206/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/4"}}},"commit":{"parents":[{"commit":"95c974c27d6e7cc92c8372e238ec9a5c60a6e4e9","subject":"Remove redundant dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/95c974c27d6e7cc92c8372e238ec9a5c60a6e4e9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-25 00:58:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-25 04:32:02.000000000","tz":0},"subject":"Accept system scope credentials for Keystone API request","message":"Accept system scope credentials for Keystone API request\n\nThis change is the first step to support secure RBAC and allows usage\nof system scope credentials for Keystone API request.\n\nThis change covers the following three items.\n - assignment of system scope roles to system user\n - credential parameters for authtoken middleware\n - credential parameters for s3token middleware\n - credential parameters for ceilometer middleware\n\nDepends-on: https://review.opendev.org/804325\nChange-Id: I1923a5aed12a503f450c7d2e4a7784e4746fb46c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/68e88f9962f0a6f2be35f74b878535a795518761"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/68e88f9962f0a6f2be35f74b878535a795518761"}]},"branch":"refs/heads/master"},"fb24b49c6a83b5606abc739d0b8fcb6d80f49e31":{"kind":"REWORK","_number":5,"created":"2022-01-25 04:34:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/826206/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/06/826206/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/5"}}},"commit":{"parents":[{"commit":"95c974c27d6e7cc92c8372e238ec9a5c60a6e4e9","subject":"Remove redundant dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/95c974c27d6e7cc92c8372e238ec9a5c60a6e4e9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-25 00:58:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-25 04:33:58.000000000","tz":540},"subject":"Accept system scope credentials for Keystone API request","message":"Accept system scope credentials for Keystone API request\n\nThis change is the first step to support secure RBAC and allows usage\nof system scope credentials for Keystone API request.\n\nThis change covers the following three items.\n - assignment of system scope roles to system user\n - credential parameters for authtoken middleware\n - credential parameters for s3token middleware\n - credential parameters for ceilometer middleware\n\nDepends-on: https://review.opendev.org/804325\nChange-Id: I1923a5aed12a503f450c7d2e4a7784e4746fb46c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fb24b49c6a83b5606abc739d0b8fcb6d80f49e31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fb24b49c6a83b5606abc739d0b8fcb6d80f49e31"}]},"branch":"refs/heads/master"},"2f2183e9d634c4d46834472ba221969ca4a88052":{"kind":"REWORK","_number":6,"created":"2022-01-25 06:20:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/826206/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/06/826206/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/06/826206/6"}}},"commit":{"parents":[{"commit":"95c974c27d6e7cc92c8372e238ec9a5c60a6e4e9","subject":"Remove redundant dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/95c974c27d6e7cc92c8372e238ec9a5c60a6e4e9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-25 00:58:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-25 06:20:39.000000000","tz":540},"subject":"Accept system scope credentials for Keystone API request","message":"Accept system scope credentials for Keystone API request\n\nThis change is the first step to support secure RBAC and allows usage\nof system scope credentials for Keystone API request.\n\nThis change covers the following three items.\n - assignment of system scope roles to system user\n - credential parameters for authtoken middleware\n - credential parameters for s3token middleware\n - credential parameters for ceilometer middleware\n\nDepends-on: https://review.opendev.org/804325\nChange-Id: I1923a5aed12a503f450c7d2e4a7784e4746fb46c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2f2183e9d634c4d46834472ba221969ca4a88052"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2f2183e9d634c4d46834472ba221969ca4a88052"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
