)]}'
{"id":"openstack%2Fpuppet-swift~826693","triplet_id":"openstack%2Fpuppet-swift~master~I370d0c7b4056ecdf3686a23ea004708b1baa73e4","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2022-03-16 09:24:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I370d0c7b4056ecdf3686a23ea004708b1baa73e4","subject":"Increase default listen_queue_size to 256","status":"ABANDONED","created":"2022-01-27 14:59:57.000000000","updated":"2022-03-16 09:24:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1c693000aa7cff1198bac1e3170ad104df97aaab","_number":826693,"virtual_id_number":826693,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-27 16:25:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-27 16:25:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a761a52e6730662df6d6d0ba0624ff6d1187b826","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-01-27 14:59:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efa7e31f621442f29746f1dd67e9bb31ce319d98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 16:25:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f20b6b1501b4abf996b5ab919951afb : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22746edad78b42c5a696d074191b958d : SUCCESS in 3m 56s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e598718787424cdaa8e7561f30ce2673 : SUCCESS in 4m 03s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0815005eba544721a6fb886806b3e0e3 : FAILURE in 15m 07s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a81eb93b89542e5ab2047f52c362b99 : FAILURE in 17m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32ef84816c6b48d196e5bb787c5e2aa4 : FAILURE in 14m 21s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7fc560ff93145018a2fd093b3fe3476 : FAILURE in 14m 22s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f04d01bc2a04d2285fb30b39278e241 : FAILURE in 15m 16s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ccf9f47d49ed4e148655021f93e33dd2 : FAILURE in 38m 07s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e370013a5fee49a68f3ead0f4ba571a8 : SUCCESS in 1h 17m 28s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/85e5d7d029924acfb07a27bf8bad10a5 : FAILURE in 52m 34s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1373d09eff554b228d7652181814558b : FAILURE in 22m 46s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b6c31f349ab457ba12905753f0b83b0 : SUCCESS in 29m 22s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91436ef449cb443680787ea200f7247c : FAILURE in 23m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15443cfcec2243d88563cc0081a00c1d : FAILURE in 26m 25s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/71df6d9ab97346ec87f9664701f8d7e5 : SUCCESS in 1h 14m 46s","accounts_in_message":[],"_revision_number":1},{"id":"1c693000aa7cff1198bac1e3170ad104df97aaab","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-03-16 09:24:24.000000000","message":"Abandoned\n\nbad idea: uwsgi only supports 128 as max.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cfe7a5619786dbb0e63a0ff8f0c3796848a039a5","revisions":{"cfe7a5619786dbb0e63a0ff8f0c3796848a039a5":{"kind":"REWORK","_number":1,"created":"2022-01-27 14:59:57.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/93/826693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/93/826693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/93/826693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/93/826693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/93/826693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/93/826693/1"}}},"commit":{"parents":[{"commit":"f82d016350e6e972a11c6f3ad617f34040ca54de","subject":"Merge \"Accept system scope credentials for Keystone API request\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f82d016350e6e972a11c6f3ad617f34040ca54de"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-27 14:58:36.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-27 14:58:36.000000000","tz":60},"subject":"Increase default listen_queue_size to 256","message":"Increase default listen_queue_size to 256\n\nIn production, we noticed that a queue of 100 is too small, and\nsometimes can be a problem. Increasing to 256 (which is now the\nnew Debian package default) fixes the problem and removes the\nmessage we have in the uwsgi logs:\n\n\"*** uWSGI listen queue of socket \"\u003cIP\u003e:6001\" (fd: 3) full !!! (101/100) ***\"\n\nChange-Id: I370d0c7b4056ecdf3686a23ea004708b1baa73e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cfe7a5619786dbb0e63a0ff8f0c3796848a039a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cfe7a5619786dbb0e63a0ff8f0c3796848a039a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
