)]}'
{"id":"openstack%2Fpuppet-swift~828536","triplet_id":"openstack%2Fpuppet-swift~master~I976acde5128a53abc5af8d19fa066667578cdcf6","project":"openstack/puppet-swift","branch":"master","topic":"c8_teardown_master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-15 08:26:19.000000000","reason":"Change was submitted"}},"hashtags":["puppet-tripleo-job"],"change_id":"I976acde5128a53abc5af8d19fa066667578cdcf6","subject":"TripleO jobs: Replace CentOS 8 jobs by CentOS 9 jobs","status":"MERGED","created":"2022-02-09 13:56:10.000000000","updated":"2022-02-15 08:26:19.000000000","submitted":"2022-02-15 08:26:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828536-c8_teardown_master","meta_rev_id":"5759d959024e073f60363610ec356ca6071f3450","_number":828536,"virtual_id_number":828536,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-15 08:26:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2022-02-14 11:57:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2022-02-14 13:41:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2022-02-14 17:59:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-14 10:24:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":1,"date":"2022-02-14 17:59:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-09 16:43:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-14 10:24:19.000000000","updated_by":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"reviewer":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"state":"REVIEWER"},{"updated":"2022-02-14 11:57:26.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-02-14 13:41:20.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2022-02-14 17:59:25.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"59c351f52c37321c05aebe67cd83caf729d7417f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-09 13:56:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e31d7aa2a581cc5abb2c345c067379a7eb765cab","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2022-02-09 14:39:53.000000000","message":"Topic tripleo-c9 removed","accounts_in_message":[],"_revision_number":1},{"id":"15f1bcc973a65c244746ae68a75a80a41ced96d5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2022-02-09 14:40:03.000000000","message":"Topic set to c8_teardown_master","accounts_in_message":[],"_revision_number":1},{"id":"7b92e734eda5437c48e1a60bbe8fbd027692ebe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 16:43:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04523e334b6a4482b3b16bec8669f2b1 : SUCCESS in 4m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/006ea76dadd84ee0bd99a4127c7ab6f2 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ddebc95bc5a4f94bc150342e7f88ddd : SUCCESS in 3m 57s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d5b1a91c99ed4fec9c5bcaddfad3b3d7 : SUCCESS in 18m 02s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/661b0792786e4880bb01cbb5ca9beefc : SUCCESS in 19m 28s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c55fcb5995f24b878cbdcf016b9da801 : SUCCESS in 16m 15s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45a08ca1b98a4dfa85bf3548bb12b16f : SUCCESS in 13m 43s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b25d40ce7d314356b4298bead19586c2 : SUCCESS in 13m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85d0fdc3aa724685b05c3427c132636f : FAILURE in 39m 08s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b183e7acea9a4804b121ffb8a83c239e : FAILURE in 1h 47m 26s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d63057329b6449383a29305b80ca002 : SUCCESS in 53m 18s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3025f3deb7804ddb92f8f2759f061e12 : FAILURE in 32m 18s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/d54ea0e6a91449c1a04d413b500e28b6 : SUCCESS in 31m 18s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/761203c138bd4564b0a2077732494b8b : FAILURE in 10m 53s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44e467744cc0442eb45500f8407eb677 : FAILURE in 25m 21s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/842c137c56014db2951ab5832037f982 : FAILURE in 23m 49s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6f81f35ab8834feb96e9af6cd4c84677 : FAILURE in 2h 20m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b3f9a932ae2c83ef786d87e66b9afcbb5e1f88e7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 06:56:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b99320b51bf6fda6d226ce983f3954767ce9307","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 08:44:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89ee19e5c2094a8986e8c3e467139129 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa5605ea9c904b6f98499ed4f5804740 : SUCCESS in 4m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f823eb5a23f8428e83f85587d81b0dd9 : SUCCESS in 3m 27s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf5d25e719a348d094eee1080bd1efaf : SUCCESS in 17m 37s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/e0e428b6746b40dea696bc50a2004071 : SUCCESS in 14m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c1e3967cc08469f9c2ae2263ab7aff9 : SUCCESS in 10m 47s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f2ba0cdb230493b809d9910595b7e51 : SUCCESS in 14m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e32abc8a0292405aa5ac2a4d2a4c99d4 : SUCCESS in 11m 36s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e612ed9b72e64a8a97dccb572fcfb0e5 : FAILURE in 29m 46s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d1a18c6bbc643219ad493fb9f53179e : SUCCESS in 1h 41m 47s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b64ec4026bd41bba816e51c99df70d9 : SUCCESS in 1h 00m 00s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33cabe4e2b9c44f1b065516b73d79d2f : FAILURE in 23m 36s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0e447ebd7a74daa9f9774c9314e144a : SUCCESS in 29m 27s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/56c57c1333f44e2f8a9ab936a9e682fd : FAILURE in 8m 25s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1ce0326feca4adabacf0186c5a5a252 : FAILURE in 23m 44s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e6718bf42414c71bdb88a832bd122a7 : FAILURE in 20m 52s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2ca3bf99b34a4ac9883ebd86b09162b9 : SUCCESS in 1h 15m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3606b8b224fca3c0de538f29e4fa17282750b409","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 15:41:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"30bbcd557f2bc9433ef7c6ef67009e8e6188bfe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 17:46:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32e2f8a336254288a51e271ddc87e530 : SUCCESS in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7460251cb40d4d4db3f02c83b34f0179 : SUCCESS in 4m 37s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be93d252166741bab039f1d244fe2da8 : SUCCESS in 3m 36s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ccf2cf9b355645b8a946b45aaf1a619e : SUCCESS in 17m 56s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b78315054e1421b88a8a748cfaa8bf4 : SUCCESS in 16m 31s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5e02da1b7a042348d595677c8eac6fb : SUCCESS in 15m 35s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e7dd7cfd46b40748f7103fba1c2e012 : SUCCESS in 14m 44s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d33959279207436d93b061c6045fcdf9 : SUCCESS in 14m 53s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67ea31aa281e48f7bf76a36c5204476e : FAILURE in 51m 29s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/75068394ba6f43fa8594c69263132ce6 : SUCCESS in 1h 56m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2be5045b261046019ff59a9010ff79d5 : SUCCESS in 1h 18m 53s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1d725ea23524d15b36255d01d8c1fe6 : FAILURE in 24m 42s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba5ed3269f8c4bbd95a2e7af8028f543 : SUCCESS in 29m 26s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/43c2f595d14648eeaf404cd3debee407 : FAILURE in 9m 31s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcc792ef8c8342ba825d35d52c438774 : FAILURE in 24m 04s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b30cb32af8084ead84ca9d6d1c6159b3 : FAILURE in 24m 08s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3c232ba0072f45c794c39d66faa0e37b : FAILURE in 41m 40s","accounts_in_message":[],"_revision_number":2},{"id":"4cdeb2c2ff27531ad65d4ce84dc3fa7b62714721","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-11 09:10:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"09b3ad09236a00a2e00e32ce3cd191bab16d802a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 11:18:53.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43f1755a34ed42b9b36a9b3967fc2f1c : SUCCESS in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c15830ab58d0469b9c7c10bac936173b : SUCCESS in 3m 16s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/260cf35aac9f4400be02536c23a49eeb : SUCCESS in 3m 28s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1cb9a55ad17d494d992fd55443bdafd0 : SUCCESS in 15m 25s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/a64339f904bd43399e5c0c523121fbce : SUCCESS in 14m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8872923beb94cd7bbe939b170b609fa : SUCCESS in 17m 11s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/588792d4ebe74f42a29e190820acc947 : SUCCESS in 12m 21s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ad52281aef245adaa690fdd1d884f68 : SUCCESS in 11m 36s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d5cb92fcd1246f885bd6f57394832f3 : FAILURE in 43m 11s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5fbf4e56eb144276b962b5472174af9a : TIMED_OUT in 2h 04m 37s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/58e46a7657a54c8f8c3ddffc849a8117 : SUCCESS in 1h 13m 59s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e7c303105124e598b225321db872fd8 : FAILURE in 22m 51s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4bd1b9caa0e41b7b754c305f31421be : SUCCESS in 38m 57s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/28ee3c43deb94ccda46b8885659abdd1 : FAILURE in 9m 07s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32a4f7de51ef49cd97639a601c637f7b : FAILURE in 27m 41s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7eeffab94024755b7df551d988f88dc : FAILURE in 23m 27s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3136526f1f024c539325e60b67cf3c75 : FAILURE in 1h 52m 53s","accounts_in_message":[],"_revision_number":2},{"id":"f1fdd9f00cf613ca5bec9c095c16ea89a525beec","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-11 11:28:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2e2495e8ff222093c2e442aaec3790536904fdd4","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-11 11:30:13.000000000","message":"Hashtag added: puppet-tripleo-job","accounts_in_message":[],"_revision_number":2},{"id":"179c3975d98b0f5a3fcbd367ac417ffaac0f430c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 13:32:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d5052e8e3164b438885c60562fac2db : SUCCESS in 4m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05bf1680e9744a6b876f77f658ea6a57 : SUCCESS in 4m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35a6d9d0176442fea04f6d89fa0fb5e9 : SUCCESS in 4m 13s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/10a53e0074654b368923dd17a53402d7 : SUCCESS in 18m 49s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c2b22fad1d6f46868a747556e2486a65 : SUCCESS in 16m 25s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cec0742c19cd498c98920ef0eff5462b : SUCCESS in 15m 39s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9553363739fa4f14ace99b016c75164a : SUCCESS in 13m 38s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a92c9d11455f454587ae32f4f6ce7c21 : SUCCESS in 12m 03s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17b51896d0ff420198941b7892f61506 : FAILURE in 40m 12s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0418add6d8c49cd97465edaf2dc6cb3 : SUCCESS in 1h 59m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5ad7610b6f641c095c37e44311fa359 : SUCCESS in 1h 05m 00s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc603a5aec4147a39c09f6e78304c44d : FAILURE in 26m 12s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c1ca021e55fc4f0e939066c59b0e598c : SUCCESS in 28m 59s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/59f2e75d65a84b88ade8a6268c49c131 : FAILURE in 9m 19s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d0b6965146f4e60b6bb99cab9f1fd79 : FAILURE in 30m 47s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86c7236132f8451d9430cddc945a02db : FAILURE in 24m 02s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d04d2b3e9ed54fd4902b916d2c353a9e : SUCCESS in 1h 27m 04s","accounts_in_message":[],"_revision_number":2},{"id":"ada6961adfaf3099a93b3d7bbbfe42d6c285bdd5","author":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"date":"2022-02-14 10:24:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e616bb5dcb625925f522d91f8448842cff0f7ee6","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-02-14 11:57:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a6348ca716490267c4977a65856d634390b9541a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-02-14 13:41:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f7dc87a720e5ce7f162e4916b7a2e3b43940246d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-02-14 17:59:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"102839bea53f39ccffb865862da84c68d550c464","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-15 01:14:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"966accf46fba7ab5eae2619cc4e99dae570cded3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 02:42:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8eafc5f1331143b1864db3ab09a07c7e : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03cbedb0304e4ca59bea8cba2ff66dd5 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9522921b6f67465e9d766073368be4bb : SUCCESS in 3m 08s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f733ed83ec6e43079d2b09e27f87e060 : SUCCESS in 13m 15s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/b98e4ccc41744675aa6ab3004748a7b5 : SUCCESS in 14m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2521859d5dfb4e1688ed37dcd2b5a966 : SUCCESS in 11m 40s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3611ba2a5f9a475abc4fcf3ede1a2641 : SUCCESS in 11m 40s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c23577e8e12145e78b0b1bbfed13e6ab : SUCCESS in 12m 10s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ccc8b08b46549d880e8f603b6d50dca : FAILURE in 35m 11s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e4c8eed3407e4858b1b34649608105f7 : SUCCESS in 1h 25m 56s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/656447809df94b128fe91a3f8e6e99a2 : SUCCESS in 57m 48s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88bd0a40d9ec470ba4b0a4eca1a3fe7e : FAILURE in 22m 38s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/683178f2dfbb4b1cafe123c60fe99aa6 : SUCCESS in 29m 33s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/242ce397393b49fba316259508f43221 : FAILURE in 8m 03s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49e0983531d84aee8c79ea4fde41ce59 : FAILURE in 19m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58b52f6191ee4a76a2a237b8a60b22d0 : FAILURE in 23m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1a2a240b06dc59246e3588265dd3d98fba5cee42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 02:44:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2ec9c26d1ebbe904e0ef69f42ce1ff308ad3d1f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 04:59:19.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31f2c6cad4004d3fb4f2ccaa884912f5 : SUCCESS in 5m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/355581bce695444b8797b1f55e45cf91 : SUCCESS in 5m 25s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fad070cb8a4b4878a9aca8d51a3a6901 : SUCCESS in 5m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d10145f8ecb84962beda61f29e65147a : SUCCESS in 20m 07s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/e8f3d1fb9a5548a895493abd29c09137 : SUCCESS in 17m 31s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b19be0fe44344756ada1535609eed149 : SUCCESS in 16m 55s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8f1824e1fef45e48663690ea288c333 : SUCCESS in 14m 58s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a9ced4b45e740bba6665a0a806c5fe8 : TIMED_OUT in 2h 07m 31s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c878c80d5daa416bbfd39a5b68c6ad28 : SUCCESS in 26m 08s","accounts_in_message":[],"_revision_number":2},{"id":"1940d7daf0085afd7a63f82acc41a7e5b505110f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-02-15 05:02:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d4636613250609af8acb6298c222fa951f141d35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 06:39:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/000871e4faf14aefa6504e8506987668 : SUCCESS in 3m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47372b7597e34e9ca7cea7fa65e345db : SUCCESS in 3m 09s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e274bbf50f574ff7ad60652c765869e9 : SUCCESS in 2m 51s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/227dbfc26c1a4fa39e39b3a712088c34 : SUCCESS in 12m 26s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ec5af2d4c3d4341bb73c0656077e52e : SUCCESS in 11m 50s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b16907631864d7e9b8c8b4cd4f671be : SUCCESS in 11m 33s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd7cd8b521cf489086cd3e6cf802b538 : SUCCESS in 11m 33s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7882202e1f014274bd13abb44ea55c65 : SUCCESS in 12m 02s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46a147fe1d404b90bb4fd2eade452ebe : FAILURE in 40m 09s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ad73d319b6e4d48a87cc322f5502dbf : SUCCESS in 1h 34m 58s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc7014c1ae1b4ac8be65d91fb37f2b61 : SUCCESS in 49m 29s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5ff7b4222ed41e58579619213571f11 : FAILURE in 24m 47s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/56e417cdecbd466b9746b4e7c3648390 : SUCCESS in 23m 25s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea98499174374f5499c8172dd5848667 : FAILURE in 7m 12s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4971b89908b948b6a6289db407fbde20 : FAILURE in 26m 20s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce98c7e12dcc4ca8ab109bc5bbb7785f : FAILURE in 19m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ac20be37df7cdf61181be53715d88e160bbb83fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 06:40:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"76c8df9578f23d5b036730fae6449a0c873d076f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 08:26:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/764b919d952643da9c0f68809baa6cc5 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85fbb8dbd8984b7293663890ac0301c9 : SUCCESS in 3m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bd5eb71c347421c979d608007909313 : SUCCESS in 3m 34s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0c895b866094accb4ae601ec6713e1e : SUCCESS in 12m 56s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/56e6cea462a1446b82514cfb97aeaba3 : SUCCESS in 18m 28s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/345b271156a54e80943740d6478d6542 : SUCCESS in 12m 06s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6221802e0ec4ffbb5bdea8480d8be77 : SUCCESS in 10m 50s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/264031c943e44a948de8106a5916eb92 : SUCCESS in 1h 44m 42s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/9fa7a6c37a914bc592aef7791f6b0b16 : SUCCESS in 27m 34s","accounts_in_message":[],"_revision_number":2},{"id":"5759d959024e073f60363610ec356ca6071f3450","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 08:26:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a42ed4509f6d12fca15f4db886c300bd1bf9ceba","revisions":{"9303caea3e8b4e489a4c248f3057b80b206c9793":{"kind":"REWORK","_number":1,"created":"2022-02-09 13:56:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/828536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/36/828536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/36/828536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/36/828536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/36/828536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/36/828536/1"}}},"commit":{"parents":[{"commit":"4683fb4f785d3c9528a9e0f50f97d1e3bea9add3","subject":"Merge \"Loose dependencies about keystone endpoints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4683fb4f785d3c9528a9e0f50f97d1e3bea9add3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-09 13:55:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-09 13:55:55.000000000","tz":540},"subject":"TripleO jobs: Replace CentOS 8 jobs by CentOS 9 jobs","message":"TripleO jobs: Replace CentOS 8 jobs by CentOS 9 jobs\n\nNow TripleO team is replacing CentOS 8 jobs in master by CentOS 9 jobs.\nThis follows that and replaces tripleo jobs in puppet repos to switch\nto CentOS 9.\n\nDepends-on: https://review.opendev.org/828525\nChange-Id: I976acde5128a53abc5af8d19fa066667578cdcf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9303caea3e8b4e489a4c248f3057b80b206c9793"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9303caea3e8b4e489a4c248f3057b80b206c9793"}]},"branch":"refs/heads/master"},"a42ed4509f6d12fca15f4db886c300bd1bf9ceba":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-10 15:41:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/828536/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/36/828536/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/36/828536/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/36/828536/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/36/828536/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/36/828536/2"}}},"commit":{"parents":[{"commit":"7e86bb55caefa824a252d590b66b9984ffb974d7","subject":"TripleO: Enable container build in standalone job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7e86bb55caefa824a252d590b66b9984ffb974d7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-09 13:55:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-10 15:41:27.000000000","tz":0},"subject":"TripleO jobs: Replace CentOS 8 jobs by CentOS 9 jobs","message":"TripleO jobs: Replace CentOS 8 jobs by CentOS 9 jobs\n\nNow TripleO team is replacing CentOS 8 jobs in master by CentOS 9 jobs.\nThis follows that and replaces tripleo jobs in puppet repos to switch\nto CentOS 9.\n\nDepends-on: https://review.opendev.org/828525\nChange-Id: I976acde5128a53abc5af8d19fa066667578cdcf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a42ed4509f6d12fca15f4db886c300bd1bf9ceba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a42ed4509f6d12fca15f4db886c300bd1bf9ceba"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
