)]}'
{"id":"openstack%2Fpuppet-swift~841825","triplet_id":"openstack%2Fpuppet-swift~master~Ic3944a18be3a02be58be799dc4b4978bdd8aa674","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-05-18 08:07:23.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"Ic3944a18be3a02be58be799dc4b4978bdd8aa674","subject":"Skip creating user/directory created by the packages","status":"ABANDONED","created":"2022-05-16 01:48:34.000000000","updated":"2023-02-03 04:09:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"33b04a2467f4143fa59b87a32814eca62a79f9df","_number":841825,"virtual_id_number":841825,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-16 03:31:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2022-05-18 08:07:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-16 03:31:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0530d1dc929e84740cc5f3075e6fa2fcdfde3e15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-16 01:48:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7652dd464567afb1c04708ff46c0278584d8095b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 03:31:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46e6634e0f5e41cab87c535590b3af47 : SUCCESS in 2m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a4b431ec7e14ab19ca3315a90352514 : SUCCESS in 2m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97c2810d99354783b57b5ed2b89ba0ec : SUCCESS in 2m 43s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6bcf9a3673b4a89857c6e56cb275a99 : FAILURE in 13m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c9cedc0ab79415590322e6df4ac298e : FAILURE in 19m 16s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/500f36231f354868a54f2cfc9e894bbb : FAILURE in 12m 24s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75683cf564b547a1b9d921035a8c086d : FAILURE in 11m 30s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ffc8a29474947819fa041bf433ed216 : FAILURE in 37m 35s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/23fadf09a0e14b9b93880f8259f9fc63 : SUCCESS in 1h 03m 36s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/751c85a4ea0c44b79b4aeb95db3385bc : FAILURE in 18m 47s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/6eb975d86e784de08c7d6e992eff6af0 : SUCCESS in 25m 44s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f0ea1c8f3d54976a1386c897f8c2378 : FAILURE in 10m 26s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09ac416f12f14ccea54050a2ef01dbab : FAILURE in 15m 18s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7bddee11f9f4777a1a6f31beac0c95e : FAILURE in 15m 40s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/de3cc8ab08bb44c5a93702cc09e0998a : SUCCESS in 1h 39m 56s","accounts_in_message":[],"_revision_number":1},{"id":"182fba5cdac21a2283f712fb93c173a538999adb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-18 08:07:23.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"33b04a2467f4143fa59b87a32814eca62a79f9df","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-03 04:09:43.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b4f280092ed95ac50297dffa9c8fe0457b6ff4a1","revisions":{"b4f280092ed95ac50297dffa9c8fe0457b6ff4a1":{"kind":"REWORK","_number":1,"created":"2022-05-16 01:48:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/841825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/25/841825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/25/841825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/25/841825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/25/841825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/25/841825/1"}}},"commit":{"parents":[{"commit":"b0c9ae29ab415ecb739b8d1a92dacfde999e983b","subject":"Revert \"Avoid duplicate package swift.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b0c9ae29ab415ecb739b8d1a92dacfde999e983b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-16 01:43:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-16 01:46:53.000000000","tz":540},"subject":"Skip creating user/directory created by the packages","message":"Skip creating user/directory created by the packages\n\nChange-Id: Ic3944a18be3a02be58be799dc4b4978bdd8aa674\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b4f280092ed95ac50297dffa9c8fe0457b6ff4a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b4f280092ed95ac50297dffa9c8fe0457b6ff4a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
