)]}'
{"id":"openstack%2Fpuppet-swift~847767","triplet_id":"openstack%2Fpuppet-swift~master~Ifdf86a6310fa8f8405579bd181e2070aa93d0b23","project":"openstack/puppet-swift","branch":"master","topic":"trivial-fix","attention_set":{},"removed_from_attention_set":{"20733":{"account":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"last_update":"2022-06-27 16:42:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifdf86a6310fa8f8405579bd181e2070aa93d0b23","subject":"Fix typo in parameter name","status":"MERGED","created":"2022-06-27 11:33:08.000000000","updated":"2022-06-27 16:42:41.000000000","submitted":"2022-06-27 16:42:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847767-trivial-fix","meta_rev_id":"56c488311bcb59632ef1491401edbd633b4d4cd3","_number":847767,"virtual_id_number":847767,"owner":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-27 16:42:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-06-27 15:17:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-06-27 13:31:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-06-27 15:17:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-27 13:30:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-27 13:31:58.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-06-27 15:17:26.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"7f4628483ef14a269ab11559b8b401b8885b78ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2022-06-27 11:33:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b26a465b175ce5ec5856dc25b4174686f4e3e68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 13:30:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd113fe2a9404002bf7a2a652c27aecd : SUCCESS in 4m 07s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c5f0e5c3d4848a8afa6cc06e097e9d2 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18c8bc8b0c074c408ac958aa1b9739d3 : SUCCESS in 3m 16s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a63327be66d48d09a927df26a83c59e : SUCCESS in 13m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/019ae95f815a40a0bd5c118b7472a38d : SUCCESS in 16m 41s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/826ec8ae894f41af96e0ebdba9b86ba7 : SUCCESS in 13m 22s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36446686b98f4a18a3fe52a83c90e8de : SUCCESS in 13m 40s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c5710844ccb402da93a339160a35293 : FAILURE in 31m 53s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a77a58bc56094508905b408b3b4fdf41 : SUCCESS in 1h 03m 39s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30f5683eb11744b2809f0d6b6b6b283f : FAILURE in 16m 44s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b09b23da70f424ca3d883dc1997addc : SUCCESS in 26m 01s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c782d2ca294944059af219e47ef27583 : FAILURE in 17m 22s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da77f2d403754b28ac1196c5dd965cd4 : FAILURE in 16m 37s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cce0919484d14fb3a34d7e51872ac353 : SUCCESS in 1h 51m 02s","accounts_in_message":[],"_revision_number":1},{"id":"2f8c60056f549b555e92ccceed34763d5b3ccfd6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-06-27 13:31:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"089043bfad8e36c65b33992a4255ca4710f55743","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-27 15:17:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"48b5b607f64ecfc4e0374ca1619117c609fb72a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 15:18:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"48fe57c2036f64778d959207eb5b08ebf2a811df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 16:42:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3a0ebb5232340169e341ef22f09ea6c : SUCCESS in 4m 55s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3aaf49e1740a467c84a5070894643158 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7c650e5d71f444096a64b14f4353312 : SUCCESS in 4m 34s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/acc68e84aa764e4d8bd5f3f4eb2656da : SUCCESS in 17m 51s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abe36311e97246f3a00f19603a71a666 : SUCCESS in 14m 12s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f4c25bb18f6450ca1e53159f6af4f2d : SUCCESS in 14m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/78bc27d08c48462994e3640ac6446713 : SUCCESS in 59m 33s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/dad77402224a4937b75a5bf1ffc6b1f5 : SUCCESS in 27m 57s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/93b1bff4b80948579ff9f4dfb18893f9 : SUCCESS in 1h 22m 50s","accounts_in_message":[],"_revision_number":1},{"id":"56c488311bcb59632ef1491401edbd633b4d4cd3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 16:42:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"814ec2de3617bde1a0423e39ef1ee987c2acc415","revisions":{"814ec2de3617bde1a0423e39ef1ee987c2acc415":{"kind":"REWORK","_number":1,"created":"2022-06-27 11:33:08.000000000","uploader":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"ref":"refs/changes/67/847767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/67/847767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/847767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/847767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/847767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/67/847767/1"}}},"commit":{"parents":[{"commit":"b4c9ec059e1eb3b62e7e5c0bc4982f76efef76ef","subject":"Merge \"Remove support for CentOS 8 Stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b4c9ec059e1eb3b62e7e5c0bc4982f76efef76ef"}]}],"author":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2022-06-27 11:29:48.000000000","tz":330},"committer":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2022-06-27 11:33:04.000000000","tz":330},"subject":"Fix typo in parameter name","message":"Fix typo in parameter name\n\nThis change fixes typo in parameter name from\n`dnspython_pakage_name` to `dnspython_package_name`.\n\nThe tests were using correct parameter name but in manifests\nincorrect name was used.\n\nChange-Id: Ifdf86a6310fa8f8405579bd181e2070aa93d0b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/814ec2de3617bde1a0423e39ef1ee987c2acc415"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/814ec2de3617bde1a0423e39ef1ee987c2acc415"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
