)]}'
{"id":"openstack%2Fpuppet-swift~847822","triplet_id":"openstack%2Fpuppet-swift~stable%2Fxena~I28ac3aacdeca8dc5803fd053e72e80256d82868f","project":"openstack/puppet-swift","branch":"stable/xena","topic":"trivial-fix","attention_set":{},"removed_from_attention_set":{"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2022-06-29 19:10:37.000000000","reason":"removed on reply"},"20733":{"account":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"last_update":"2022-06-30 03:12:20.000000000","reason":"\u003cGERRIT_ACCOUNT_20733\u003e replied on the change","reason_account":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}}},"hashtags":[],"change_id":"I28ac3aacdeca8dc5803fd053e72e80256d82868f","subject":"Fix typo in swiftinit service type","status":"MERGED","created":"2022-06-28 11:20:37.000000000","updated":"2022-06-30 05:57:11.000000000","submitted":"2022-06-30 05:57:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847822-trivial-fix","cherry_pick_of_change":847812,"cherry_pick_of_patch_set":1,"meta_rev_id":"143b9e4c7406afbf74f9b02fe05b887a73f24e55","_number":847822,"virtual_id_number":847822,"owner":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-30 05:57:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-06-29 10:39:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-06-29 19:10:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-06-29 19:10:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-28 13:08:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-28 13:28:49.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-06-28 17:20:16.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"7b3b96bdb934896487ca7e5e95a28f9bffb27da6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2022-06-28 11:20:37.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"10f869c009f1c140d5e0200437ce4ad4858759e5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2022-06-28 11:20:44.000000000","message":"Topic trivial-fix-stable/xena removed","accounts_in_message":[],"_revision_number":1},{"id":"cf26264d66f3aaff8107106b869618edf6ca01d7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2022-06-28 11:20:48.000000000","message":"Topic set to trivial-fix","accounts_in_message":[],"_revision_number":1},{"id":"66031d86577d7df7402df40311a9dc837c4b9491","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 13:08:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad874d2e1716450b8752a1388260b38f : SUCCESS in 4m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3371ff5d5ad48c1aec2f817fba23758 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbf9c1590efb425fbb1e9394598914d3 : SUCCESS in 3m 34s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8c87b835595f485c91d3071416f4e30f : SUCCESS in 18m 12s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0aa1f59150a422d9632d85509b136b2 : SUCCESS in 17m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2887f9eb1303464a93080b9d720c410e : SUCCESS in 15m 08s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d888203049fe428fa0b4da771c479108 : SUCCESS in 15m 57s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e60a89bf88941c4a040f03782f0c7eb : SUCCESS in 14m 16s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f9e2423b41e425da0d87bd88d6281b9 : FAILURE in 1h 02m 42s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f6ef5551563b4fe393987db46e43b54f : SUCCESS in 1h 37m 54s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24a48f3a8cbd41619399204bac6ba946 : FAILURE in 24m 25s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/32090e819d77442aa9dd1aca035ee506 : SUCCESS in 31m 24s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3fbd3d2bce1460fbbddbe8dab34eb6e : FAILURE in 23m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/895935f2d3d541d3acf62f7fa9a4b775 : FAILURE in 23m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e8da465e186dcf19bc80ad9b3ef381118003d42","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-28 13:28:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7bff8682d6e09f4233bfb81b70492564508a55a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2022-06-29 10:39:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8bbc6502715942eb45cd8d220ca2cdeb36849858","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2022-06-29 10:40:07.000000000","message":"Topic trivial-fix-stable/xena removed","accounts_in_message":[],"_revision_number":2},{"id":"1ffd5c75b819dcd68218902d00949e6192925a2c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2022-06-29 10:40:12.000000000","message":"Topic set to trivial-fix","accounts_in_message":[],"_revision_number":2},{"id":"c1877e1de92f45dd9c34a6ee31ee9d261c490cec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 11:57:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48ceb4f3f2fb4912bb4b7b9971bab68f : SUCCESS in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ff3ba8341ba4cb08571f9efa2aa81bb : SUCCESS in 3m 24s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46aac7dff4264e0b89647625aba56b3f : SUCCESS in 3m 41s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2fbe58d3a2434a05a4896588718aab3d : SUCCESS in 16m 15s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/231f7afc49324a168a4ca38b6d58dc69 : SUCCESS in 15m 23s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18d7c7a9b0684897bbba68964c9f28c7 : SUCCESS in 14m 07s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa80fd4ae0a74681bddb962533a00f5f : SUCCESS in 13m 32s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64a58a24d2a84a189dd608954c85c457 : SUCCESS in 13m 51s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac07886116f04a19a1450db7e0992d07 : FAILURE in 58m 49s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bad606897a5148afa1965abf018aab71 : SUCCESS in 1h 16m 25s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4dea7b4484a34bc190fe0d6be0523a11 : FAILURE in 24m 53s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2b6bf8fe0e14c62880f5e9d6a0c07dd : SUCCESS in 30m 42s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50385a45fd9545aa8ff1f6a0f8c07afe : FAILURE in 25m 31s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91c099adca944a5ba933118bfb3340c1 : FAILURE in 22m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8020a181693af5960d21a39cc495620e7b6a9789","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-06-29 19:10:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6ba8d2197711511ec7a53dd27aab5c5cb0ffa0a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 19:11:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"de4587c04d6ed7ddf3c0b292c4d3b0a3144136b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 21:02:51.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49114cafb61a44189f09ca99f32d791c : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12eb6ad2203044368022a3108ad9923a : SUCCESS in 3m 24s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69e0352b65224a50aa6005f71f49c738 : SUCCESS in 3m 45s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e048b7ba7dac480db42624e53f524ae4 : SUCCESS in 15m 43s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/826361e947f046b1aa106a72b509dbb1 : SUCCESS in 16m 59s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3dab89d3e844bfbbfa6e4cf0f61ff09 : SUCCESS in 13m 54s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16093eca5d5f4d15be12507d0d7bf429 : SUCCESS in 15m 25s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ff6b3a695f142e9b06753a3fae21d3c : POST_FAILURE in 1h 48m 16s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/32d13c266b9442bcacc6fc9ad983ee7a : SUCCESS in 28m 31s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de03ceaa8af14bb78bed8eb77d566008 : FAILURE in 21m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c296edeafcc13e92da2a0167890fee40b8c8dcfd","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2022-06-30 03:12:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed0c919c9ae6ef00201727cc98fde34bf655b53c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 04:45:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e60261dd2e444068cf4c8e5bb360159 : SUCCESS in 2m 55s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f577f649fc914b05b4235534b90f3e9e : SUCCESS in 3m 50s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0488e19ec7c542c389c5efa245781ca8 : SUCCESS in 3m 17s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c98f4b24960c4d36b8a9c76756c48db0 : SUCCESS in 14m 17s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/4f8c8e96c531440c888df854854fd852 : SUCCESS in 15m 26s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c424747fff74e94a2577f37f0255fff : SUCCESS in 14m 11s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93c8933a11b44253801038ec3dada136 : SUCCESS in 12m 39s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbf3c1e1c0274ef795e36aacac96d52f : SUCCESS in 13m 13s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b199f82baff43f5a1a0836112aede9b : FAILURE in 56m 45s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a2a15bfd9274faeb63fe71ebb2a7a1d : SUCCESS in 1h 31m 04s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b46dcf656ec146dc871fef250c0a792d : FAILURE in 23m 10s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/722820aa17e8409e94f3fe835983d170 : SUCCESS in 26m 23s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cebcf105d39745548d18a72c2ac5ac01 : FAILURE in 23m 06s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc47fd294bea42d590b881e91f49c17a : FAILURE in 23m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d938bd85e15a82c3364991bd8508ab54cdbcebae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 04:46:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8047e7c537e98230a05ac8308980fb11edbe6ce4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 05:57:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a71254dec974d1d984aa967a8fb562c : SUCCESS in 3m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/821af716505347528582c5de1ab562f0 : SUCCESS in 2m 56s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0b3a963c7764014bf5b18ed72019612 : SUCCESS in 3m 00s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f535a3cf0b12444cb33e66dfe773a11c : SUCCESS in 14m 52s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ccbe2dc8b2414d21a37b7b1d1d4bda12 : SUCCESS in 15m 34s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/128e1c0c7d994afe908f6cd6e46d956b : SUCCESS in 12m 35s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b53702c32df4ccdacfcdee2ffd19b9c : SUCCESS in 13m 16s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6f51fcbef264376876fc54241624ece : SUCCESS in 1h 09m 11s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/818e03f3f29e470e8f39a2b9b8b8be3e : SUCCESS in 29m 59s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/901f4a9263d648648bf91e98e3ae06db : FAILURE in 22m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"143b9e4c7406afbf74f9b02fe05b887a73f24e55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 05:57:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"331b92688edfb4e919d86f03395f0e4b68a5be5b","revisions":{"f7e5e1d0dff295ea632f5ba1d4bee33e6e7728cf":{"kind":"REWORK","_number":1,"created":"2022-06-28 11:20:37.000000000","uploader":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"ref":"refs/changes/22/847822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/22/847822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/847822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/847822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/847822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/22/847822/1"}}},"commit":{"parents":[{"commit":"e7d560738d5e02d32e63a0411211ddbb8afdcd9c","subject":"Fix missing argument of builder_file_path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e7d560738d5e02d32e63a0411211ddbb8afdcd9c"}]}],"author":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2022-06-27 11:34:54.000000000","tz":330},"committer":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2022-06-28 11:20:37.000000000","tz":0},"subject":"Fix typo in swiftinit service type","message":"Fix typo in swiftinit service type\n\nChange-Id: I28ac3aacdeca8dc5803fd053e72e80256d82868f\n(cherry picked from commit 9fcfcb2f2a7c5b9923f30264e28cfeaf4ead28b1)\n(cherry picked from commit 447044525358bcc79f1380ca4429f9ef7a22b542)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f7e5e1d0dff295ea632f5ba1d4bee33e6e7728cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f7e5e1d0dff295ea632f5ba1d4bee33e6e7728cf"}]},"branch":"refs/heads/stable/xena"},"331b92688edfb4e919d86f03395f0e4b68a5be5b":{"kind":"NO_CHANGE","_number":2,"created":"2022-06-29 10:39:05.000000000","uploader":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"ref":"refs/changes/22/847822/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/22/847822/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/847822/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/847822/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/847822/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/22/847822/2"}}},"commit":{"parents":[{"commit":"e7d560738d5e02d32e63a0411211ddbb8afdcd9c","subject":"Fix missing argument of builder_file_path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e7d560738d5e02d32e63a0411211ddbb8afdcd9c"}]}],"author":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2022-06-27 11:34:54.000000000","tz":330},"committer":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2022-06-29 10:39:05.000000000","tz":0},"subject":"Fix typo in swiftinit service type","message":"Fix typo in swiftinit service type\n\nChange-Id: I28ac3aacdeca8dc5803fd053e72e80256d82868f\n(cherry picked from commit 9fcfcb2f2a7c5b9923f30264e28cfeaf4ead28b1)\n(cherry picked from commit 447044525358bcc79f1380ca4429f9ef7a22b542)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/331b92688edfb4e919d86f03395f0e4b68a5be5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/331b92688edfb4e919d86f03395f0e4b68a5be5b"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
