)]}'
{"id":"openstack%2Fpuppet-swift~847974","triplet_id":"openstack%2Fpuppet-swift~master~Ia990bffee54f6c0344060f16cf418a6c39ee81fc","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-01 11:08:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia990bffee54f6c0344060f16cf418a6c39ee81fc","subject":"Remove deprecated swift::test_file","status":"MERGED","created":"2022-06-28 13:35:38.000000000","updated":"2022-08-01 11:08:19.000000000","submitted":"2022-08-01 11:08:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847974","meta_rev_id":"d36b5fab91f9b3970911c2e0f424850fd755d559","_number":847974,"virtual_id_number":847974,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-01 11:08:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-06-29 19:10:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-08-01 07:44:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-08-01 07:44:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-28 15:51:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-29 19:10:47.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-08-01 07:44:11.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"db03854f8ed39a0f56e9d31bc6bfb6527e2467fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-28 13:35:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aac40687f0fb6bc23a77eb439a48e2afbce4084d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 15:51:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70278b14d2b742f58530d3d28732e67d : SUCCESS in 3m 33s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8337c71a703b4c9dae0b551fdcd52fb4 : SUCCESS in 3m 04s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf127cf5ef754bf09c27b4bf91b2daa9 : SUCCESS in 3m 29s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb53d152b5624bc4b889cadc4f5d6994 : SUCCESS in 15m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d757ada7ec874737b81e8282a22fcf44 : SUCCESS in 14m 28s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3f1fc8ed21d42f399b094271faf1f42 : SUCCESS in 13m 20s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44c9d9c5dce34d3586f7b1383a06e8f7 : SUCCESS in 14m 19s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c77b04c1939a46e59a23e3c0e61d9e7a : FAILURE in 43m 24s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6092bc89c654462a9397b35f914fa607 : SUCCESS in 1h 00m 52s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8212110e9dc467c9601db61a1545fdf : FAILURE in 17m 53s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c5893043ab947d28b8f8a2fa0966278 : SUCCESS in 24m 06s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32d03ebcc7664d81a8dfce688a91769b : FAILURE in 18m 35s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5077dc8b3df04ac399b8c0724079007d : FAILURE in 18m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a4517d9dfa947d48c838c6c32ab0a8d : SUCCESS in 3m 37s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/84f87e9741ad421faa29f63b9a556cbb : SUCCESS in 2h 09m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f9d54d3acad47c71a6b9b76f8885960b39eb25a9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-06-29 19:10:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f513681e3baf86bf78359e3f16c7b9435b0bb14d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-01 07:44:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5118470ccba62539f817cc8770467621873b7a1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 07:46:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5313aaec7784df8537fed530d9cae293e95b03be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 11:08:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ad66ccffd3846dea613a22526d125ea\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb05ce7ea1a5477397e2afa21cf358db : SUCCESS in 4m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a35a2f18c4b24e34acc352229cbd5522 : SUCCESS in 4m 41s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f74147b5bf445caa4814d2eaf05ffd7 : SUCCESS in 4m 00s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/38b5f2bfbee948d69ef9fb7094c7cdf8 : SUCCESS in 20m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5f4f1b0d6dd43e69f1ea6fac6ee8a43 : SUCCESS in 16m 59s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e40c09d070254311872553a586de0e76 : SUCCESS in 15m 31s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6f7f12a7fef4af2ae8897de641e2bde : SUCCESS in 1h 12m 27s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/907da0c9aa33452388c237699d632fc1 : SUCCESS in 34m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80edc700751d41b1b772ca224cb6fcf1 : SUCCESS in 5m 04s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9a4c0cc3bf5c4429938d15d23983df04 : SUCCESS in 1h 38m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d36b5fab91f9b3970911c2e0f424850fd755d559","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 11:08:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1414bd7d85f0efddc96ba3c0b2355fada2943a8e","revisions":{"1414bd7d85f0efddc96ba3c0b2355fada2943a8e":{"kind":"REWORK","_number":1,"created":"2022-06-28 13:35:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/847974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/74/847974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/74/847974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/74/847974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/74/847974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/74/847974/1"}}},"commit":{"parents":[{"commit":"718155cc227c0da39f8c1951a9aa276a0b57719f","subject":"Merge \"Fix outdated constraint of puppet-rsync and puppet-xinetd\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/718155cc227c0da39f8c1951a9aa276a0b57719f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-28 13:34:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-28 13:35:27.000000000","tz":540},"subject":"Remove deprecated swift::test_file","message":"Remove deprecated swift::test_file\n\n... because it was deprecated during Yoga cycle[1].\n\n[1] dd3a5bbf0dfd98fe9bf27e130a317640a1621bf5\n\nChange-Id: Ia990bffee54f6c0344060f16cf418a6c39ee81fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1414bd7d85f0efddc96ba3c0b2355fada2943a8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1414bd7d85f0efddc96ba3c0b2355fada2943a8e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
