)]}'
{"id":"openstack%2Fpuppet-swift~847976","triplet_id":"openstack%2Fpuppet-swift~master~I237aa3196d24a287fd891c0a818dfd20bfd64d04","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-29 22:38:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I237aa3196d24a287fd891c0a818dfd20bfd64d04","subject":"Remove deprecated allow_versions","status":"MERGED","created":"2022-06-28 13:39:09.000000000","updated":"2022-06-29 22:39:53.000000000","submitted":"2022-06-29 22:38:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847976","meta_rev_id":"1c800518544f94cda72fe434d5c899c1c4c512e6","_number":847976,"virtual_id_number":847976,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-29 22:38:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-06-28 14:28:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-06-29 19:10:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-06-29 19:10:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-28 14:28:02.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-06-28 15:30:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-29 19:10:50.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"a707f6051a5ab0957509114f186f2bc01db6e44c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-28 13:39:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b25c9ab27530818a045ffc783d9be51011ac2a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-28 13:40:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4435ca22dd223f592fbb82a5b352f19594c70fca","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-06-28 14:28:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a01c4a8c605839237b75823dcbba2cea311f008f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 15:30:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/905b35c41452402ea8aa6a2631a8797f : SUCCESS in 3m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ca4f16269aa47f8928d227572a8418d : SUCCESS in 3m 30s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7331634022dc4694813d36f8b813a606 : SUCCESS in 3m 28s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/cccca57ec1974ee191c1f2e3cc4090f3 : SUCCESS in 15m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/baf0b39b48bd483aa1963ac4ad1bc0ab : SUCCESS in 13m 41s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25b6c5adfd7a442bb5ce2f42eefee971 : SUCCESS in 13m 39s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b48e0a043ce453dbd32c28097931d62 : SUCCESS in 12m 46s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d37d3a1b40bd45328bc79af57d8401c3 : FAILURE in 40m 48s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b1dd0a9c0be4afe807e42c0b7859401 : SUCCESS in 1h 00m 24s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d892e3768574adca4120832aa641018 : FAILURE in 17m 54s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/43ed0c27a43c4315a13a318bfa2a3380 : SUCCESS in 25m 33s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0d8242e90634cd9833d2ddeb6e560bf : FAILURE in 18m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c51d1561a21479a89ea285895167655 : FAILURE in 16m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1825f7193f3b42c897b65e405c636a37 : SUCCESS in 3m 22s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/429be145b4684da59bdeb014daf33bb7 : SUCCESS in 1h 44m 06s","accounts_in_message":[],"_revision_number":2},{"id":"206fb373b5c8254dd73fd75edea255443d5eff48","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-06-29 19:10:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"07b88b41c5be7bbc5da0b8035e2964aaba8d78ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 19:12:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bc387993d76a99a6517a6fbf46174baa3875b1d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 22:38:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12170eca749945d98d166ba52699ec2e : SUCCESS in 3m 07s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/358083b1e8b840ef8483299616ffb14e : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/142bb6047f0445439ecc7b782114a23d : SUCCESS in 3m 15s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb8fcd6aa8d9410cae3db8920c49ef52 : SUCCESS in 14m 56s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ff7dcf6c5364da99d2f9b335301af8d : SUCCESS in 13m 20s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b62ce77921746f0be9f4bb9a31bc56e : SUCCESS in 11m 57s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8c81100820d41c1b8c3d137418b4748 : SUCCESS in 1h 07m 02s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/68e3a9fd11204f92b3177c4720c87c47 : SUCCESS in 25m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05ffdf6acdfc4a9f8d37fa38c4f3ce2f : SUCCESS in 3m 01s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/93ee73154ffd4bd78c9a9900ebe0760b : SUCCESS in 1h 33m 14s","accounts_in_message":[],"_revision_number":2},{"id":"89558f504d0632d15ff5c140a024dbbf980e3e32","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 22:38:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1c800518544f94cda72fe434d5c899c1c4c512e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 22:39:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d39209bd39d4d738feac86f5e6afac6 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e004e4627384f2141e2b1a1178496725117315b8","revisions":{"55ead03abacd75aafdb7c414f0ace812ae5d6cc2":{"kind":"REWORK","_number":1,"created":"2022-06-28 13:39:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/847976/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/76/847976/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/76/847976/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/76/847976/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/76/847976/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/76/847976/1"}}},"commit":{"parents":[{"commit":"718155cc227c0da39f8c1951a9aa276a0b57719f","subject":"Merge \"Fix outdated constraint of puppet-rsync and puppet-xinetd\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/718155cc227c0da39f8c1951a9aa276a0b57719f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-28 13:38:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-28 13:38:42.000000000","tz":540},"subject":"Remove deprecated allow_versions","message":"Remove deprecated allow_versions\n\n... because the parameters were deprecated during Yoga cycle[1].\n\n[1] a1088139ec254f3e8283a854cc78534abfb6ac08\n\nChange-Id: I237aa3196d24a287fd891c0a818dfd20bfd64d04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/55ead03abacd75aafdb7c414f0ace812ae5d6cc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/55ead03abacd75aafdb7c414f0ace812ae5d6cc2"}]},"branch":"refs/heads/master"},"e004e4627384f2141e2b1a1178496725117315b8":{"kind":"REWORK","_number":2,"created":"2022-06-28 13:40:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/847976/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/76/847976/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/76/847976/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/76/847976/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/76/847976/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/76/847976/2"}}},"commit":{"parents":[{"commit":"718155cc227c0da39f8c1951a9aa276a0b57719f","subject":"Merge \"Fix outdated constraint of puppet-rsync and puppet-xinetd\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/718155cc227c0da39f8c1951a9aa276a0b57719f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-28 13:38:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-28 13:40:26.000000000","tz":540},"subject":"Remove deprecated allow_versions","message":"Remove deprecated allow_versions\n\n... because the parameters were deprecated during Yoga cycle[1].\n\n[1] a1088139ec254f3e8283a854cc78534abfb6ac08\n\nChange-Id: I237aa3196d24a287fd891c0a818dfd20bfd64d04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e004e4627384f2141e2b1a1178496725117315b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e004e4627384f2141e2b1a1178496725117315b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
