)]}'
{"id":"openstack%2Fpuppet-swift~848543","triplet_id":"openstack%2Fpuppet-swift~master~I9cdf102cc517222d64dc09cb5369fc6ba675304c","project":"openstack/puppet-swift","branch":"master","hashtags":[],"change_id":"I9cdf102cc517222d64dc09cb5369fc6ba675304c","subject":"Remove remaining hard-code of username in each manifest","status":"ABANDONED","created":"2022-07-03 16:41:06.000000000","updated":"2022-08-25 06:23:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"51532fd49c718dfada96e9f9fc8e0db46b4a9410","_number":848543,"virtual_id_number":848543,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-03 18:27:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2022-07-04 01:12:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-03 18:27:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb3d1d8dff2c804923844c1e699dfed69bda1ea6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-03 16:41:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1df85976815eb9f2727257253cc5c49d5159024c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-03 16:57:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a42f71ddebfc1349f838c26f050b1ee2fadd02f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-03 17:04:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6e406d4e731d3c5d0c7c2be5d593b038a770b2cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-03 18:27:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/208dd51656184bedaab9af617dc1e390\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/531839cc9d3a4c37a9277f54930e3dad : SUCCESS in 3m 17s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b5ea0cbe89b4069b84ccc7a37c054d7 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98f4bde23c584c3e941634b76debe7b2 : SUCCESS in 3m 31s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab81bb881b9a49179407e963cb084473 : SUCCESS in 13m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0272cf2776824e12ad3fb635f991e163 : SUCCESS in 10m 18s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adf7a0e76af945f5aa02dc0671065b8c : SUCCESS in 13m 03s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19d0c912e2404f2a9ba380d41d637ad1 : SUCCESS in 11m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25a6440eb2a84280a5268867ddfafb68 : FAILURE in 31m 17s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f3d9f8229ce405d8f0cdd7865c5259d : SUCCESS in 52m 15s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/199f04a2e4434a1b903a96d1820e0c51 : FAILURE in 20m 08s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/90bc6ba9e515499b9ced4978a795c7fd : SUCCESS in 22m 29s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d1d635ef4bd4a7fa8687182121dbb03 : FAILURE in 14m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6657014440754ed6b50a440ab7698dc5 : FAILURE in 18m 11s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/37fb3bfbf7814e7fbf3a39e8c8814c3c : SUCCESS in 1h 20m 12s","accounts_in_message":[],"_revision_number":3},{"id":"492eab80e9f77a3ae719a6f50ab3bac540078ba4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-04 01:12:57.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"51532fd49c718dfada96e9f9fc8e0db46b4a9410","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-25 06:23:16.000000000","message":"Abandoned\n\ndone by https://review.opendev.org/c/openstack/puppet-swift/+/854490","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"57d0f6da9ff7fe999d58888c4c7ac33e4c04c407","revisions":{"5118e29e608035412f0abd28bc5ffce340d56a3a":{"kind":"REWORK","_number":1,"created":"2022-07-03 16:41:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/848543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/43/848543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/43/848543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/43/848543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/43/848543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/43/848543/1"}}},"commit":{"parents":[{"commit":"718155cc227c0da39f8c1951a9aa276a0b57719f","subject":"Merge \"Fix outdated constraint of puppet-rsync and puppet-xinetd\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/718155cc227c0da39f8c1951a9aa276a0b57719f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-03 16:40:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-03 16:40:26.000000000","tz":540},"subject":"Remove remaining hard-code of username in each manifest","message":"Remove remaining hard-code of username in each manifest\n\nThis is follow-up of f781eb9e34501b651d3f2069fc8848e7cacef679 and\nreplaces remaining hard-coded user name by the central definition in\nswift::params.\n\nChange-Id: I9cdf102cc517222d64dc09cb5369fc6ba675304c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5118e29e608035412f0abd28bc5ffce340d56a3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5118e29e608035412f0abd28bc5ffce340d56a3a"}]},"branch":"refs/heads/master"},"640ec3a3eb5860f7a09847058a5ae6f3ea06b9f5":{"kind":"REWORK","_number":2,"created":"2022-07-03 16:57:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/848543/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/43/848543/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/43/848543/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/43/848543/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/43/848543/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/43/848543/2"}}},"commit":{"parents":[{"commit":"152587fd2a1d0c0bc7ad67585fe2ecdd6abde995","subject":"Remove unnecessary group assignment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/152587fd2a1d0c0bc7ad67585fe2ecdd6abde995"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-03 16:40:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-03 16:57:11.000000000","tz":540},"subject":"Remove remaining hard-code of username in each manifest","message":"Remove remaining hard-code of username in each manifest\n\nThis is follow-up of f781eb9e34501b651d3f2069fc8848e7cacef679 and\nreplaces remaining hard-coded user name by the central definition in\nswift::params.\n\nChange-Id: I9cdf102cc517222d64dc09cb5369fc6ba675304c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/640ec3a3eb5860f7a09847058a5ae6f3ea06b9f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/640ec3a3eb5860f7a09847058a5ae6f3ea06b9f5"}]},"branch":"refs/heads/master"},"57d0f6da9ff7fe999d58888c4c7ac33e4c04c407":{"kind":"REWORK","_number":3,"created":"2022-07-03 17:04:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/848543/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/43/848543/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/43/848543/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/43/848543/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/43/848543/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/43/848543/3"}}},"commit":{"parents":[{"commit":"152587fd2a1d0c0bc7ad67585fe2ecdd6abde995","subject":"Remove unnecessary group assignment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/152587fd2a1d0c0bc7ad67585fe2ecdd6abde995"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-03 16:40:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-03 17:04:11.000000000","tz":540},"subject":"Remove remaining hard-code of username in each manifest","message":"Remove remaining hard-code of username in each manifest\n\nThis is follow-up of f781eb9e34501b651d3f2069fc8848e7cacef679 and\nreplaces remaining hard-coded user name by the central definition in\nswift::params.\n\nChange-Id: I9cdf102cc517222d64dc09cb5369fc6ba675304c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/57d0f6da9ff7fe999d58888c4c7ac33e4c04c407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/57d0f6da9ff7fe999d58888c4c7ac33e4c04c407"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
