)]}'
{"id":"openstack%2Fpuppet-swift~849055","triplet_id":"openstack%2Fpuppet-swift~master~I375634214e88acc5075f64090287f934ce8451d1","project":"openstack/puppet-swift","branch":"master","topic":"bug/1986599","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-18 23:42:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I375634214e88acc5075f64090287f934ce8451d1","subject":"Use ini provider to manage storage config files","status":"MERGED","created":"2022-07-08 01:51:27.000000000","updated":"2022-08-18 23:43:56.000000000","submitted":"2022-08-18 23:42:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849055-bug/1986599","meta_rev_id":"d25bfbfd794895c110f0bdb465a8a9c5ce3e7291","_number":849055,"virtual_id_number":849055,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-18 23:42:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"all":[{"value":1,"date":"2022-08-17 01:39:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2022-08-17 01:36:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":2,"date":"2022-08-16 18:23:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-08-18 21:55:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-08-18 21:55:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-08 03:04:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-16 18:23:47.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-08-17 01:36:19.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2022-08-17 01:39:57.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2022-08-18 21:55:39.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"55dab7c1f7f7ec0eb7d3cf3d8384ace1f0369c96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-08 01:51:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35373ec1d50bee3a98f619b326006d752d118baa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 03:04:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/57fbaf0c29fe4ef9b666f8900d2743bb\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b20f99dff6348f0b8ee3bb09423641d : FAILURE in 2m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61070d2c2dc048d88531a99d171689d6 : FAILURE in 2m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a90b7658e4b48d4bce9d930d3ab6c7c : FAILURE in 2m 41s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/08f51d290ffc4837a30d7982b6b967e8 : FAILURE in 14m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddd559402bb749608f7b76add7be0eb0 : FAILURE in 14m 30s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0338b67ccec6463eb20fd1c844dfcbfe : FAILURE in 13m 24s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cfe79a41e8c4fe8ac4c25095b7b0945 : FAILURE in 13m 13s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb0b4521eb6644e78d5d639ef92ce394 : FAILURE in 7m 04s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/76a70459fedf4cbb9cc0470687ed25ca : FAILURE in 10m 44s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99d95ec119614345a2c06b54670e2b8a : FAILURE in 7m 02s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ab004bafcaa46af9aa6dbdda9e2ecc2 : FAILURE in 10m 15s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d0e30769f4e4c7498ec23844f9188e6 : FAILURE in 7m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb299812ff0941f5bbc36d334c49ac40 : FAILURE in 9m 07s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6a9c57e313614a2b9689bdd7a6b2a295 : FAILURE in 1h 05m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b52ee281507a3a684985a861cf2fa3705a0afb1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-08 03:39:38.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"77c486d966dadc96021227d489eca059262d104e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 05:07:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2b17703aca04493a2dbb5d18e9c5255\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/908fe09c3d534eab89cefa0784863706 : FAILURE in 2m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c638bf1c5094d6b8fa9521a4f064746 : FAILURE in 2m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac48208bcaf04d0fbc5a491207c8a9a0 : FAILURE in 2m 47s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/6969b3ee01854e0892569e8b89036104 : FAILURE in 15m 13s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23792ad602d64831858335a8931d157f : FAILURE in 13m 09s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93acce36c0b742e0a92fa3e885ce61fb : FAILURE in 13m 34s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c699cf5271641d0a0d62b251faa92c7 : FAILURE in 12m 50s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/633b35b98aa4400cb32550896f4caa3e : FAILURE in 6m 47s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6afaef16a934c8da7295ea8a7aed77f : FAILURE in 11m 09s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73ec34779e9946ca81506eed855510f3 : FAILURE in 7m 18s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8f904467b2b4d8eacd2994fffd1a978 : FAILURE in 9m 02s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbbd18259481465db6cfa78433237647 : FAILURE in 7m 37s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bd7d7611f8d44d988ec3199539ebab5 : FAILURE in 6m 56s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/73460f2849f24c4f82046b43437bb4d2 : FAILURE in 1h 21m 26s","accounts_in_message":[],"_revision_number":2},{"id":"6553e9cc35a0b1b084ab49ba0c839eeea859cfdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-08 08:09:11.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"81f11c94615ec2f008238b50c1bb8c0179e5b02a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-08 08:14:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ee379efdea916aa34c5346f403cb259e6c96b713","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 09:13:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/939654ea55d449129d0428e4a2ea0166\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcbbc8348f2940b787982940f740cbca : SUCCESS in 3m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecfcd8c2f5004476b8fa230786f2c5fa : FAILURE in 3m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1815f47f6878437fbdfc9533e62b17a4 : FAILURE in 2m 37s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b64f3bfd580748a49814f4794ff9a762 : FAILURE in 12m 18s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36c0dd655ee04569a2dc92a4707fb9a6 : FAILURE in 11m 19s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d19be4c99b541d68781b5764dd6d3af : FAILURE in 11m 36s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d21ab2d6e554044b216997a1e186a78 : FAILURE in 15m 23s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97945fb7b7bf4efca87dcc866e04d326 : FAILURE in 6m 02s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c37f64e56f0d4757ab07e56f23598b58 : FAILURE in 9m 25s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3af43674ecf644b383d91e3b46277df2 : FAILURE in 6m 10s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/20497e94d366457c9e35ef18c91f92c7 : FAILURE in 10m 03s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e73a3c04df4e4ccf8bec55ebcdaad05f : FAILURE in 7m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0987132f7594eb290bc22a66dec41e2 : FAILURE in 6m 08s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3d4fdf4e90ec4bb58c87fc6e60c62943 : FAILURE in 54m 56s","accounts_in_message":[],"_revision_number":4},{"id":"49da489b7df05a1e9b0b13a70a1b67b554d776f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-08 09:34:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d74c008b815a86d4eb1402bda1a5f095881db5d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-08 09:49:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"236f40b891581061221ad3ae0cf5e992fcd9fb84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 10:12:46.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b30f7f152c0e44ee925bdbfa152be43b\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bf8b6f02f9a48e0afa01df5a4484d10 : FAILURE in 2m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f924ff5f899540218df38201aeb3e72c : SUCCESS in 4m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8e4fa125e064bcfabab98a31775a3c3 : SUCCESS in 2m 46s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/945a41b3b7464f018fd4a8ed5bada538 : FAILURE in 16m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3153bffb75a418396851f617d5b0a71 : FAILURE in 14m 49s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c865cefdc9454367a9a42c726dc60cba : FAILURE in 13m 03s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0b6f4cbfadc49a9bcea6d03a4b089ff : FAILURE in 12m 09s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/433b7deb1e4046c9b62e623be353b2f1 : FAILURE in 8m 11s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9fa57e4fa1d34bf1a31b978962193e5e : FAILURE in 11m 43s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88796a7c0fb84509a799bfc4176ac0d8 : FAILURE in 7m 07s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/c589f779af5e41fbac1df53ddac1a351 : FAILURE in 10m 36s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27c2e693312a46e99f3c70b5a0fe8e13 : FAILURE in 7m 10s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ddcec165c624ca18590bdd9a51f56fd : FAILURE in 7m 16s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4017be646f7d4841a21aa87286d12dcc : FAILURE in 20m 48s","accounts_in_message":[],"_revision_number":6},{"id":"034dfe9f5ed5f44806b0ad6a2a78878d956f591d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 15:12:37.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6a3ad25a68f245a62c1cdf1d52220579c929c455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 16:24:33.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7720386c056f46c3b9ab9eeb41f3cf1d\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea608767e0f9470380e1988b94745130 : FAILURE in 3m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8dba3caabd74b1a892b9107e28b94e8 : SUCCESS in 2m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68278cf1a97948f9a1501500301c48d8 : SUCCESS in 3m 11s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5edbf8e08e5427a997aa598a1b5310c : FAILURE in 13m 00s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8be8c10d4136409783b87b10b7ddef5b : FAILURE in 12m 26s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc118219420f4aa09bc7cfc5f5d489d5 : FAILURE in 9m 49s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68e4f298a1b546efa681430b004e6dc9 : FAILURE in 10m 48s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fddffcad8bb40db819aad96e9278afa : FAILURE in 7m 13s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/91e41ccb5ea343a4a686081c146ba968 : FAILURE in 8m 55s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b34636a7791c45e9989397f649351ce9 : FAILURE in 6m 41s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/ecc43fcfad6041e3a9bae7ced67b3fff : FAILURE in 8m 29s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/702a0c022d1541f596005e209e15c644 : FAILURE in 7m 21s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5706c63ed5e1414788be255db8f298fd : FAILURE in 6m 27s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6add930ab8f448e58bd1682e4d5916be : FAILURE in 1h 09m 23s","accounts_in_message":[],"_revision_number":7},{"id":"e5db360df08c32069d8144ccfeb08fbf7917afc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 16:32:38.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f9a38b8498effba8fc262b5bfcbd41eb1ac20711","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 17:27:36.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/822dd09d9a8d49a48a190e2e407537d0\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f05286d076f44d94838c1a9efcb826c9 : FAILURE in 2m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08c2b5ebf10646c8bb8fc6e254295dc9 : SUCCESS in 3m 07s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3e029f2ff4a4d40af323e32ab176de8 : SUCCESS in 2m 30s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/aac09e2d4a064b16af50e2e50f3279c4 : FAILURE in 14m 05s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf7ad1e8bbf5489fab1cb52679aa8146 : FAILURE in 12m 45s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/526a539ba84e48979a7b72a111e52697 : FAILURE in 12m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e8901715bd84b98a0a5f33b0fcb1bc4 : FAILURE in 11m 15s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2da4a29b0de7459e9f426fe1a1ade502 : FAILURE in 36m 04s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/69bac42ca82a4b03aeafa0b417f15417 : FAILURE in 35m 38s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a9591aee6cd4e0999990a41db6366c2 : FAILURE in 15m 26s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4cb07b9759b34769b63d3458326fc6dd : FAILURE in 18m 49s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8dbd699d78d743f3a803ab872ee9fe10 : FAILURE in 19m 02s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32f8cdf3885e4b2a97ff507a2952d7ea : FAILURE in 16m 17s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/80a92343bbf746f4a55fe64fe0a64889 : FAILURE in 52m 40s","accounts_in_message":[],"_revision_number":8},{"id":"205696651b5f43e43e0c85d9e5c6738ecedeb8aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-10 04:27:23.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"b58688f8916ab600f937d9302ed86ccc18cb1afd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-10 05:23:30.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d673b9b5e0e45bfbe078f0984a81466\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c017ded9d184408a6b5ba2ec88455d1 : FAILURE in 2m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/448bc0a095e54400b177ce93760e85ff : SUCCESS in 2m 59s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/414894988f23499283ecbe307e501adb : SUCCESS in 2m 43s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/100ffb0db94847179f6da251109286ec : FAILURE in 12m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5714724b3a784763b01f6e8683a2ed66 : FAILURE in 11m 09s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0288230975641e283925c594ab71579 : FAILURE in 11m 25s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39f51368993c4ce9a2035e604974d272 : FAILURE in 11m 51s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa0c4763fce74dc7a9213a08d10e50eb : FAILURE in 35m 43s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d25effceaf5477694af7511602f6258 : FAILURE in 37m 50s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e49751d700d64994a20c4afded309d25 : FAILURE in 18m 49s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/521bf538df334010bd30e79460a24f4e : FAILURE in 19m 12s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd0aa70e257c441b97e1ba8db284cb4e : FAILURE in 18m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3923fe6026664562803fc23c8d17a9f2 : FAILURE in 15m 40s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3f92aef52ce549738569db16b3671fac : FAILURE in 53m 41s","accounts_in_message":[],"_revision_number":9},{"id":"89ff3de53be612e00a48c2b654f873eca2dc982d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-10 06:55:04.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"782e9f712eb7f9ca1b2640dd8b984595fb478c6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-10 07:20:13.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"7ccd78b97939a3eae00c73d9ca65c8a40c222fd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-10 07:31:21.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"14bacaea0cf01d6c37ac15ce2b4570080ff8aa3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-10 07:47:52.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"258b97e7a959d6e5bb71e89c4c7058c62b3002c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-10 08:03:39.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"b0bb5526f763ecc68fee98a3794643418deb0a69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-10 08:40:02.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"be7f3282984e555d55ba2b7d938e88867c9d95e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-10 08:57:38.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"448466c6a6c2e9c37e9c855285cf71c4b1ae78ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-10 10:14:24.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"2c6bc1803cd32b8abb6ea0d4a70fd57b16187189","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-10 12:02:15.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e92fb45ca3b4bc6847ac4fa01f2cc9f\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7634c37914a4d88afa0f569b457e952 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35a7f7a3c195429c8dbaba924fec6602 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81f3f7181026445dbbb31f79f31cb03e : SUCCESS in 3m 00s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/696b3384c7084fb7851a8fdc76887439 : FAILURE in 13m 47s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d864002e76af4daf9b2fb66908ef6f53 : FAILURE in 13m 57s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/192368d253444926800e5e95d41fc6a3 : FAILURE in 12m 04s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7d9f6fa7d5f4d42a40fd556c8b9ec91 : FAILURE in 11m 57s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f913a11e94b4444e8186317ee7663dd1 : FAILURE in 36m 44s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3fbf20ad11714486a3dee895b8386a0f : FAILURE in 46m 41s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9d45f575c044984ae55de99a016312d : FAILURE in 15m 27s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/47fce114350f4320a5fa5693e8e291b9 : FAILURE in 25m 37s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/355ac70c3698464f9e9fba3fd27ac49f : FAILURE in 16m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/821c8d4f399a498dac4bd6dda413b0d7 : FAILURE in 15m 52s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b1a55dc5323247d3a00acbd4118622c2 : SUCCESS in 1h 42m 19s","accounts_in_message":[],"_revision_number":17},{"id":"7de8cec5dd98f454342a14e78d87b93fa022f0ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-10 12:21:44.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"32384026383782705828eced9ce8b885a7689836","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-10 13:03:45.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"4e4f4b8a1255d3911e1639e19ae722b5e05f3762","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-10 14:00:25.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/62a65eff1b814b2ea3ce6a3bc110db00\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31fd0fc09838463eb2cd5cd330c6af1c : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17ab0b2ddb8349f8a6630c7b09303bda : FAILURE in 4m 57s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36866fe52f864bcd8fd73bfd6b9d288a : FAILURE in 2m 57s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/10b71bbd8dca4fe1bcb46b23f89b1dc8 : FAILURE in 11m 23s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/502c064337624aa49dcb50219313f19e : FAILURE in 11m 37s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a01d59480f134a10b149f29b57a8cec3 : FAILURE in 11m 37s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c289027ec1764e21a2097370a8f71bae : FAILURE in 9m 32s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2eccad73b89438db494717ddedfa282 : FAILURE in 5m 51s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/964d019fc5a149a5a5ac46c35b129b1e : FAILURE in 9m 15s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7bbdeb950bf4609bd5eb5a439945a75 : FAILURE in 6m 24s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b89536dd6b22405a82255037208d0955 : FAILURE in 8m 52s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4dec24083fdb44d0afe865ab74cb6bea : FAILURE in 6m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/533544e1750f45e8aa697e1bae4502d9 : FAILURE in 7m 24s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bc5d74e7d7e14c2dbca336fdf337f90a : FAILURE in 52m 42s","accounts_in_message":[],"_revision_number":19},{"id":"a8e4915c2a2df66918a5870c44f6779367cb590a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-10 14:10:57.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"dd6e2f83bb15d549579f62befce148bc0d8440c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-10 14:35:48.000000000","message":"Patch Set 21: Patch Set 20 was rebased","accounts_in_message":[],"_revision_number":21},{"id":"2460702230354f02e9188586b47db91f5523e428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-10 15:29:08.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e45218ab93614664a17cac775095511c\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65cc2d1f50674d3495fc47d4833fdd49 : SUCCESS in 3m 09s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2df2963af8384e25bce7d0541442e182 : FAILURE in 3m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6044f1c8875442ca97c97a99d2cac196 : FAILURE in 3m 05s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/41689b37c7b24dd1a3e7833e3da1c984 : FAILURE in 13m 52s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55e562761cc6499aa0f25ae0401c662f : FAILURE in 9m 59s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32d04eb186144939b5517e881482f1ff : FAILURE in 13m 21s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aac4a02c6fde465fbf8c4371f55b9425 : FAILURE in 9m 48s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/508baccbdffc4df4becf1c7b0208d342 : FAILURE in 8m 12s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/91a7b59759e042d5b20a8fe6dc31a97c : FAILURE in 17m 06s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ada78b63746741b49da2d4ad28e7be69 : FAILURE in 7m 43s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/00e7db4958fa49f08c20065e1eedf621 : FAILURE in 10m 25s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/032024800d5b4dc7b28982ae134e0895 : FAILURE in 6m 48s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b65ad84e48445faa116264830d643c8 : FAILURE in 8m 13s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8f325aed4b844d0092f72c8d895f80e9 : FAILURE in 51m 07s","accounts_in_message":[],"_revision_number":21},{"id":"e40f66cd26f3becc7679ee1e59bfe7e23234ae44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-10 16:21:40.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"6ebe2d4c78f77e71241f4d6bd54e348805c8bd5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-10 17:46:44.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a54746604fe745d48fd13dabb2461865\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94564f78eeb143a2a5f25c8bf9babd27 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21dd891e16f8435db09c25d4c95666d5 : FAILURE in 2m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72b04687cba645629b319eef7a2ffff3 : FAILURE in 3m 07s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/637f6ed46f294ab58a37a47952d6d61b : FAILURE in 13m 50s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c3a9e372cf54846a271c00dc7504462 : FAILURE in 14m 13s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba2296b758dc4f8cacfbe4bff766c95e : FAILURE in 14m 16s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edebb05800204d9ba9af2a5dd01a7a91 : FAILURE in 13m 18s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d940e85e44e47abb8b93f97aac9af50 : FAILURE in 8m 30s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ddd79ac7ad2642fd8ff94c9a15653401 : FAILURE in 11m 25s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb53530f63fd4523a99b6b6e9fa210bc : FAILURE in 7m 44s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c09b690e15946a298329cbb0e802b3b : FAILURE in 10m 34s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/327f2dd8334d42479c3c8cfe8a436204 : FAILURE in 6m 42s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94915e5cbd4842c1b226e1778f3e75c9 : FAILURE in 7m 40s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0e254dae14c349cebc75ec2a1f91cb0e : FAILURE in 1h 19m 50s","accounts_in_message":[],"_revision_number":22},{"id":"f2e53b6d734d1e760d2354dfd98a04bc4a7a0ac2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-11 03:07:17.000000000","message":"Patch Set 23: Patch Set 22 was rebased","accounts_in_message":[],"_revision_number":23},{"id":"bed02cbdc83fb92fec9adf5486110729482aa530","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-11 03:40:06.000000000","message":"Patch Set 24: Patch Set 23 was rebased","accounts_in_message":[],"_revision_number":24},{"id":"de2063f326f578f784d34a50c7e5d44ad10b98c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 04:55:27.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/408ab8a2c40d4d5eae1a5af2b0f182a9\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b34e3b6ba0a44d9ae1e74ba66128995 : SUCCESS in 2m 47s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/507f4ba6b81d4444aa021dc964c119e1 : FAILURE in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e4fca2a0d1447fcb2ee94434a9562ad : FAILURE in 3m 33s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2cc3c186f81b42f69347adc5c53c5c49 : FAILURE in 11m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82628e67be9f4efeb3d14feb06ff672f : FAILURE in 10m 27s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30ce2a4c842c405a82fa3dc309ea2571 : FAILURE in 10m 20s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47faa276bf364b71b6157ccee7bd8090 : FAILURE in 13m 21s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f8831ff68da47f8b543681491610df6 : FAILURE in 6m 49s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/57226bba39c848b7b4794a35f982d9d7 : FAILURE in 8m 19s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10b2c4812d96456a9d2ac80bf5ac6caf : FAILURE in 7m 08s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/23ce5eac32f842d5bb6f267371eaa36e : FAILURE in 11m 33s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13dd7aa89fde45a1afc85af84661dc7a : FAILURE in 5m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89e9d83127974000aa7e2399a25a0e2e : FAILURE in 6m 13s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/eee111cd18fd4506a1a5a44fc7b22057 : FAILURE in 1h 08m 59s","accounts_in_message":[],"_revision_number":24},{"id":"797b182e25441a6d1f42b1ee4e7b516d7ce15adf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-11 08:23:08.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"957f4dc075ed5c1ddf876ca41d56848376a799ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 09:55:52.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3477ab11e30a411bb7ce219489cc37dd\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62f7f9aaf7c54f8d8802d9db74b59428 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f9f3673dc4749f19960b34450622150 : SUCCESS in 5m 08s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e230a2042e44ae1a24d8b051f52477a : SUCCESS in 4m 02s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e25ea63ba26427c82e9360bd0b72b34 : FAILURE in 19m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/792bc0e2cd58420fa90fe74eb6340335 : FAILURE in 16m 02s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62920f5439f6400ea74a360a67147f6d : FAILURE in 12m 39s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6039f7c5188046f5ad229d8897c854b3 : FAILURE in 15m 08s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95c757ac59014171a28057942f70c840 : FAILURE in 41m 52s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/89c918cef2bf4d1493db6aa3f372bbe1 : SUCCESS in 1h 14m 31s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f82bba740404093ba5831d36043d281 : FAILURE in 16m 26s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/53a526627238454b9882d77a9175e03f : SUCCESS in 32m 13s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/137d6251395a4eb6872bec04f6e17c1b : FAILURE in 19m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc3857b64659463d9c820ee474b9c5d8 : FAILURE in 22m 35s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e8bb9cfda810483a8df7988ecc744a17 : FAILURE in 1h 25m 35s","accounts_in_message":[],"_revision_number":25},{"id":"0f22deabc9faf164f1283920d099b81a4655dc92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-11 10:11:35.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"0893da18f29355487f20ed4be3428722f679541b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-11 10:14:11.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"7823da0737010ddecfa86dc23b7fb8540b462109","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 11:32:50.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75ded007f3144ef5a7ac9a845b5ab6b1\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/280b2f58dd00444a843a782199f7a7de : SUCCESS in 3m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac99e66c7cbd4de6bf3a556ba8825dd9 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bed2c02779440fab6c6277dc58b4a3f : SUCCESS in 3m 17s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/0bae592b9cf84eb8a370fdba491754e5 : FAILURE in 17m 55s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c1f01bae0c740488d456f64cc36aa7c : FAILURE in 13m 47s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed118e7996534a32b35c52132f9a7fbe : FAILURE in 15m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85d7bf63134c43388ed1c1bef52060c2 : FAILURE in 14m 23s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0435aa9e78a2450488f27e15deb04c3a : FAILURE in 38m 45s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c330d95fdf54bb1af39daee5c504e73 : SUCCESS in 1h 11m 21s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30e93c8c1f5e41adb4d2697875df3955 : FAILURE in 19m 33s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/0aef3cf3ae9642799232aed1243d1dc3 : SUCCESS in 32m 05s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65aaec12712a4d84be29f94529a4cf1c : FAILURE in 18m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a25413411a44af6a3cb5be5f2043308 : FAILURE in 18m 34s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0a5a3bd16df24b208e676a628f6fb46f : FAILURE in 1h 17m 00s","accounts_in_message":[],"_revision_number":27},{"id":"bf426bf3cceb2165689695548d013f8da1de2610","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-16 14:24:37.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"2bb87647b7584890423473ab89d0fa5589b52e0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-16 15:49:39.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c55c0409d074eca8a4525c3e3314de9\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d13ffeebfee441aa7a616a1c78bbd56 : SUCCESS in 3m 21s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db3c30326bd748a88daa35147fc7f0d5 : SUCCESS in 3m 27s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa707b79ffbb45e088fecdf0ee88e222 : SUCCESS in 3m 47s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc7b51a06b8e4e23aaffd5f7ee47bdef : FAILURE in 17m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/feab93b9bf554b65b7bdeb51c1f614af : FAILURE in 15m 28s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c0b9a4b7f144b4ca4bb82effc63364e : FAILURE in 15m 20s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9280bffdd20426586c5b0c0c7814c1b : FAILURE in 14m 51s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61d1f5ef3a884578b1626e595427a2f6 : FAILURE in 58m 01s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/68d1bdc11020481b8516e8aabb81c2e7 : SUCCESS in 1h 06m 38s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6685c00287854f36ba032a2a46185d96 : FAILURE in 18m 28s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea9c11dff349425c9e3f73fe218ef6e7 : SUCCESS in 29m 48s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37d7306aa49947f69acd05420721e276 : FAILURE in 18m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/112a65123af247fdb113d296a21a0fe5 : FAILURE in 17m 44s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/53a868d3b0024c84a83ac6b4b1bcbeb7 : FAILURE in 1h 19m 20s","accounts_in_message":[],"_revision_number":28},{"id":"bbff9840267d23247dd3eb4b327a81ab2cebc7dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-16 18:03:20.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"ae67f689560a27e62eb189cbe3a5347564653308","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-16 18:05:21.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"610c91903eb97fe6d41c591b671500665eaab028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-16 18:57:31.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/310098590c484cca95ba952a4948d545\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ced0b67aa4ca43d79472ee0c1f910f74 : SUCCESS in 2m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f9fb3d842b549b2b80d1255e97fa4e6 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a9fd01ddb1a4f079629850dee1d0ba0 : SUCCESS in 2m 59s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/135fbfc67de34fb8b63676d731b234f3 : FAILURE in 13m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7db448d93ce94cf5829497f0dcd2dbd4 : FAILURE in 11m 11s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af6b746ffb1e4e868437ee99d8e60884 : FAILURE in 10m 48s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05b35a35cede46d8a340b7a8f5a441ce : FAILURE in 11m 17s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/686ee4fa63574ed38bfbac479a2c1a3e : FAILURE in 9m 07s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f6e6f57977a400e9e1cae6f7fae7d7b : FAILURE in 12m 42s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d966d998cf4f4a978d767191b5d3ff20 : FAILURE in 6m 55s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/c86dd7d2928842898f20b19531d88ce4 : FAILURE in 11m 57s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2f62123c5a4449fb6ff376dcd7021a9 : FAILURE in 9m 55s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8945828678914b9a97c34435d0111e84 : FAILURE in 8m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13c433be9757440194b0c5d663143b4c : SUCCESS in 2m 54s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3c485be770ac4f0c82f96c7af3732bda : FAILURE in 50m 00s","accounts_in_message":[],"_revision_number":30},{"id":"5c089839270dad69483e15b60015f2129ae0d439","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-17 02:06:25.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"ba85074627a001b84f3bb94478bc844528fcc39b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-17 03:38:24.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9bde87e04f57449fb79928d246cbe3b4\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b74cb2d5f7534312813065331c1882f2 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87b9c8ebdc5846528816933bb85f410c : SUCCESS in 3m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a86135e359b4e0aac93037090e92492 : SUCCESS in 3m 55s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/e23cb600937a4188b894da18f77465e7 : FAILURE in 20m 19s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f108e51fe94b41da9cf14c42e1a7b6aa : FAILURE in 15m 14s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45f1dd4357b540a8ae0e3ac8cec0bf7a : FAILURE in 14m 57s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6dad9a8cc7fc49988edfa7fc6cee0354 : FAILURE in 14m 30s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/719320cba39a465cb0d0ff7dcafbe856 : FAILURE in 1h 08m 19s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/22ef19f4e9c04f1d910dfab74516685c : SUCCESS in 1h 15m 37s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36b03eaac382456bacb16fe2eac5e3c2 : FAILURE in 22m 34s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/4bb6bece0de04de4a99d21aac68c9556 : SUCCESS in 32m 34s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abea59b89b56443eb75a43969e57861c : FAILURE in 18m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7c46e71632f42d69ca8d3e26e9a66fc : FAILURE in 20m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5633593369a476cb80bba2f04c7bcb3 : SUCCESS in 4m 07s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/503676a543bb4f61b27f8425eaf51e4d : FAILURE in 1h 03m 56s","accounts_in_message":[],"_revision_number":31},{"id":"4b67b1a38d8e977c83b80ca7388befd2491fd23d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-17 13:55:16.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"5d2bdfc92796f14a1585e65405b88fb1be8c13ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-17 15:12:23.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d6ceac9e8f094df2933ab80d80417efb\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b25dbc31e2f47f395e730fe6d5067c2 : SUCCESS in 6m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d665ad0005d47edb81a1b59d4272fd0 : SUCCESS in 5m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3eb846906184a858fac87217dc47063 : SUCCESS in 6m 40s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/92c2709f49394df78992d988d3605630 : FAILURE in 24m 05s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4563981f3403438b89787fc2509015e5 : FAILURE in 20m 15s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee95a9db01fa4da7b400635d2814cc92 : FAILURE in 19m 45s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/571643b802d94a70927e4c820d32a547 : FAILURE in 19m 52s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d52318f6f4964cd79f68d4fb2a8edf16 : FAILURE in 42m 32s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2cb3f1f02e846e596f04d8121dde335 : SUCCESS in 1h 08m 15s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef41514178f64db880f69ec0a90f126b : FAILURE in 27m 18s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e1afcff4e4a4ec1a225dd1ba1b3457f : SUCCESS in 36m 00s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba6ac599a2b249e39606c7927a553158 : FAILURE in 26m 12s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e72245ca053a4a25b80d970147749ff5 : FAILURE in 25m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38aefe92b4fe4e28bb52f4acc1aeb1b5 : SUCCESS in 6m 20s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/39c4e129cd154d3cb09004da28105fb8 : FAILURE in 1h 03m 42s","accounts_in_message":[],"_revision_number":32},{"id":"44ac1b351763c65976826f55da0b27ceb88d56ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 14:44:43.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"e944eec6798d2a8db5d39d0166f44cf28f02375f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 15:30:56.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"c6d2e1d4d1e1d1fe47ff5bf5f0d167c42217220e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 16:30:18.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"26bf6086a10f4b8383210719e6bc4991e310e644","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 16:31:39.000000000","message":"Topic set to ini-provider","accounts_in_message":[],"_revision_number":35},{"id":"1564e30f43cb05bbc3da74079435d1a99dc14f4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 16:54:34.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"ea9ea5bd6f4b94343a4e62cab0001871428db627","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 17:32:07.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"68eca409e9db46f07b49040a2ad149571fa7742c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 18:09:05.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"a53deca8768c252dcd6b2e99489eec9c59276429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 20:07:00.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2978fe5e18e4f5aa1304e139458e0a9\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1803fa744a204a92b2b4b657414c9356 : SUCCESS in 4m 03s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd4f93ad862e44c1bba4979cb6eaa473 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28cbc25f206647548de0cd8a4d5aac72 : SUCCESS in 4m 06s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/5343bfc441284db6ad75880bfe453fdb : SUCCESS in 17m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1399c60541e4dcb8b5156247d019db9 : SUCCESS in 13m 31s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e50567728abe4162b5bfc388ef786ba7 : SUCCESS in 14m 31s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/555a9b034d8f4b5987dfba1e3d843394 : SUCCESS in 12m 01s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b82700fdd5a43b99d1ed4ffc5fcf10d : FAILURE in 50m 58s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/87df40cebd614e4dbe5830b133b7f78d : SUCCESS in 1h 13m 07s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b36eceac83e64812b52c0f3107ca7531 : FAILURE in 24m 34s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbe018b7438c4b3d988d99666b141983 : SUCCESS in 31m 08s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/819b9088f5224488ac0c1407d8ce14d3 : FAILURE in 22m 47s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bcf150671b64f7aa5fc634351c4e8d8 : FAILURE in 23m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fc266c29ed04310b4aff59255ae5d32 : SUCCESS in 3m 36s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f9e97b23cfbb4091bee91010595a28e3 : SUCCESS in 1h 51m 27s","accounts_in_message":[],"_revision_number":38},{"id":"942961d1f132bb137e4c931859a76f9612dbcf01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-16 01:55:19.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"b791eba298ea2521abcf87579684d2630ea90930","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-16 01:55:56.000000000","message":"Topic bug/1986599 removed","accounts_in_message":[],"_revision_number":39},{"id":"f516a4227a6f4a7e1e9cf02e68a1d6b736cacfb5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-16 01:56:01.000000000","message":"Topic set to ini-provider","accounts_in_message":[],"_revision_number":39},{"id":"4952ebd914f45f71346dfdf5cdb2da0e3549f821","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-16 02:27:12.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"5532f2c78ef622dbdfcef66a6ca6dfc92532bc85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 04:12:54.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62ba16184c4d405f960d4c8cca65ecee\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/445cce0566ce4251b5232443f1378295 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d27033672194708a2c5670369b91a29 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2f4d458a3dc401bbf8c2e0283f483f4 : SUCCESS in 3m 59s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/1126473216094b19b5ba75f155264d48 : SUCCESS in 20m 05s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/039f4329fbc14b859a69e826af9ed14a : SUCCESS in 14m 12s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74d23938c9dd449297b492510e87cc28 : SUCCESS in 14m 03s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3020ded5abbb4bbd94f0a2cdbd75a03d : SUCCESS in 14m 13s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04bbb40e69124d55a734b64e58710f4b : FAILURE in 54m 41s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/88119e246b96479786b8362f51560aa2 : SUCCESS in 1h 15m 02s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21efcdabdc2441348a0327a7c20b3ec6 : FAILURE in 26m 47s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/b92882d4fe5d444ba9dcb75e4b3f5892 : SUCCESS in 35m 05s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90e02df9074f41309fc5d1a6b068f122 : FAILURE in 24m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6efbba4b7fd34a9a9a8ebc0eead35ca5 : FAILURE in 21m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/781239bd526343f7b0bbd1d8aaa994af : SUCCESS in 3m 17s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/324e412e45574cf99c55308da4994555 : SUCCESS in 1h 39m 07s","accounts_in_message":[],"_revision_number":40},{"id":"e306a032efe0fef4d8451400a5bdbcd5cb04b3fe","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-16 18:23:47.000000000","message":"Patch Set 40: Code-Review+2","accounts_in_message":[],"_revision_number":40},{"id":"62dfdd7e49d3f89197b6b50ac738d886f6e853eb","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2022-08-17 01:36:19.000000000","message":"Patch Set 40: Code-Review+1","accounts_in_message":[],"_revision_number":40},{"id":"8234329431e9e3ab23dd666b692fcfbab3482be6","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-08-17 01:39:57.000000000","message":"Patch Set 40: Code-Review+1","accounts_in_message":[],"_revision_number":40},{"id":"5b8b6968656652947d212fe13fa7554afff07e24","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-18 21:55:39.000000000","message":"Patch Set 40: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":40},{"id":"0ae55a06ecec73f2b11ef769037ad870f55376bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 21:57:01.000000000","message":"Patch Set 40: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":40},{"id":"ecd7c595e21ab44f080dc4ca4fb7fdacc5bc6051","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 23:42:33.000000000","message":"Patch Set 40: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a5e2f6e6f2b4762bc80c2f6045a7735\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5cf61fc65484c02818a76e0fd42122c : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e57fadc58434492eb1bd26854df2bb48 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5c18ccd31af4e0ea8c9bc2c233739c4 : SUCCESS in 3m 28s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/35d336f5d5ef47f187ed3637a1d203a3 : SUCCESS in 17m 43s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6123107e6e544fe2bce48015e8dd926b : SUCCESS in 13m 42s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b61f8504131426b8ae1b69c1999d4e2 : SUCCESS in 13m 10s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c51e329665384d2a880976191a1cfcb0 : SUCCESS in 1h 15m 52s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/e825336d43a244e6a92109ec86843750 : SUCCESS in 31m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/088fa5413c5341f29a48aaccea028e87 : SUCCESS in 4m 11s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/924109eff5394f2e930c53ea63b65b0f : SUCCESS in 1h 41m 44s","accounts_in_message":[],"_revision_number":40},{"id":"4d0d9b20cc2e3592556844aa2bdcf6fe286b2ff0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 23:42:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":40},{"id":"d25bfbfd794895c110f0bdb465a8a9c5ce3e7291","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 23:43:56.000000000","message":"Patch Set 40:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34544fac7b4e44349e06a5977002e3d4\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0641674a15694198a9c055d19c9de18e : SUCCESS in 37s","accounts_in_message":[],"_revision_number":40}],"current_revision_number":40,"current_revision":"453b1a97311dfa9b3fc186e13af037e2ce27b117","revisions":{"d761ef4469819796f649f0c7b05deb447b229e4b":{"kind":"REWORK","_number":1,"created":"2022-07-08 01:51:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/1"}}},"commit":{"parents":[{"commit":"d44df8e3a49236a4bec9c110301fdf123b5463d0","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d44df8e3a49236a4bec9c110301fdf123b5463d0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:51:14.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d761ef4469819796f649f0c7b05deb447b229e4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d761ef4469819796f649f0c7b05deb447b229e4b"}]},"branch":"refs/heads/master"},"d820982cf4a03f99993109cbff15f7e607cced2a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-08 03:39:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/2"}}},"commit":{"parents":[{"commit":"73bfcd95636cdcdfd20089d426b6364800a52e51","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/73bfcd95636cdcdfd20089d426b6364800a52e51"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 03:39:30.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d820982cf4a03f99993109cbff15f7e607cced2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d820982cf4a03f99993109cbff15f7e607cced2a"}]},"branch":"refs/heads/master"},"b716dfff479e4f55fe8e36b5278b6975d90ae769":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-07-08 08:09:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/3"}}},"commit":{"parents":[{"commit":"662d4f7dafbb4d31588305168f2dbb13d4630b41","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/662d4f7dafbb4d31588305168f2dbb13d4630b41"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 08:09:11.000000000","tz":0},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b716dfff479e4f55fe8e36b5278b6975d90ae769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b716dfff479e4f55fe8e36b5278b6975d90ae769"}]},"branch":"refs/heads/master","description":"Rebase"},"774572aebf89c0e50e1969f3caeff76f197f7c15":{"kind":"REWORK","_number":4,"created":"2022-07-08 08:14:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/4"}}},"commit":{"parents":[{"commit":"662d4f7dafbb4d31588305168f2dbb13d4630b41","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/662d4f7dafbb4d31588305168f2dbb13d4630b41"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 08:14:36.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/774572aebf89c0e50e1969f3caeff76f197f7c15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/774572aebf89c0e50e1969f3caeff76f197f7c15"}]},"branch":"refs/heads/master"},"fd62c3df1360cc905ea2e31bd65e595549e92242":{"kind":"REWORK","_number":5,"created":"2022-07-08 09:34:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/5"}}},"commit":{"parents":[{"commit":"662d4f7dafbb4d31588305168f2dbb13d4630b41","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/662d4f7dafbb4d31588305168f2dbb13d4630b41"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 09:34:11.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fd62c3df1360cc905ea2e31bd65e595549e92242"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fd62c3df1360cc905ea2e31bd65e595549e92242"}]},"branch":"refs/heads/master"},"d431a54a8335e2cb63926c45ecc99fe715766169":{"kind":"REWORK","_number":6,"created":"2022-07-08 09:49:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/6"}}},"commit":{"parents":[{"commit":"662d4f7dafbb4d31588305168f2dbb13d4630b41","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/662d4f7dafbb4d31588305168f2dbb13d4630b41"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 09:49:01.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d431a54a8335e2cb63926c45ecc99fe715766169"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d431a54a8335e2cb63926c45ecc99fe715766169"}]},"branch":"refs/heads/master"},"3ebfe52f0d57a300f69d3ec93eab4c0d5418fac7":{"kind":"REWORK","_number":7,"created":"2022-07-09 15:12:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/7"}}},"commit":{"parents":[{"commit":"662d4f7dafbb4d31588305168f2dbb13d4630b41","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/662d4f7dafbb4d31588305168f2dbb13d4630b41"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 15:11:15.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3ebfe52f0d57a300f69d3ec93eab4c0d5418fac7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3ebfe52f0d57a300f69d3ec93eab4c0d5418fac7"}]},"branch":"refs/heads/master"},"85852dc7c4f826a5b3445160b418a52d4ce26b82":{"kind":"REWORK","_number":8,"created":"2022-07-09 16:32:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/8"}}},"commit":{"parents":[{"commit":"662d4f7dafbb4d31588305168f2dbb13d4630b41","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/662d4f7dafbb4d31588305168f2dbb13d4630b41"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 16:32:05.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/85852dc7c4f826a5b3445160b418a52d4ce26b82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/85852dc7c4f826a5b3445160b418a52d4ce26b82"}]},"branch":"refs/heads/master"},"39568e9c85681c8170a656d67962d1fea2be9e65":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2022-07-10 04:27:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/9"}}},"commit":{"parents":[{"commit":"5cff39d694e331a0d591cf2d213993fac6ef2f1c","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5cff39d694e331a0d591cf2d213993fac6ef2f1c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-10 04:27:23.000000000","tz":0},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/39568e9c85681c8170a656d67962d1fea2be9e65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/39568e9c85681c8170a656d67962d1fea2be9e65"}]},"branch":"refs/heads/master","description":"Rebase"},"18b97e0cf742620ec25321ea7cb55f8f608cd100":{"kind":"REWORK","_number":10,"created":"2022-07-10 06:55:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/10"}}},"commit":{"parents":[{"commit":"5cff39d694e331a0d591cf2d213993fac6ef2f1c","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5cff39d694e331a0d591cf2d213993fac6ef2f1c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-10 06:54:46.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/18b97e0cf742620ec25321ea7cb55f8f608cd100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/18b97e0cf742620ec25321ea7cb55f8f608cd100"}]},"branch":"refs/heads/master"},"76cfcc1304deb0c06044d925ca1e7a4b9bc661ba":{"kind":"REWORK","_number":11,"created":"2022-07-10 07:20:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/11"}}},"commit":{"parents":[{"commit":"5cff39d694e331a0d591cf2d213993fac6ef2f1c","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5cff39d694e331a0d591cf2d213993fac6ef2f1c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-10 07:20:02.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/76cfcc1304deb0c06044d925ca1e7a4b9bc661ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/76cfcc1304deb0c06044d925ca1e7a4b9bc661ba"}]},"branch":"refs/heads/master"},"686909c131c645666bb8f9c377c8f5247bb43812":{"kind":"REWORK","_number":12,"created":"2022-07-10 07:31:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/12"}}},"commit":{"parents":[{"commit":"5cff39d694e331a0d591cf2d213993fac6ef2f1c","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5cff39d694e331a0d591cf2d213993fac6ef2f1c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-10 07:30:58.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/686909c131c645666bb8f9c377c8f5247bb43812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/686909c131c645666bb8f9c377c8f5247bb43812"}]},"branch":"refs/heads/master"},"bc7c855128ad0aff0555e5d7555092945ca04d56":{"kind":"REWORK","_number":13,"created":"2022-07-10 07:47:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/13"}}},"commit":{"parents":[{"commit":"5cff39d694e331a0d591cf2d213993fac6ef2f1c","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5cff39d694e331a0d591cf2d213993fac6ef2f1c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-10 07:47:41.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/bc7c855128ad0aff0555e5d7555092945ca04d56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/bc7c855128ad0aff0555e5d7555092945ca04d56"}]},"branch":"refs/heads/master"},"0e4be96538abc18f93d5940047cf64bdc445f52e":{"kind":"REWORK","_number":14,"created":"2022-07-10 08:03:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/14"}}},"commit":{"parents":[{"commit":"5cff39d694e331a0d591cf2d213993fac6ef2f1c","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5cff39d694e331a0d591cf2d213993fac6ef2f1c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-10 08:03:19.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0e4be96538abc18f93d5940047cf64bdc445f52e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0e4be96538abc18f93d5940047cf64bdc445f52e"}]},"branch":"refs/heads/master"},"fa640e93c446ce911fd592c6185f7f483c44dbe7":{"kind":"REWORK","_number":15,"created":"2022-07-10 08:40:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/15"}}},"commit":{"parents":[{"commit":"5cff39d694e331a0d591cf2d213993fac6ef2f1c","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5cff39d694e331a0d591cf2d213993fac6ef2f1c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-10 08:39:52.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fa640e93c446ce911fd592c6185f7f483c44dbe7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fa640e93c446ce911fd592c6185f7f483c44dbe7"}]},"branch":"refs/heads/master"},"abdcb03807663f25d949e23bf32609f62fb5019f":{"kind":"REWORK","_number":16,"created":"2022-07-10 08:57:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/16"}}},"commit":{"parents":[{"commit":"5cff39d694e331a0d591cf2d213993fac6ef2f1c","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5cff39d694e331a0d591cf2d213993fac6ef2f1c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-10 08:57:28.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/abdcb03807663f25d949e23bf32609f62fb5019f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/abdcb03807663f25d949e23bf32609f62fb5019f"}]},"branch":"refs/heads/master"},"d8487ef20e3207476cd68918471b88ef84c94bfa":{"kind":"NO_CHANGE","_number":17,"created":"2022-07-10 10:14:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/17"}}},"commit":{"parents":[{"commit":"1792814498380f731420516b4b46d6f8252ff634","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1792814498380f731420516b4b46d6f8252ff634"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-10 10:14:24.000000000","tz":0},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d8487ef20e3207476cd68918471b88ef84c94bfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d8487ef20e3207476cd68918471b88ef84c94bfa"}]},"branch":"refs/heads/master","description":"Rebase"},"f98deb9cba17fcf5526569258acf95ff623e8b1c":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2022-07-10 12:21:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/18"}}},"commit":{"parents":[{"commit":"ef64ff4ae9cffb6715717a8ec80019f3ef4fbf77","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ef64ff4ae9cffb6715717a8ec80019f3ef4fbf77"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-10 12:21:34.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f98deb9cba17fcf5526569258acf95ff623e8b1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f98deb9cba17fcf5526569258acf95ff623e8b1c"}]},"branch":"refs/heads/master"},"2c3a6256275eb7dbaadf9826e6a57885e2a4460a":{"kind":"REWORK","_number":19,"created":"2022-07-10 13:03:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/19"}}},"commit":{"parents":[{"commit":"e004e4627384f2141e2b1a1178496725117315b8","subject":"Remove deprecated allow_versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e004e4627384f2141e2b1a1178496725117315b8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-10 13:03:09.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2c3a6256275eb7dbaadf9826e6a57885e2a4460a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2c3a6256275eb7dbaadf9826e6a57885e2a4460a"}]},"branch":"refs/heads/master"},"bcb2fb7739558044b2117b584f9cc7b5fa5d4932":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2022-07-10 14:10:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/20"}}},"commit":{"parents":[{"commit":"ef64ff4ae9cffb6715717a8ec80019f3ef4fbf77","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ef64ff4ae9cffb6715717a8ec80019f3ef4fbf77"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-10 14:10:57.000000000","tz":0},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/bcb2fb7739558044b2117b584f9cc7b5fa5d4932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/bcb2fb7739558044b2117b584f9cc7b5fa5d4932"}]},"branch":"refs/heads/master","description":"Rebase"},"010a75a117487de2e07dab2f5a02f11e661028cf":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2022-07-10 14:35:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/21"}}},"commit":{"parents":[{"commit":"874cc367da5815eb38ce40aae632073fb91832b4","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/874cc367da5815eb38ce40aae632073fb91832b4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-10 14:35:48.000000000","tz":0},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/010a75a117487de2e07dab2f5a02f11e661028cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/010a75a117487de2e07dab2f5a02f11e661028cf"}]},"branch":"refs/heads/master","description":"Rebase"},"8a266d9da319a1788f027e7e4f9b868ed57e9b04":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2022-07-10 16:21:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/22"}}},"commit":{"parents":[{"commit":"f5217309de6db6571ed2069707548f3bc60054ac","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f5217309de6db6571ed2069707548f3bc60054ac"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-10 16:21:40.000000000","tz":0},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8a266d9da319a1788f027e7e4f9b868ed57e9b04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8a266d9da319a1788f027e7e4f9b868ed57e9b04"}]},"branch":"refs/heads/master","description":"Rebase"},"bbb3b8b0bef836b6273fc66e2b555b2ebd579216":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2022-07-11 03:07:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/23"}}},"commit":{"parents":[{"commit":"1a254615e9be479dbccce6a0ea6b00b4cc0b5935","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1a254615e9be479dbccce6a0ea6b00b4cc0b5935"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-11 03:07:17.000000000","tz":0},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/bbb3b8b0bef836b6273fc66e2b555b2ebd579216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/bbb3b8b0bef836b6273fc66e2b555b2ebd579216"}]},"branch":"refs/heads/master","description":"Rebase"},"adf6ebe8497b307e98a18182e80a4341a890a380":{"kind":"NO_CHANGE","_number":24,"created":"2022-07-11 03:40:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/24"}}},"commit":{"parents":[{"commit":"cf38ff6ab03098cdaf24306473ea4241062f2dde","subject":"Allow customizing storage config file paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cf38ff6ab03098cdaf24306473ea4241062f2dde"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-11 03:40:06.000000000","tz":0},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/adf6ebe8497b307e98a18182e80a4341a890a380"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/adf6ebe8497b307e98a18182e80a4341a890a380"}]},"branch":"refs/heads/master","description":"Rebase"},"d62d3dc2373e08c231bc4c0a26cfbc93d327738a":{"kind":"REWORK","_number":25,"created":"2022-07-11 08:23:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/25"}}},"commit":{"parents":[{"commit":"e004e4627384f2141e2b1a1178496725117315b8","subject":"Remove deprecated allow_versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e004e4627384f2141e2b1a1178496725117315b8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-11 08:22:53.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d62d3dc2373e08c231bc4c0a26cfbc93d327738a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d62d3dc2373e08c231bc4c0a26cfbc93d327738a"}]},"branch":"refs/heads/master"},"4c49b5b80e455c43f32386876b52b42a8621be2c":{"kind":"REWORK","_number":26,"created":"2022-07-11 10:11:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/26"}}},"commit":{"parents":[{"commit":"e004e4627384f2141e2b1a1178496725117315b8","subject":"Remove deprecated allow_versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e004e4627384f2141e2b1a1178496725117315b8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-11 10:11:25.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4c49b5b80e455c43f32386876b52b42a8621be2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4c49b5b80e455c43f32386876b52b42a8621be2c"}]},"branch":"refs/heads/master"},"3abe8bbb3c87a09b057704ce837e4e53e2569330":{"kind":"REWORK","_number":27,"created":"2022-07-11 10:14:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/27"}}},"commit":{"parents":[{"commit":"e004e4627384f2141e2b1a1178496725117315b8","subject":"Remove deprecated allow_versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e004e4627384f2141e2b1a1178496725117315b8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-11 10:14:02.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3abe8bbb3c87a09b057704ce837e4e53e2569330"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3abe8bbb3c87a09b057704ce837e4e53e2569330"}]},"branch":"refs/heads/master"},"a988b75f0d5962d7d780bdc300b3ddcc0249cc0e":{"kind":"REWORK","_number":28,"created":"2022-07-16 14:24:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/28"}}},"commit":{"parents":[{"commit":"e004e4627384f2141e2b1a1178496725117315b8","subject":"Remove deprecated allow_versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e004e4627384f2141e2b1a1178496725117315b8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-16 14:24:28.000000000","tz":540},"subject":"Use swift_*_config to manage config files","message":"Use swift_*_config to manage config files\n\nStill WIP.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a988b75f0d5962d7d780bdc300b3ddcc0249cc0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a988b75f0d5962d7d780bdc300b3ddcc0249cc0e"}]},"branch":"refs/heads/master"},"4079a49be9c61514796c45a5c02ccc0de2761d88":{"kind":"REWORK","_number":29,"created":"2022-07-16 18:03:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/29"}}},"commit":{"parents":[{"commit":"e004e4627384f2141e2b1a1178496725117315b8","subject":"Remove deprecated allow_versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e004e4627384f2141e2b1a1178496725117315b8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-16 18:02:58.000000000","tz":540},"subject":"Use ini provider to manage storage config files","message":"Use ini provider to manage storage config files\n\nWith this change now config files for storage services(account,\ncontainer and object) are managed by ini provider instead of concat\nresources. This allows us to use the default file provided by packages\ninstead of replacing the whole file content by our templates, and use\nthe $::os_service_default value instead of hard-coding own defaults.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4079a49be9c61514796c45a5c02ccc0de2761d88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4079a49be9c61514796c45a5c02ccc0de2761d88"}]},"branch":"refs/heads/master"},"3dd986e63bad093b660c2b9496bce02053a17c74":{"kind":"REWORK","_number":30,"created":"2022-07-16 18:05:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/30"}}},"commit":{"parents":[{"commit":"e004e4627384f2141e2b1a1178496725117315b8","subject":"Remove deprecated allow_versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e004e4627384f2141e2b1a1178496725117315b8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-16 18:05:12.000000000","tz":540},"subject":"Use ini provider to manage storage config files","message":"Use ini provider to manage storage config files\n\nWith this change now config files for storage services(account,\ncontainer and object) are managed by ini provider instead of concat\nresources. This allows us to use the default file provided by packages\ninstead of replacing the whole file content by our templates, and use\nthe $::os_service_default value instead of hard-coding own defaults.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3dd986e63bad093b660c2b9496bce02053a17c74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3dd986e63bad093b660c2b9496bce02053a17c74"}]},"branch":"refs/heads/master"},"38b476a0b3558d6821bae9422489aa687dda96a5":{"kind":"REWORK","_number":31,"created":"2022-07-17 02:06:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/31"}}},"commit":{"parents":[{"commit":"e004e4627384f2141e2b1a1178496725117315b8","subject":"Remove deprecated allow_versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e004e4627384f2141e2b1a1178496725117315b8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-17 02:06:14.000000000","tz":540},"subject":"Use ini provider to manage storage config files","message":"Use ini provider to manage storage config files\n\nWith this change now config files for storage services(account,\ncontainer and object) are managed by ini provider instead of concat\nresources. This allows us to use the default file provided by packages\ninstead of replacing the whole file content by our templates, and use\nthe $::os_service_default value instead of hard-coding own defaults.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/38b476a0b3558d6821bae9422489aa687dda96a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/38b476a0b3558d6821bae9422489aa687dda96a5"}]},"branch":"refs/heads/master"},"b67793cc01cdae122e5643b2dcb9e3a55617f8ff":{"kind":"REWORK","_number":32,"created":"2022-07-17 13:55:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/32"}}},"commit":{"parents":[{"commit":"e004e4627384f2141e2b1a1178496725117315b8","subject":"Remove deprecated allow_versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e004e4627384f2141e2b1a1178496725117315b8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-17 13:55:06.000000000","tz":540},"subject":"Use ini provider to manage storage config files","message":"Use ini provider to manage storage config files\n\nWith this change now config files for storage services(account,\ncontainer and object) are managed by ini provider instead of concat\nresources. This allows us to use the default file provided by packages\ninstead of replacing the whole file content by our templates, and use\nthe $::os_service_default value instead of hard-coding own defaults.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b67793cc01cdae122e5643b2dcb9e3a55617f8ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b67793cc01cdae122e5643b2dcb9e3a55617f8ff"}]},"branch":"refs/heads/master"},"85b3d77cbda62936601cf649c6d3ca941c599236":{"kind":"REWORK","_number":33,"created":"2022-08-15 14:44:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/33"}}},"commit":{"parents":[{"commit":"e004e4627384f2141e2b1a1178496725117315b8","subject":"Remove deprecated allow_versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e004e4627384f2141e2b1a1178496725117315b8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-15 14:44:33.000000000","tz":540},"subject":"Use ini provider to manage storage config files","message":"Use ini provider to manage storage config files\n\nWith this change now config files for storage services(account,\ncontainer and object) are managed by ini provider instead of concat\nresources. This allows us to use the default file provided by packages\ninstead of replacing the whole file content by our templates, and use\nthe $::os_service_default value instead of hard-coding own defaults.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/85b3d77cbda62936601cf649c6d3ca941c599236"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/85b3d77cbda62936601cf649c6d3ca941c599236"}]},"branch":"refs/heads/master"},"d401e335b0d4e3b9286fb200fad5bd74e943ec1a":{"kind":"REWORK","_number":34,"created":"2022-08-15 15:30:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/34"}}},"commit":{"parents":[{"commit":"e004e4627384f2141e2b1a1178496725117315b8","subject":"Remove deprecated allow_versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e004e4627384f2141e2b1a1178496725117315b8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-15 15:30:44.000000000","tz":540},"subject":"Use ini provider to manage storage config files","message":"Use ini provider to manage storage config files\n\nWith this change now config files for storage services(account,\ncontainer and object) are managed by ini provider instead of concat\nresources. This allows us to use the default file provided by packages\ninstead of replacing the whole file content by our templates, and use\nthe $::os_service_default value instead of hard-coding own defaults.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d401e335b0d4e3b9286fb200fad5bd74e943ec1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d401e335b0d4e3b9286fb200fad5bd74e943ec1a"}]},"branch":"refs/heads/master"},"9b22ab0159da179be700e1ff60ee942b2a1ca111":{"kind":"REWORK","_number":35,"created":"2022-08-15 16:30:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/35"}}},"commit":{"parents":[{"commit":"e004e4627384f2141e2b1a1178496725117315b8","subject":"Remove deprecated allow_versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e004e4627384f2141e2b1a1178496725117315b8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-15 16:30:09.000000000","tz":540},"subject":"Use ini provider to manage storage config files","message":"Use ini provider to manage storage config files\n\nWith this change now config files for storage services(account,\ncontainer and object) are managed by ini provider instead of concat\nresources. This allows us to use the default file provided by packages\ninstead of replacing the whole file content by our templates, and use\nthe $::os_service_default value instead of hard-coding own defaults.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9b22ab0159da179be700e1ff60ee942b2a1ca111"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9b22ab0159da179be700e1ff60ee942b2a1ca111"}]},"branch":"refs/heads/master"},"ec6d69f917e8a3d717f72496dc8fa93ef632b2b5":{"kind":"REWORK","_number":36,"created":"2022-08-15 16:54:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/36"}}},"commit":{"parents":[{"commit":"e004e4627384f2141e2b1a1178496725117315b8","subject":"Remove deprecated allow_versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e004e4627384f2141e2b1a1178496725117315b8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-15 16:54:18.000000000","tz":540},"subject":"Use ini provider to manage storage config files","message":"Use ini provider to manage storage config files\n\nWith this change now config files for storage services(account,\ncontainer and object) are managed by ini provider instead of concat\nresources. This allows us to use the default file provided by packages\ninstead of replacing the whole file content by our templates, and use\nthe $::os_service_default value instead of hard-coding own defaults.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ec6d69f917e8a3d717f72496dc8fa93ef632b2b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ec6d69f917e8a3d717f72496dc8fa93ef632b2b5"}]},"branch":"refs/heads/master"},"1477c3b7b52c50b0ae7a3c24c08e5f4fc7dce3f3":{"kind":"REWORK","_number":37,"created":"2022-08-15 17:32:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/37"}}},"commit":{"parents":[{"commit":"e004e4627384f2141e2b1a1178496725117315b8","subject":"Remove deprecated allow_versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e004e4627384f2141e2b1a1178496725117315b8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-15 17:31:57.000000000","tz":540},"subject":"Use ini provider to manage storage config files","message":"Use ini provider to manage storage config files\n\nWith this change now config files for storage services(account,\ncontainer and object) are managed by ini provider instead of concat\nresources. This allows us to use the default file provided by packages\ninstead of replacing the whole file content by our templates, and use\nthe $::os_service_default value instead of hard-coding own defaults.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1477c3b7b52c50b0ae7a3c24c08e5f4fc7dce3f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1477c3b7b52c50b0ae7a3c24c08e5f4fc7dce3f3"}]},"branch":"refs/heads/master"},"652afd2b879ff6bc6e378c5baa393af53ae6a03b":{"kind":"REWORK","_number":38,"created":"2022-08-15 18:09:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/38"}}},"commit":{"parents":[{"commit":"e004e4627384f2141e2b1a1178496725117315b8","subject":"Remove deprecated allow_versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e004e4627384f2141e2b1a1178496725117315b8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-15 18:08:56.000000000","tz":540},"subject":"Use ini provider to manage storage config files","message":"Use ini provider to manage storage config files\n\nWith this change now config files for storage services(account,\ncontainer and object) are managed by ini provider instead of concat\nresources. This allows us to use the default file provided by packages\ninstead of replacing the whole file content by our templates, and use\nthe $::os_service_default value instead of hard-coding own defaults.\n\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/652afd2b879ff6bc6e378c5baa393af53ae6a03b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/652afd2b879ff6bc6e378c5baa393af53ae6a03b"}]},"branch":"refs/heads/master"},"3900953f39d47271cf766407fff7b549a7705079":{"kind":"REWORK","_number":39,"created":"2022-08-16 01:55:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/39"}}},"commit":{"parents":[{"commit":"040a509b278f5d02d09e8f9a8d094aa58f78f0af","subject":"Support customizing internal-client.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/040a509b278f5d02d09e8f9a8d094aa58f78f0af"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-16 01:54:39.000000000","tz":540},"subject":"Use ini provider to manage storage config files","message":"Use ini provider to manage storage config files\n\nWith this change now config files for storage services(account,\ncontainer and object) are managed by ini provider instead of concat\nresources. This allows us to use the default file provided by packages\ninstead of replacing the whole file content by our templates, and use\nthe $::os_service_default value instead of hard-coding own defaults.\n\nRelated-Bug: #1921319\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3900953f39d47271cf766407fff7b549a7705079"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3900953f39d47271cf766407fff7b549a7705079"}]},"branch":"refs/heads/master"},"453b1a97311dfa9b3fc186e13af037e2ce27b117":{"kind":"REWORK","_number":40,"created":"2022-08-16 02:27:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/849055/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/849055/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/849055/40"}}},"commit":{"parents":[{"commit":"efcef86de7ba859197aecd791b3b390a220a66d8","subject":"Support customizing internal-client.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/efcef86de7ba859197aecd791b3b390a220a66d8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 01:40:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-16 02:27:03.000000000","tz":540},"subject":"Use ini provider to manage storage config files","message":"Use ini provider to manage storage config files\n\nWith this change now config files for storage services(account,\ncontainer and object) are managed by ini provider instead of concat\nresources. This allows us to use the default file provided by packages\ninstead of replacing the whole file content by our templates, and use\nthe $::os_service_default value instead of hard-coding own defaults.\n\nRelated-Bug: #1921319\nChange-Id: I375634214e88acc5075f64090287f934ce8451d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/453b1a97311dfa9b3fc186e13af037e2ce27b117"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/453b1a97311dfa9b3fc186e13af037e2ce27b117"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
