)]}'
{"id":"openstack%2Fpuppet-swift~849063","triplet_id":"openstack%2Fpuppet-swift~master~I4f462027417369d8666bf55c7564cf5df2452995","project":"openstack/puppet-swift","branch":"master","topic":"memcached","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-01 09:46:32.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I4f462027417369d8666bf55c7564cf5df2452995","subject":"Use p-o-i manifest to deploy memcached","status":"MERGED","created":"2022-07-08 04:25:47.000000000","updated":"2022-08-01 11:08:23.000000000","submitted":"2022-08-01 11:08:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849063-memcached","meta_rev_id":"66e28f05ea81602666ceb27f5a8951f83bb50318","_number":849063,"virtual_id_number":849063,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-01 11:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-08-01 09:46:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-08-01 09:46:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-08 04:47:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-11 22:01:52.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-08-01 07:34:58.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"98bfc3e83d80a74945042da875486af8ac98fffc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-08 04:25:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"420e055488d650b8c0eb84932fbfb5129dbc3368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 04:47:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59ef76c7459e4d1fb2ff721ce75db06e\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2898c09470044f90b640d9e06babfaf9 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da759a88bdde4a35847447c536330cde : SUCCESS in 2m 39s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e858dc944f0f4a7fbe111ce7e2908a41 : SUCCESS in 2m 39s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/667735f57eb1414ca8f797a05197a963 : FAILURE in 14m 18s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c3151575cbc4aac856782b8d5e87ca8 : SUCCESS in 19m 59s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1f533709a194b7b8d959b3434ef84b4 : FAILURE in 13m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/119250b6945d4895b8bf97c1985b3073 : FAILURE in 13m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f97ba14cf06f34a095711ba99786f546c50f0e9e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-11 22:01:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a38101db17d415a7a60293b597ef1a7b20b8283f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-01 07:34:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"86e3b0c93bd237f0795988f9a14ce0fd57fb9d8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 09:41:08.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/puppet-swift for 849063,1","accounts_in_message":[],"_revision_number":1},{"id":"cb870e9b5358bb9c49ac1a990adaaf66a7d8d1b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-01 09:45:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4bdb840bdb29ec5a66adb4b74a15f15784142ebf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-01 09:46:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"672a8b2e9e0cec27ba22db63053e8709e4084853","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 10:23:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9080589d38a4985ac733f8670525e8b\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa0cddcd856446c282f6542d071b24b0 : SUCCESS in 3m 49s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/002c5e3007814a8dbd4b6474a5425a4b : SUCCESS in 3m 40s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2744ff5b52ae4ffcbaba3970a4b770b3 : SUCCESS in 3m 45s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46a3c3240ed04dcea5efc7b88654d436 : FAILURE in 20m 40s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e9039a44c7041ce924db57f48afa8a4 : SUCCESS in 28m 27s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9ec0cb41772452e89f6f59706657d0e : FAILURE in 21m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a2ab141760e486ea64edd9a6e1a4851 : FAILURE in 20m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d0725172cd16b68a18e33314cd9aa9bcba290242","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 10:24:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0b3d9f9db2bc34d46f6a9feb68506388066879f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 11:08:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41a185092ac6477bad29dd3746e20874\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2aeebd0702444bd58f5dee31a9d77003 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0b54f764782490d8c8c788cbc7a6601 : SUCCESS in 3m 27s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62cf41ed67154c74830114605e8bdf3f : SUCCESS in 3m 39s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b574666b3ed47d1a06bc4075be16d87 : SUCCESS in 34m 25s","accounts_in_message":[],"_revision_number":2},{"id":"66e28f05ea81602666ceb27f5a8951f83bb50318","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 11:08:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0439bd22d6d0aeeeb2092d2056162e8c89cf5c4b","revisions":{"79f299761390cd741f419496f252c930727fe122":{"kind":"REWORK","_number":1,"created":"2022-07-08 04:25:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/849063/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/63/849063/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/63/849063/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/63/849063/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/63/849063/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/63/849063/1"}}},"commit":{"parents":[{"commit":"e004e4627384f2141e2b1a1178496725117315b8","subject":"Remove deprecated allow_versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e004e4627384f2141e2b1a1178496725117315b8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 04:25:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 04:25:05.000000000","tz":540},"subject":"Use p-o-i manifest to deploy memcached","message":"Use p-o-i manifest to deploy memcached\n\n... instead of own implementation locally managed in this repo.\n\nChange-Id: I4f462027417369d8666bf55c7564cf5df2452995\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/79f299761390cd741f419496f252c930727fe122"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/79f299761390cd741f419496f252c930727fe122"}]},"branch":"refs/heads/master"},"0439bd22d6d0aeeeb2092d2056162e8c89cf5c4b":{"kind":"REWORK","_number":2,"created":"2022-08-01 09:45:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/849063/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/63/849063/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/63/849063/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/63/849063/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/63/849063/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/63/849063/2"}}},"commit":{"parents":[{"commit":"0adeebce9432f8d4fd15328bdf0476191bfbcde0","subject":"Merge \"Rename acceptance spec files to explain order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0adeebce9432f8d4fd15328bdf0476191bfbcde0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-08 04:25:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-01 09:45:07.000000000","tz":540},"subject":"Use p-o-i manifest to deploy memcached","message":"Use p-o-i manifest to deploy memcached\n\n... instead of own implementation locally managed in this repo.\n\nChange-Id: I4f462027417369d8666bf55c7564cf5df2452995\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0439bd22d6d0aeeeb2092d2056162e8c89cf5c4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0439bd22d6d0aeeeb2092d2056162e8c89cf5c4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
