)]}'
{"id":"openstack%2Fpuppet-swift~849203","triplet_id":"openstack%2Fpuppet-swift~master~Ic10cd3632e1b52945195cfefe904257e66958eba","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-23 18:13:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic10cd3632e1b52945195cfefe904257e66958eba","subject":"Do not define service resource when service management is disabled","status":"MERGED","created":"2022-07-09 15:19:08.000000000","updated":"2023-03-23 18:13:40.000000000","submitted":"2023-03-23 18:13:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849203","meta_rev_id":"d4d2873fe826b7029b717d4db4b487dff1809048","_number":849203,"virtual_id_number":849203,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-23 18:13:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-03-22 10:37:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-03-23 16:49:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-03-23 16:49:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-09 16:54:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-22 10:37:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-03-23 16:49:58.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"0453c6d001b2b2ed8a03817af0898f8fac62f02b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 15:19:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"800843dabafecef08879d5582bde7820f3b3a506","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 16:54:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eccc89f870f040a49e2ef8f4306627ef\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60f8c4e50d8d43cd8ba60ef4ddd32af2 : SUCCESS in 3m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76048164852143f8a60cf32fa6a1969e : SUCCESS in 2m 59s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5b61a8872c14a7ba554e0bbcc2e0080 : SUCCESS in 2m 40s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f907708b7df4b7dbea10bd9d3887aee : FAILURE in 13m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c4cf92853004ee2ab4aa65153424e56 : FAILURE in 12m 29s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8737b2fce0a14d5da2abc66bb33004dd : FAILURE in 12m 26s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f29425efe1a248b3affbceb838c1b6bb : FAILURE in 12m 09s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cf4677933ed405392866f32af243c7e : FAILURE in 37m 09s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/71b6768f1810499a9d8b95d1aa168c2e : SUCCESS in 1h 05m 03s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/120059286ff64922b378632d128364cb : FAILURE in 18m 12s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/76f4b6ef0be341cb8755809f222602ed : SUCCESS in 37m 51s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48a4ef58df3642959868a4fe1daf52e5 : FAILURE in 18m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c2c35d47a874fa29dc9c81398dc11f8 : FAILURE in 19m 00s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2eb63305291e4ab7a04757e628c611e4 : SUCCESS in 1h 32m 51s","accounts_in_message":[],"_revision_number":1},{"id":"2748fd55dd53bb94fbcc7eb41ddf77eef975237b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-22 01:56:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0dcfdc16060f447d059cc63078b4d2c339a4baa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 03:15:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e42b1fed75a54ca7ad5e6f434572fe6f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/463db616cc984b2e8941999f1e1f8d5f : SUCCESS in 2m 36s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f00be90dc8fd4fc39bfef4b3e5fa800c : SUCCESS in 2m 15s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/df5f28e2bef140cf88a72e40027debc5 : FAILURE in 13m 12s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74bb3d07aa80482582c0acfd22882450 : FAILURE in 7m 06s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24b0498e9392494fac2983587b08c2bb : FAILURE in 6m 52s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53d71c89a3d64748ae258d7459c3b1b6 : SUCCESS in 1h 04m 42s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/37111b548a55408b83d6be88e1f18362 : SUCCESS in 1h 18m 10s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/314540d47a8c4d438ea393b2520eeee9 : SUCCESS in 30m 29s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eabd018775f44ef8b14f9b11b87502fc : SUCCESS in 19m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6f3bdb8d70748cabd6af3bbc379ecd1 : SUCCESS in 19m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"817a7fc0e94f348b3cb0cc89e0ea419e3c04c30a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-22 07:15:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2c5526163a86f57c45c67c31ad10e102e678c66d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-22 07:17:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4dd519c720336da7155ebeae9a331b396119a82e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-22 07:18:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"03f5725d91889f8aa27a5f583dd179f2052b8767","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-22 07:20:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"182512ff686d61c43e03621b9f9a4f899d91aea3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 08:48:22.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/43e00712057c4c2da79dcace0b956cae\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f4d2b5690244264a733ec2d92e32971 : SUCCESS in 2m 41s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff2085a6280a44a8ad5793c901bc7ef9 : SUCCESS in 2m 20s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/987241d46d104bd0931230496026ce6f : FAILURE in 12m 31s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f86d389760a4c1b944394a5b897c31d : FAILURE in 11m 53s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74ee93a9d97548d68a36119c3eda47b6 : FAILURE in 10m 05s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae70c2879bb146779bf937f9160d0790 : SUCCESS in 59m 46s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/93805cfe0ab9465ba27b752575f8c50e : SUCCESS in 1h 20m 25s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/30e2ec221add4cc48cfd764002d548fe : SUCCESS in 29m 15s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4de6f8eedae41658e83956a1daa8e8d : SUCCESS in 15m 49s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/346b349447614424abb6d79d612303eb : SUCCESS in 19m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"37cf514ca25ae67f24f220f99fc9e27cc546e247","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-22 08:58:17.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b9cc6a05080c4398bb958291559b005087946641","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 10:25:14.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9f75e13357343eab799dfcd6b6394c2\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/259aa9043329424aa3b62620b2557ee4 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/024264b0f3b944d98d24d2a0926e3ba7 : SUCCESS in 3m 22s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce3bf4b8b12f45e1a977f3f849cf4d9e : SUCCESS in 14m 09s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75e94491d0f049c595f6bd670b1c364e : SUCCESS in 10m 27s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c1ee5abb3914ca4b0886fbf046f416e : SUCCESS in 10m 34s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50fb33689dcf40c0bce7569929ccec46 : SUCCESS in 1h 05m 13s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/656e6cc7310a4dd4837ab5bd9fb5a47f : SUCCESS in 1h 18m 38s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/fef2f2f01a6b447991dc3844b5f806a3 : SUCCESS in 35m 53s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea25ab3b0ba849508a641d7e51ed8ecb : SUCCESS in 18m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f955991dd41486f93059cb06b58fae7 : SUCCESS in 18m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a02dfa26d2c0a7f704c0c34bb30e997d6386f4e6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-22 10:37:23.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"00b307f8739e70c77ee6e974fa97aa7d2c9c1295","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-03-23 16:49:58.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"424a0d3ce028cec74d33cbed57bd9f3f72e22169","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 16:50:52.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"8f0abed44b630342136292ffa7bd919030b72987","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 18:13:40.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a33be28ebb5b44148ea9a3d8af4cf088\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/454fca48985b4f0d9f9eb78c9bcb830e : SUCCESS in 3m 04s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23cc5336bd4b47249f9749d8f1abcbb9 : SUCCESS in 3m 00s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/866a0ffe2d9845439d8355f9458227c6 : SUCCESS in 13m 26s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7367884d149433f808897f9eabd4a2c : SUCCESS in 10m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce9ffc725ad54f51b5854fe760de6ef8 : SUCCESS in 1h 16m 30s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/47505d8953664479a99b24d2fc859e02 : SUCCESS in 28m 47s","accounts_in_message":[],"_revision_number":7},{"id":"d4d2873fe826b7029b717d4db4b487dff1809048","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 18:13:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"d1cc6938da211c3488101d90b7c146f15e11d620","revisions":{"223d447f03dcf749f42b0dcd842358fbab4777ae":{"kind":"REWORK","_number":1,"created":"2022-07-09 15:19:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/849203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/03/849203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/1"}}},"commit":{"parents":[{"commit":"e004e4627384f2141e2b1a1178496725117315b8","subject":"Remove deprecated allow_versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e004e4627384f2141e2b1a1178496725117315b8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 15:17:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 15:17:58.000000000","tz":540},"subject":"Do not define service resource when service management is disabled","message":"Do not define service resource when service management is disabled\n\nChange-Id: Ic10cd3632e1b52945195cfefe904257e66958eba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/223d447f03dcf749f42b0dcd842358fbab4777ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/223d447f03dcf749f42b0dcd842358fbab4777ae"}]},"branch":"refs/heads/master"},"c0ee047b3f02436c093c4a8c3e493b13fe7761ff":{"kind":"REWORK","_number":2,"created":"2023-03-22 01:56:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/849203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/03/849203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/2"}}},"commit":{"parents":[{"commit":"9b05f2ecd75db5738a363df995618cb78569c779","subject":"Remove TripleO job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9b05f2ecd75db5738a363df995618cb78569c779"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 15:17:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-22 01:56:41.000000000","tz":540},"subject":"Do not define service resource when service management is disabled","message":"Do not define service resource when service management is disabled\n\nChange-Id: Ic10cd3632e1b52945195cfefe904257e66958eba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c0ee047b3f02436c093c4a8c3e493b13fe7761ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c0ee047b3f02436c093c4a8c3e493b13fe7761ff"}]},"branch":"refs/heads/master"},"5b9acbff3491b32611fa19a6d4a6fd755360c6aa":{"kind":"REWORK","_number":3,"created":"2023-03-22 07:15:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/849203/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/03/849203/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/3"}}},"commit":{"parents":[{"commit":"9b05f2ecd75db5738a363df995618cb78569c779","subject":"Remove TripleO job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9b05f2ecd75db5738a363df995618cb78569c779"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 15:17:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-22 07:15:06.000000000","tz":540},"subject":"Do not define service resource when service management is disabled","message":"Do not define service resource when service management is disabled\n\nChange-Id: Ic10cd3632e1b52945195cfefe904257e66958eba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5b9acbff3491b32611fa19a6d4a6fd755360c6aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5b9acbff3491b32611fa19a6d4a6fd755360c6aa"}]},"branch":"refs/heads/master"},"1e3da182aa0990d10ac490f02b64a45ee97f1872":{"kind":"REWORK","_number":4,"created":"2023-03-22 07:17:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/849203/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/03/849203/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/4"}}},"commit":{"parents":[{"commit":"9b05f2ecd75db5738a363df995618cb78569c779","subject":"Remove TripleO job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9b05f2ecd75db5738a363df995618cb78569c779"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 15:17:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-22 07:17:23.000000000","tz":540},"subject":"Do not define service resource when service management is disabled","message":"Do not define service resource when service management is disabled\n\nChange-Id: Ic10cd3632e1b52945195cfefe904257e66958eba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1e3da182aa0990d10ac490f02b64a45ee97f1872"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1e3da182aa0990d10ac490f02b64a45ee97f1872"}]},"branch":"refs/heads/master"},"bf35ab32cefe44451b5db3396920a56de381f076":{"kind":"REWORK","_number":5,"created":"2023-03-22 07:18:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/849203/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/03/849203/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/5"}}},"commit":{"parents":[{"commit":"9b05f2ecd75db5738a363df995618cb78569c779","subject":"Remove TripleO job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9b05f2ecd75db5738a363df995618cb78569c779"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 15:17:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-22 07:18:21.000000000","tz":540},"subject":"Do not define service resource when service management is disabled","message":"Do not define service resource when service management is disabled\n\nChange-Id: Ic10cd3632e1b52945195cfefe904257e66958eba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/bf35ab32cefe44451b5db3396920a56de381f076"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/bf35ab32cefe44451b5db3396920a56de381f076"}]},"branch":"refs/heads/master"},"84024d2f4e4500ec1906281a1071c3961e79e03d":{"kind":"REWORK","_number":6,"created":"2023-03-22 07:20:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/849203/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/03/849203/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/6"}}},"commit":{"parents":[{"commit":"9b05f2ecd75db5738a363df995618cb78569c779","subject":"Remove TripleO job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9b05f2ecd75db5738a363df995618cb78569c779"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 15:17:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-22 07:20:17.000000000","tz":540},"subject":"Do not define service resource when service management is disabled","message":"Do not define service resource when service management is disabled\n\nChange-Id: Ic10cd3632e1b52945195cfefe904257e66958eba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/84024d2f4e4500ec1906281a1071c3961e79e03d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/84024d2f4e4500ec1906281a1071c3961e79e03d"}]},"branch":"refs/heads/master"},"d1cc6938da211c3488101d90b7c146f15e11d620":{"kind":"REWORK","_number":7,"created":"2023-03-22 08:58:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/849203/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/03/849203/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/03/849203/7"}}},"commit":{"parents":[{"commit":"9b05f2ecd75db5738a363df995618cb78569c779","subject":"Remove TripleO job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9b05f2ecd75db5738a363df995618cb78569c779"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 15:17:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-22 08:58:09.000000000","tz":540},"subject":"Do not define service resource when service management is disabled","message":"Do not define service resource when service management is disabled\n\nChange-Id: Ic10cd3632e1b52945195cfefe904257e66958eba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d1cc6938da211c3488101d90b7c146f15e11d620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d1cc6938da211c3488101d90b7c146f15e11d620"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
