)]}'
{"id":"openstack%2Fpuppet-swift~850122","triplet_id":"openstack%2Fpuppet-swift~master~I200e3a4ff323a07778ec6c4d7dd96824c72a12cf","project":"openstack/puppet-swift","branch":"master","topic":"bug/1986599","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-18 23:13:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I200e3a4ff323a07778ec6c4d7dd96824c72a12cf","subject":"Support customizing internal-client.conf","status":"MERGED","created":"2022-07-18 02:47:26.000000000","updated":"2022-08-18 23:14:35.000000000","submitted":"2022-08-18 23:13:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850122-bug/1986599","meta_rev_id":"0297148a0ae46f9e13acf179c6eb77cbc78c9f62","_number":850122,"virtual_id_number":850122,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-18 23:13:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"all":[{"value":1,"date":"2022-08-17 01:39:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2022-08-17 01:37:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-08-16 18:23:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-08-18 21:53:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-08-18 21:53:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-18 04:55:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-16 18:23:34.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-08-17 01:37:12.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2022-08-17 01:39:45.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2022-08-18 21:53:09.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"516b7258e7f483b97c4b1691b0ac648c37416404","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-18 02:47:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ab6aa74ceec9f7519bbf41cc3502523607bc3e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 04:55:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e33d5f9639ed40368096837934c9f6a7\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c97142b52334e56bd6f459e5bba1718 : FAILURE in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55e6adc5ee7a4cc99499486266bec640 : SUCCESS in 3m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cda503a2d4b439b8126b4963dc172ea : SUCCESS in 4m 17s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/86a386973ac6432b9486692c0c7174e1 : FAILURE in 15m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5eaacb6420db4812a3c9f9ed71283a2f : FAILURE in 18m 51s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/672e75e0301f46378d57f623875a2654 : FAILURE in 18m 08s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9fa13ce3f174409ad4e8f75f3fe7a51 : FAILURE in 11m 06s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34dc9579dba347b6b53e17031845faa3 : FAILURE in 43m 01s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f44fd4c8faf496ba1a6d3681f340aa1 : SUCCESS in 59m 17s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe8722b6e9a445c3a21c6d006139063d : FAILURE in 21m 16s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/227407b1ec6f45eb9547a7c29b440d57 : SUCCESS in 26m 59s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ea0d8d1c629453bae71011850ef63f6 : FAILURE in 23m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52f6bfeda1c24875bd70b5cc5cc67bdb : FAILURE in 29m 03s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7b29d4c829854c4abe53448f06a47e81 : SUCCESS in 1h 34m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a55d2eebad22c37ef7d6a8cb81766b5c5c34da73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 14:57:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"01925d413ab24cc289a9c58ba1e367db67f9b046","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 15:19:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a663256b0bd44f2608e1b47537d4690c4b7fc30a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 15:42:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b30db5192c82c5f68e3e6da472ec0d1a73903e96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 16:25:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5e19ef6cab73d2340cc3b051a387668c63dc4fe5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 16:51:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"07b97d3492b09f1179a4024ae0fdec9cd341baaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 17:30:47.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b00a9eeb24d643122f1bf8544b00b63fa213c3bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 18:07:37.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"12bdfac722f95a9056be42e2fd7f007c92fd6ae0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 19:51:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9abbf89772814a6b9fbced31b48dddb2\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87d04fc466c64c4988ad0ac5d1c23052 : SUCCESS in 4m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29b6af13926f4add807fa270af08754b : SUCCESS in 3m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcf37b5162fa443cb66c5aa17641d481 : SUCCESS in 3m 49s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/075fde3b2d7041a08e31623c493eaa8f : SUCCESS in 19m 43s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afc7be3a9e6247279396bb2f7a79fd3e : SUCCESS in 15m 57s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ded9fbb7eae47628b3d3035eb6d2cae : SUCCESS in 15m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85e6313e4e4f4ac78eb2832954803d49 : SUCCESS in 16m 04s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58d2a94b93ce44848d66699b7695e524 : FAILURE in 54m 15s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8241eb4f29154df9b5e177c4c856f37b : SUCCESS in 1h 13m 51s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3508a68720d041eeadda99ad9287eb0c : FAILURE in 23m 28s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6c47e98aba649dd939579393ac5d024 : SUCCESS in 31m 01s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e6c934f370e459b9f7c0b8db2989fff : FAILURE in 22m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33b6678699a64c1e9b2d894b54115410 : FAILURE in 22m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58e35c65b68a4f45ae52bc0fd209aef8 : SUCCESS in 3m 43s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5e93e776c7ff456890c6a9b2e4347d8b : SUCCESS in 1h 39m 01s","accounts_in_message":[],"_revision_number":8},{"id":"4771ab0afa9b57ef459e18a0ba1711b8d5c19726","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 23:11:28.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"f178142c1b12397209b27b9f231da72ff22e5a03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-16 00:17:49.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2b5f8f8ba6bea6b6e624a0c1f5fa09a07262264f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 01:47:11.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98b2843147654c27b0d6a072bc92525b\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3b8f41005e64034a408d7e3d8e13e2a : SUCCESS in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b655b1fdf55418bb2c5fd42ac50857f : SUCCESS in 3m 15s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4eb50e96242e4daaad9b76c6e5d56154 : SUCCESS in 3m 28s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c1281c8263846cfb0435fa93e8387eb : SUCCESS in 17m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a1dd62a7ac2455d967c71a7387e2e64 : SUCCESS in 14m 30s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46821b68c6cb429fb5f1d2f5a15f4027 : SUCCESS in 13m 37s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f193a41bba9f4de2a5864ecb9aeb16d6 : SUCCESS in 13m 03s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9075e092f1947ba832db7537d9c53a3 : FAILURE in 42m 53s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/19c92bd0060a479e857e2311c80721e6 : SUCCESS in 1h 02m 54s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/019c13e5dec44d1aafa7655898fb53d2 : FAILURE in 21m 06s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/052ff5db93d24652964adee7a3ee33c3 : SUCCESS in 28m 55s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29fc7d8219bf45e983a9108a5b458296 : FAILURE in 21m 34s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc15c0ca087643198da8385bb5a35f22 : FAILURE in 21m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/915e4964803845449b1a61e30bc8175a : SUCCESS in 3m 43s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6d051e1abee4460a8e932f499e159fea : SUCCESS in 1h 20m 48s","accounts_in_message":[],"_revision_number":9},{"id":"087f450e5bf3d76a42ff14bd728d55ebfc001b13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-16 01:50:06.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a824af634094147bdbd1c27129a702ea8e087f20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-16 01:51:09.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"e5013de5122b6a9e10a85c4a3f5a8e5f416d5c52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-16 01:53:07.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"d1a2a77d455d5fd94d3428ce746165508ec81518","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-16 02:27:12.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"4c8dd610a0dc69e4686c167fb19388320cae0497","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 03:48:47.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/637998e3cf504082a3a9ce863fd1e7f7\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15de37323a8d4ea290b61b3f721adddf : SUCCESS in 5m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4a0d918bd2247ff8bd163d5439a033a : SUCCESS in 3m 53s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98067b3f77044130b405382a6f851cd5 : SUCCESS in 3m 46s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/de530f5135234eeda61d0ad859aa5af8 : SUCCESS in 25m 16s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b0e187823444616a18d94fe18f8595b : SUCCESS in 16m 36s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdf855a710e544498d801a468dc9f80a : SUCCESS in 16m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5ca1e585f364ff791b63f9d428277b4 : SUCCESS in 16m 42s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd43615d231447699f5562d2a96eac31 : FAILURE in 52m 23s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/71bd3751b9e44cb58122b047553cf94c : SUCCESS in 1h 11m 37s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe83745b47aa4189b201f3633141c9fc : FAILURE in 26m 49s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7fa2820e98e46f49918b00be7cf9257 : SUCCESS in 26m 01s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06b8d6d8c517420797baa18f24f4871f : FAILURE in 20m 24s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8e938cef4274fe98cf8596fbfebc3ce : FAILURE in 21m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e920f35ed6e54b09ace57b6e8a951e27 : SUCCESS in 4m 39s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/93f9b3195acd47298192041fb9a3b5f2 : SUCCESS in 1h 11m 33s","accounts_in_message":[],"_revision_number":13},{"id":"e5bb8617848b5ca42bf906c9e5752505537ad88d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-16 18:23:34.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"81344f72f66787e82922ba36f422a86df4f7fde3","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2022-08-17 01:37:12.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"885a6fef4d5242b4ad66a12d7d809cf0380eb935","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-08-17 01:39:45.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"35c84973dd89b297b9af6e7dbc994140406c068e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-18 21:53:09.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"b656872c6d12d7ee81c59bdbf361c9edb46fe4b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 21:54:29.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"09fe6ca103e88cadc7dd1e38bb95eb944364fd20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 23:13:34.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fbb2ce0333d4db8a5dd269348abd3e3\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddd1baf3ce32406a91acf87bc842365e : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/febfa037490f43fcafe26c27ab17d2d9 : SUCCESS in 3m 30s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d7b8da4996047f08c5863b015c57eb4 : SUCCESS in 3m 36s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/acba9ac86a3f43499c166b3bf4449767 : SUCCESS in 17m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e14cc547212e4b7782c63779ef9cca09 : SUCCESS in 16m 50s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/353b409e3be044209c446866c3452de3 : SUCCESS in 15m 43s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/29bfa00f56b94ce79f9a0b3443289d14 : SUCCESS in 1h 15m 01s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/6581f64d4a654115ba15a359bb4d3661 : SUCCESS in 29m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/639ae8a28de44192a78ef49d15a61509 : SUCCESS in 4m 01s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/34e0a17633b246edbb92b13ab478a652 : SUCCESS in 1h 15m 14s","accounts_in_message":[],"_revision_number":13},{"id":"53e8b39c73078c469335128f8a07f79a4dbcbae9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 23:13:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"0297148a0ae46f9e13acf179c6eb77cbc78c9f62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 23:14:35.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1900b41d441a49589f00afdf7a7d899b\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c5e4d75f84c46f38d01b9d54545c1fb : SUCCESS in 43s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"efcef86de7ba859197aecd791b3b390a220a66d8","revisions":{"3ff9790728852d8db5dcaa1979b538fe16eed06a":{"kind":"REWORK","_number":1,"created":"2022-07-18 02:47:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/850122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/22/850122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/1"}}},"commit":{"parents":[{"commit":"68c875beb1056038fb1bc0df61c9e7812cf758b0","subject":"Fix wrong test description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/68c875beb1056038fb1bc0df61c9e7812cf758b0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-18 02:47:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-18 02:47:09.000000000","tz":540},"subject":"wip: Support customizing internal-client.conf","message":"wip: Support customizing internal-client.conf\n\nChange-Id: I200e3a4ff323a07778ec6c4d7dd96824c72a12cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3ff9790728852d8db5dcaa1979b538fe16eed06a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3ff9790728852d8db5dcaa1979b538fe16eed06a"}]},"branch":"refs/heads/master"},"d6a84ea7ad0dfeb900f53ece24442be005cdc887":{"kind":"REWORK","_number":2,"created":"2022-08-15 14:57:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/850122/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/22/850122/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/2"}}},"commit":{"parents":[{"commit":"68c875beb1056038fb1bc0df61c9e7812cf758b0","subject":"Fix wrong test description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/68c875beb1056038fb1bc0df61c9e7812cf758b0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-18 02:47:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-15 14:55:07.000000000","tz":540},"subject":"Support customizing internal-client.conf","message":"Support customizing internal-client.conf\n\nThis change introduces capability to customize internal-client.conf,\nwhich is used by several daemon processes like container-sharder.\n\nChange-Id: I200e3a4ff323a07778ec6c4d7dd96824c72a12cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d6a84ea7ad0dfeb900f53ece24442be005cdc887"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d6a84ea7ad0dfeb900f53ece24442be005cdc887"}]},"branch":"refs/heads/master"},"08a262356abd50ed5e071a6c4671dc3031d02658":{"kind":"REWORK","_number":3,"created":"2022-08-15 15:19:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/850122/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/22/850122/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/3"}}},"commit":{"parents":[{"commit":"68c875beb1056038fb1bc0df61c9e7812cf758b0","subject":"Fix wrong test description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/68c875beb1056038fb1bc0df61c9e7812cf758b0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-18 02:47:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-15 15:19:13.000000000","tz":540},"subject":"Support customizing internal-client.conf","message":"Support customizing internal-client.conf\n\nThis change introduces capability to customize internal-client.conf,\nwhich is used by several daemon processes like container-sharder.\n\nChange-Id: I200e3a4ff323a07778ec6c4d7dd96824c72a12cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/08a262356abd50ed5e071a6c4671dc3031d02658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/08a262356abd50ed5e071a6c4671dc3031d02658"}]},"branch":"refs/heads/master"},"fbbfa6fe601f90c64133f61e99f317f4c8cb64dd":{"kind":"REWORK","_number":4,"created":"2022-08-15 15:42:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/850122/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/22/850122/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/4"}}},"commit":{"parents":[{"commit":"68c875beb1056038fb1bc0df61c9e7812cf758b0","subject":"Fix wrong test description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/68c875beb1056038fb1bc0df61c9e7812cf758b0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-18 02:47:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-15 15:42:05.000000000","tz":540},"subject":"Support customizing internal-client.conf","message":"Support customizing internal-client.conf\n\nThis change introduces capability to customize internal-client.conf,\nwhich is used by several daemon processes like container-sharder.\n\nChange-Id: I200e3a4ff323a07778ec6c4d7dd96824c72a12cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fbbfa6fe601f90c64133f61e99f317f4c8cb64dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fbbfa6fe601f90c64133f61e99f317f4c8cb64dd"}]},"branch":"refs/heads/master"},"f049c1ffabb4bfbdb70723f351389230e47e5949":{"kind":"REWORK","_number":5,"created":"2022-08-15 16:25:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/850122/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/22/850122/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/5"}}},"commit":{"parents":[{"commit":"68c875beb1056038fb1bc0df61c9e7812cf758b0","subject":"Fix wrong test description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/68c875beb1056038fb1bc0df61c9e7812cf758b0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-18 02:47:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-15 16:25:45.000000000","tz":540},"subject":"Support customizing internal-client.conf","message":"Support customizing internal-client.conf\n\nThis change introduces capability to customize internal-client.conf,\nwhich is used by several daemon processes like container-sharder.\n\nChange-Id: I200e3a4ff323a07778ec6c4d7dd96824c72a12cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f049c1ffabb4bfbdb70723f351389230e47e5949"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f049c1ffabb4bfbdb70723f351389230e47e5949"}]},"branch":"refs/heads/master"},"3ee737fa78494eaa503e004f624e93ff5f8c4ad9":{"kind":"REWORK","_number":6,"created":"2022-08-15 16:51:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/850122/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/22/850122/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/6"}}},"commit":{"parents":[{"commit":"68c875beb1056038fb1bc0df61c9e7812cf758b0","subject":"Fix wrong test description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/68c875beb1056038fb1bc0df61c9e7812cf758b0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-18 02:47:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-15 16:51:25.000000000","tz":540},"subject":"Support customizing internal-client.conf","message":"Support customizing internal-client.conf\n\nThis change introduces capability to customize internal-client.conf,\nwhich is used by several daemon processes like container-sharder.\n\nChange-Id: I200e3a4ff323a07778ec6c4d7dd96824c72a12cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3ee737fa78494eaa503e004f624e93ff5f8c4ad9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3ee737fa78494eaa503e004f624e93ff5f8c4ad9"}]},"branch":"refs/heads/master"},"5b03ade1787c7ce337dc42cc715de0672a0736c3":{"kind":"REWORK","_number":7,"created":"2022-08-15 17:30:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/850122/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/22/850122/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/7"}}},"commit":{"parents":[{"commit":"68c875beb1056038fb1bc0df61c9e7812cf758b0","subject":"Fix wrong test description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/68c875beb1056038fb1bc0df61c9e7812cf758b0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-18 02:47:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-15 17:30:38.000000000","tz":540},"subject":"Support customizing internal-client.conf","message":"Support customizing internal-client.conf\n\nThis change introduces capability to customize internal-client.conf,\nwhich is used by several daemon processes like container-sharder.\n\nChange-Id: I200e3a4ff323a07778ec6c4d7dd96824c72a12cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5b03ade1787c7ce337dc42cc715de0672a0736c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5b03ade1787c7ce337dc42cc715de0672a0736c3"}]},"branch":"refs/heads/master"},"18f68ea5246b5b8fc02a4230a32317090fd06cb1":{"kind":"REWORK","_number":8,"created":"2022-08-15 18:07:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/850122/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/22/850122/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/8"}}},"commit":{"parents":[{"commit":"68c875beb1056038fb1bc0df61c9e7812cf758b0","subject":"Fix wrong test description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/68c875beb1056038fb1bc0df61c9e7812cf758b0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-18 02:47:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-15 18:07:28.000000000","tz":540},"subject":"Support customizing internal-client.conf","message":"Support customizing internal-client.conf\n\nThis change introduces capability to customize internal-client.conf,\nwhich is used by several daemon processes like container-sharder.\n\nChange-Id: I200e3a4ff323a07778ec6c4d7dd96824c72a12cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/18f68ea5246b5b8fc02a4230a32317090fd06cb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/18f68ea5246b5b8fc02a4230a32317090fd06cb1"}]},"branch":"refs/heads/master"},"975bb0d11df7eb9b9ba8420ebfcca231d1ded11d":{"kind":"REWORK","_number":9,"created":"2022-08-16 00:17:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/850122/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/22/850122/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/9"}}},"commit":{"parents":[{"commit":"68c875beb1056038fb1bc0df61c9e7812cf758b0","subject":"Fix wrong test description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/68c875beb1056038fb1bc0df61c9e7812cf758b0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-18 02:47:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-16 00:17:36.000000000","tz":540},"subject":"Support customizing internal-client.conf","message":"Support customizing internal-client.conf\n\nThis change introduces capability to customize internal-client.conf,\nwhich is used by several daemon processes like container-sharder.\n\nChange-Id: I200e3a4ff323a07778ec6c4d7dd96824c72a12cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/975bb0d11df7eb9b9ba8420ebfcca231d1ded11d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/975bb0d11df7eb9b9ba8420ebfcca231d1ded11d"}]},"branch":"refs/heads/master"},"8a7502cadd416cf36602ecec8867ecb9dc5f386d":{"kind":"REWORK","_number":10,"created":"2022-08-16 01:50:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/850122/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/22/850122/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/10"}}},"commit":{"parents":[{"commit":"dd88f5625d25f8c6c32cf3146ae2d6fbbf3010c9","subject":"Merge \"Restart proxy-server when ceilometermiddleware is updated.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/dd88f5625d25f8c6c32cf3146ae2d6fbbf3010c9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-18 02:47:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-16 01:49:43.000000000","tz":540},"subject":"Support customizing internal-client.conf","message":"Support customizing internal-client.conf\n\nThis change introduces capability to customize internal-client.conf,\nwhich is used by several daemon processes like container-sharder.\n\nCloses-Bug: #1986599\nRelated-Bug: #1921319\nChange-Id: I200e3a4ff323a07778ec6c4d7dd96824c72a12cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8a7502cadd416cf36602ecec8867ecb9dc5f386d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8a7502cadd416cf36602ecec8867ecb9dc5f386d"}]},"branch":"refs/heads/master"},"843f03447e3a4e5220fe6e3fbb7fb75c8ee1d1eb":{"kind":"REWORK","_number":11,"created":"2022-08-16 01:51:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/850122/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/22/850122/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/11"}}},"commit":{"parents":[{"commit":"dd88f5625d25f8c6c32cf3146ae2d6fbbf3010c9","subject":"Merge \"Restart proxy-server when ceilometermiddleware is updated.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/dd88f5625d25f8c6c32cf3146ae2d6fbbf3010c9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-18 02:47:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-16 01:50:58.000000000","tz":540},"subject":"Support customizing internal-client.conf","message":"Support customizing internal-client.conf\n\nThis change introduces capability to customize internal-client.conf,\nwhich is used by several daemon processes like container-sharder.\n\nCloses-Bug: #1986599\nRelated-Bug: #1921319\nChange-Id: I200e3a4ff323a07778ec6c4d7dd96824c72a12cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/843f03447e3a4e5220fe6e3fbb7fb75c8ee1d1eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/843f03447e3a4e5220fe6e3fbb7fb75c8ee1d1eb"}]},"branch":"refs/heads/master"},"040a509b278f5d02d09e8f9a8d094aa58f78f0af":{"kind":"REWORK","_number":12,"created":"2022-08-16 01:53:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/850122/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/22/850122/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/12"}}},"commit":{"parents":[{"commit":"dd88f5625d25f8c6c32cf3146ae2d6fbbf3010c9","subject":"Merge \"Restart proxy-server when ceilometermiddleware is updated.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/dd88f5625d25f8c6c32cf3146ae2d6fbbf3010c9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-18 02:47:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-16 01:52:34.000000000","tz":540},"subject":"Support customizing internal-client.conf","message":"Support customizing internal-client.conf\n\nThis change introduces capability to customize internal-client.conf,\nwhich is used by several daemon processes like container-sharder.\n\nCloses-Bug: #1986599\nRelated-Bug: #1921319\nChange-Id: I200e3a4ff323a07778ec6c4d7dd96824c72a12cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/040a509b278f5d02d09e8f9a8d094aa58f78f0af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/040a509b278f5d02d09e8f9a8d094aa58f78f0af"}]},"branch":"refs/heads/master"},"efcef86de7ba859197aecd791b3b390a220a66d8":{"kind":"REWORK","_number":13,"created":"2022-08-16 02:27:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/850122/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/22/850122/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/22/850122/13"}}},"commit":{"parents":[{"commit":"dd88f5625d25f8c6c32cf3146ae2d6fbbf3010c9","subject":"Merge \"Restart proxy-server when ceilometermiddleware is updated.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/dd88f5625d25f8c6c32cf3146ae2d6fbbf3010c9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-18 02:47:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-16 02:26:43.000000000","tz":540},"subject":"Support customizing internal-client.conf","message":"Support customizing internal-client.conf\n\nThis change introduces capability to customize internal-client.conf,\nwhich is used by several daemon processes like container-sharder.\n\nCloses-Bug: #1986599\nRelated-Bug: #1921319\nChange-Id: I200e3a4ff323a07778ec6c4d7dd96824c72a12cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/efcef86de7ba859197aecd791b3b390a220a66d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/efcef86de7ba859197aecd791b3b390a220a66d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
