)]}'
{"id":"openstack%2Fpuppet-swift~851378","triplet_id":"openstack%2Fpuppet-swift~stable%2Fyoga~Ifdf86a6310fa8f8405579bd181e2070aa93d0b23","project":"openstack/puppet-swift","branch":"stable/yoga","topic":"trivial-fix-stable/yoga","attention_set":{},"removed_from_attention_set":{"20733":{"account":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"last_update":"2022-08-01 15:54:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifdf86a6310fa8f8405579bd181e2070aa93d0b23","subject":"Fix typo in parameter name","status":"MERGED","created":"2022-07-29 03:33:57.000000000","updated":"2022-08-01 15:54:37.000000000","submitted":"2022-08-01 15:54:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851378-trivial-fix-stable/yoga","cherry_pick_of_change":847767,"cherry_pick_of_patch_set":1,"meta_rev_id":"57664624467d42d89eed521485010f3ef2bc0331","_number":851378,"virtual_id_number":851378,"owner":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-01 15:54:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-07-29 13:29:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-08-01 14:34:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-08-01 14:34:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-29 04:53:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-29 10:47:37.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2022-07-29 13:29:09.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-08-01 14:34:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"a33ffcccfc4df227e47162133e0ef2f332c769f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2022-07-29 03:33:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cc1dcec9d23e172ea624961c12b15abbc7f0241b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 04:53:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f56dc51f5d924acab6423593d27fc239\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b73a2b22f6cc4cd7967e69ae6d6394cf : SUCCESS in 3m 32s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce356b31fa6d444992f3c28401a4befc : SUCCESS in 3m 32s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efd3ffa972724fe2a30edd8b75f02c56 : SUCCESS in 3m 36s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd6603fee9ba44c79f39ea1ec0a4d4d1 : RETRY_LIMIT in 17m 19s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c9547811f6c426da019af80853a772b : SUCCESS in 18m 54s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/440b2d425126493387d9f1df9899e12f : SUCCESS in 19m 09s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68291a9ca3e443d7a8f6a751a3961a14 : SUCCESS in 17m 33s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90529458fb0643b089d7d2a6615b431c : FAILURE in 56m 19s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7da87484bc4a40ca9191522adceeccc1 : RETRY_LIMIT in 10m 58s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5ff820625724d5da57855e7c5eca6e4 : SUCCESS in 1h 18m 02s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad7caac93dff47828412b7b52b677c8c : FAILURE in 25m 17s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c1b30aaa78e64a589755cb5fd9c60484 : RETRY_LIMIT in 8m 47s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a73d86b85434261bd4a0946e0a8b87e : SUCCESS in 32m 20s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f24e4ce10cdc4ee08ce2e0b3cde4b5f3 : FAILURE in 25m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/639f806ed3e44cfe8406f4a7976344ce : FAILURE in 26m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23627860b5ea411c528931f22db80f3987c09888","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-29 10:47:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7709c9a5a190ac570afc0d39a0e21680e6668d05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 12:13:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/702f01524af14ff2b1397a5c938944f5\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2afff8b96a7c423699a10beff3c86e4b : SUCCESS in 3m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c21a261c78834e49b308cb8b61d71222 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28223ef058f64362b0c0921e337670e2 : SUCCESS in 3m 42s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa26195874fb4c82ae86d0f16e3edce0 : SUCCESS in 21m 22s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51c1c2786269443db9a84a375b29ab38 : SUCCESS in 19m 58s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c66638bdf47847c7aea228e42257662f : SUCCESS in 18m 51s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44783574ef284cac81f9c4b12fdc9fe3 : SUCCESS in 21m 55s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e12cb8e1b64a4bf6a21aa8da67786263 : FAILURE in 52m 27s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d35331412ddc492592e2703b1b48a2df : SUCCESS in 1h 20m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a87b466088704b22817be34f2db28ddc : SUCCESS in 1h 19m 19s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35793a7ec4214f32914eadca30600330 : FAILURE in 22m 27s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/14f4fbc40db74f8c9d28c9564f38db00 : SUCCESS in 28m 30s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/1200b31d80a44d449f0f82413190478e : SUCCESS in 27m 48s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fdac64181e847798f9396049d8b2de5 : FAILURE in 22m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62a594dcfb1d49b7bf97e19be8718d22 : FAILURE in 20m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"994ca45d1293d95b173335f68abe7acac56ed2b2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-29 13:29:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4b9c5a41d812aa499723a92042f74345fe0431d8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-01 14:34:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"af988e3cc8455345294b200fea6cc501f4993ef4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 14:36:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b88a62604a8b41d9a5b039aa0b47e1f84a4d9b12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 15:54:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a320ca94ce24e19abbe37d6897e7191\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d4ea9434c744f39b44d51da51d24f55 : SUCCESS in 3m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/475e3f05992d4a6ba6f8ff8a11496c9a : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b7f3aac27f5481792028f79801bd09a : SUCCESS in 3m 50s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/e0f15a9867b24c96af03132552f5cea2 : SUCCESS in 20m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b150f51f2cb4b6faca4fe875f25b6e5 : SUCCESS in 19m 55s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc3d2cbc9df9427c80b9cc611aebc34c : SUCCESS in 17m 04s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f085907bf5304756bc78bccebd49cad0 : SUCCESS in 1h 16m 33s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d56efcf4c01413b98201c932426a5fd : SUCCESS in 32m 23s","accounts_in_message":[],"_revision_number":1},{"id":"57664624467d42d89eed521485010f3ef2bc0331","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 15:54:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9cc8b8d2d4d81041e2bff664c0555d149a1e02a1","revisions":{"9cc8b8d2d4d81041e2bff664c0555d149a1e02a1":{"kind":"REWORK","_number":1,"created":"2022-07-29 03:33:57.000000000","uploader":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"ref":"refs/changes/78/851378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/78/851378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/78/851378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/78/851378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/78/851378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/78/851378/1"}}},"commit":{"parents":[{"commit":"447044525358bcc79f1380ca4429f9ef7a22b542","subject":"Fix typo in swiftinit service type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/447044525358bcc79f1380ca4429f9ef7a22b542"}]}],"author":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2022-06-27 11:29:48.000000000","tz":330},"committer":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2022-07-29 03:33:57.000000000","tz":0},"subject":"Fix typo in parameter name","message":"Fix typo in parameter name\n\nThis change fixes typo in parameter name from\n`dnspython_pakage_name` to `dnspython_package_name`.\n\nThe tests were using correct parameter name but in manifests\nincorrect name was used.\n\nChange-Id: Ifdf86a6310fa8f8405579bd181e2070aa93d0b23\n(cherry picked from commit 814ec2de3617bde1a0423e39ef1ee987c2acc415)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9cc8b8d2d4d81041e2bff664c0555d149a1e02a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9cc8b8d2d4d81041e2bff664c0555d149a1e02a1"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
