)]}'
{"id":"openstack%2Fpuppet-swift~852708","triplet_id":"openstack%2Fpuppet-swift~stable%2Fyoga~I2a1d8e2d559bda4fda229f40aec7d97d14b10b11","project":"openstack/puppet-swift","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-12 07:56:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2a1d8e2d559bda4fda229f40aec7d97d14b10b11","subject":"Support more [oslo_messaging_rabbit] parameters","status":"MERGED","created":"2022-08-10 08:25:08.000000000","updated":"2022-08-12 07:57:11.000000000","submitted":"2022-08-12 07:56:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852708","meta_rev_id":"b5803e725e5bfd810e8680c7f5324a242fcbacc2","_number":852708,"virtual_id_number":852708,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-12 07:56:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-08-10 08:35:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-08-10 14:28:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-08-12 06:33:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-08-12 06:33:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-10 08:35:47.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-08-10 09:48:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-10 14:28:49.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-08-12 06:33:57.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"f2707ffb4b0dd5b2dec583ec1cf188745cda2462","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-10 08:25:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3027c60a87f42ffa91549ab61ba8078c25a68669","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-10 08:27:58.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8b513ba86586ed442c0e376fedd68b63968747d5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-10 08:35:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8903d6c509389f436aa50c50861df37d313fe459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 09:48:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad15f4a269b4479386e5bdc42ba52a07\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4dcfa48ae04d402094bc2f7652cb3fdb : SUCCESS in 3m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f99cf6390ad24f018e9b1b49ca250335 : SUCCESS in 3m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6223c54bd8a747b2b397870aa24f01b8 : SUCCESS in 3m 38s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/dbabd25dd7a04922ab6add8d491b3b7f : SUCCESS in 18m 28s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6279519048a24a06a76cf01ed3604c2b : SUCCESS in 15m 20s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92a81b8477d24fbc943b6793dcd6422c : SUCCESS in 15m 39s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ace687cab3d44e949d4ea4f8f27dd2c8 : SUCCESS in 19m 06s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b285e853a4640939a2dbe87aeff453d : FAILURE in 50m 13s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d12461ca86f3402084a6e4dd46e850ad : SUCCESS in 1h 18m 49s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7caf9749b7414437be5899ba38cd0cce : SUCCESS in 1h 13m 36s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6509c9697ef443c8eefcdef22a63eea : FAILURE in 24m 41s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/841f02220c364d3d9d0b213a11283845 : SUCCESS in 32m 56s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4a85c998eb946b48848ee14717b978e : SUCCESS in 30m 56s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4525e0cbd084d77a8eb5eab96ecc6b7 : FAILURE in 25m 36s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f8a95f766284ecc86dde18aab6a0eb9 : FAILURE in 24m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d31280f5d9e4c81be3a959ce41cae11 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"49686283cd38c2a4694316c132030e365de7571b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-10 14:28:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fe1771331a115bc9c5088e3006c876e042acaa70","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-12 06:33:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8c5ba22e0db4baf1678a780d00ba40ee84f91976","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 06:35:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d2001de4c589a1e99769ff31c27de7ae61b3b9b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 07:56:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df3ff195b92347f7b4c8cffb3d7ef651\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dc7852d87d14e04b2f05ed036d33b79 : SUCCESS in 3m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d59890243604504b0ad0a90d400f9ca : SUCCESS in 6m 01s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/547488bff46e4d48aa179d5b878e21da : SUCCESS in 3m 34s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/13a54f71c7d14c56a0eb71923a3bf36f : SUCCESS in 21m 37s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/578654a1428b47a7bd7107a4758e7522 : SUCCESS in 16m 29s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b08ef2cf7cc4b898e3ee68612784840 : SUCCESS in 16m 24s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a54c1a534424487b8c619c89f7e6834e : SUCCESS in 1h 17m 03s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/e32a6fa907294e6fbd1c070e3d9abd11 : SUCCESS in 31m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8baecf86cb294946be2105466756ea8f : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":2},{"id":"65c77a3aef23cb741829c8b3db4faf34db4b5a4b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 07:56:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b5803e725e5bfd810e8680c7f5324a242fcbacc2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 07:57:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/600a271c6cb141ae8654139d5e2deed1\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a95933b928464e96bb821428c0bda7b0 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6e834c34b35fb8c75168012d29ccd7d6f037db0f","revisions":{"c94d765c4b0276abe88c72550c16df1eb126d6aa":{"kind":"REWORK","_number":1,"created":"2022-08-10 08:25:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/852708/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/08/852708/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/08/852708/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/08/852708/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/08/852708/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/08/852708/1"}}},"commit":{"parents":[{"commit":"34d148dd272282524e22087a58816a35b569e2b5","subject":"Merge \"Use the root user to generate ring builder files\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/34d148dd272282524e22087a58816a35b569e2b5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-09 08:49:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-10 08:24:50.000000000","tz":540},"subject":"Support more [oslo_messaging_rabbit] parameters","message":"Support more [oslo_messaging_rabbit] parameters\n\nThe ceilometer middleware uses these options when creating notification\nclient instances.\n\nConflicts:\n\tmanifests/proxy/ceilometer.pp\n\nChange-Id: I2a1d8e2d559bda4fda229f40aec7d97d14b10b11\n(cherry picked from commit 67454cf7ff499c7c11b0a0ae8d1563dcc904f2ed)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c94d765c4b0276abe88c72550c16df1eb126d6aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c94d765c4b0276abe88c72550c16df1eb126d6aa"}]},"branch":"refs/heads/stable/yoga"},"6e834c34b35fb8c75168012d29ccd7d6f037db0f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-08-10 08:27:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/852708/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/08/852708/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/08/852708/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/08/852708/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/08/852708/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/08/852708/2"}}},"commit":{"parents":[{"commit":"34d148dd272282524e22087a58816a35b569e2b5","subject":"Merge \"Use the root user to generate ring builder files\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/34d148dd272282524e22087a58816a35b569e2b5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-09 08:49:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-10 08:27:42.000000000","tz":540},"subject":"Support more [oslo_messaging_rabbit] parameters","message":"Support more [oslo_messaging_rabbit] parameters\n\nThe ceilometer middleware uses these options when creating notification\nclient instances.\n\nConflicts:\n\tmanifests/proxy/ceilometer.pp\n\nResolved conflicts caused by 152587fd2a1d0c0bc7ad67585fe2ecdd6abde995 .\n\nChange-Id: I2a1d8e2d559bda4fda229f40aec7d97d14b10b11\n(cherry picked from commit 67454cf7ff499c7c11b0a0ae8d1563dcc904f2ed)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/6e834c34b35fb8c75168012d29ccd7d6f037db0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/6e834c34b35fb8c75168012d29ccd7d6f037db0f"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
