)]}'
{"id":"openstack%2Fpuppet-swift~852710","triplet_id":"openstack%2Fpuppet-swift~stable%2Fwallaby~I2a1d8e2d559bda4fda229f40aec7d97d14b10b11","project":"openstack/puppet-swift","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-12 08:47:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2a1d8e2d559bda4fda229f40aec7d97d14b10b11","subject":"Support more [oslo_messaging_rabbit] parameters","status":"MERGED","created":"2022-08-10 08:30:27.000000000","updated":"2022-08-12 08:48:07.000000000","submitted":"2022-08-12 08:47:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852710","meta_rev_id":"08661d98394e11155cfde8c19fcf4394475ac978","_number":852710,"virtual_id_number":852710,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-12 08:47:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-08-10 08:36:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-08-10 14:28:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-08-12 06:34:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-08-12 06:34:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-10 08:36:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-08-10 11:01:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-10 14:28:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-08-12 06:34:20.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"3bd9ec4915f77ee78b57b8bab1c0a129da811eaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-10 08:30:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b9e846b879ad5ab4e0dd24251a78530a55e721a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-10 08:36:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bf0adb42a00e484e811116f05e18ebbaa469764a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 11:01:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/180ffc25331f440ca6dd8270c27ed6a5\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8fe3b6b19154add9be82a85655c7062 : SUCCESS in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/039b7218063347b5ad5f2f52b244b524 : SUCCESS in 3m 43s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/21cdf979af60489f9de05ebedc53cd63 : SUCCESS in 18m 50s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f19a6210fae74caabc85545bcf8cdc04 : SUCCESS in 15m 23s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e16317b0192b4d3181f64db64f8bc666 : SUCCESS in 14m 13s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/876994c803894131aeb6e39458c89cf7 : FAILURE in 42m 18s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd08458e94f34742bc3b711a2ae628d6 : SUCCESS in 1h 45m 18s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/f88a65e812814ab3b92de560556317c3 : SUCCESS in 33m 42s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29708ba00bd9423ab66a5725980ece76 : FAILURE in 27m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/450d8ae8e14f4cecbe6810750e1dd0e3 : SUCCESS in 3m 38s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c6f7f966034648eb979aa0a0fc68271b : SUCCESS in 2h 23m 22s\n- puppet-swift-tripleo-centos-9-standalone https://zuul.opendev.org/t/openstack/build/42a698cc42b94a0084bb8e3dfca376bb : SUCCESS in 1h 43m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40f70f8600c6d0468244f63700226c1ac8cd66d8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-10 14:28:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c3ad8c68d3eed34df10a7d179babf1577b1817ab","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-12 06:34:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3510e5d245a43e163ceb1a92088a4dcea3fb546e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 06:35:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6faab033ba0c307ec0d1f705fcb4503aec3360b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 08:47:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f7fa6c3057a4688befd9a862fd088c0\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/204824bbfde0449d83a1d26688b9fe78 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a4c063a055e4c72917784f94acbd523 : SUCCESS in 3m 55s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/acef4f12e9b74c44adfaf11b34526729 : SUCCESS in 19m 41s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/330aee75f8154b9ea9d8ee6eb9a6daab : SUCCESS in 15m 37s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/749e54d87f30455ea425d4167414f9bc : SUCCESS in 1h 39m 20s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/82ebadea3ac940d78d6db030afd49af0 : SUCCESS in 34m 22s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fb5927ca9a344e693e2bf6b8e8d147f : FAILURE in 26m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e01fc3d47024ef2979447aab4859353 : SUCCESS in 4m 02s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d821f1b78bd245879656813ec2ad7502 : SUCCESS in 2h 06m 22s","accounts_in_message":[],"_revision_number":1},{"id":"70fbf2db76822367fec0da18d3dc83153cab79ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 08:47:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"08661d98394e11155cfde8c19fcf4394475ac978","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 08:48:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a37852853a2444a8465e96e83f8c16b\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/494f887050e2467d916f12b22b1bd659 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba44c317149d389c0fb0726b82e9a73dbdf7870e","revisions":{"ba44c317149d389c0fb0726b82e9a73dbdf7870e":{"kind":"REWORK","_number":1,"created":"2022-08-10 08:30:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/852710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/10/852710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/10/852710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/10/852710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/10/852710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/10/852710/1"}}},"commit":{"parents":[{"commit":"195425d3f14427f11d909861501227995b4be0fe","subject":"Fix typo in swiftinit service type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/195425d3f14427f11d909861501227995b4be0fe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-09 08:49:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-10 08:30:18.000000000","tz":540},"subject":"Support more [oslo_messaging_rabbit] parameters","message":"Support more [oslo_messaging_rabbit] parameters\n\nThe ceilometer middleware uses these options when creating notification\nclient instances.\n\nConflicts:\n\tmanifests/proxy/ceilometer.pp\n\nResolved conflicts caused by 152587fd2a1d0c0bc7ad67585fe2ecdd6abde995 .\n\nConflicts:\n\tmanifests/proxy/ceilometer.pp\n\nResolved conflicts caused by 2f2183e9d634c4d46834472ba221969ca4a88052 .\n\nChange-Id: I2a1d8e2d559bda4fda229f40aec7d97d14b10b11\n(cherry picked from commit 67454cf7ff499c7c11b0a0ae8d1563dcc904f2ed)\n(cherry picked from commit 6e834c34b35fb8c75168012d29ccd7d6f037db0f)\n(cherry picked from commit ff757c6eb3f84ecbc97b2db1cd8b6bcbb30b2f67)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ba44c317149d389c0fb0726b82e9a73dbdf7870e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ba44c317149d389c0fb0726b82e9a73dbdf7870e"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
