)]}'
{"id":"openstack%2Fpuppet-swift~853929","triplet_id":"openstack%2Fpuppet-swift~master~Ibc7479e7defbfa4c49bf3c9f1574cdcf61b90ab9","project":"openstack/puppet-swift","branch":"master","topic":"bug/1987260","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-25 07:39:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibc7479e7defbfa4c49bf3c9f1574cdcf61b90ab9","subject":"Fix type validation warning for storage policy rings","status":"MERGED","created":"2022-08-22 08:15:00.000000000","updated":"2022-08-25 07:39:07.000000000","submitted":"2022-08-25 07:39:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853929-bug/1987260","meta_rev_id":"e5fb25eb4815d3faedb3fa3728bbebea78b5eeb1","_number":853929,"virtual_id_number":853929,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-25 07:39:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2022-08-23 01:16:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":2,"date":"2022-08-22 15:42:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-08-25 05:59:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-08-25 05:59:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-22 12:27:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-22 15:42:46.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-08-23 01:16:43.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2022-08-25 05:59:19.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"111fc579c4d65412614f2b3f61a34993cbe1fc76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-22 08:15:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f885275fdd519b04324d0577cffdac695c652bc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-22 08:28:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cfb9562a12b6ce05d384e0607798ae2282a5cf1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-22 08:51:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a81e864d42a8caf84b8a63846517979f9c20992f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-22 08:56:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"20e3a1f65ae324dc2c96d7da36da9d3d8784aa18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-22 09:54:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3d3a56976f3c0e49edf3cd6129d0b3481e43aa2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-22 10:37:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1a966bbd604af0e1aa98d233a590787846530d8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 12:27:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5949db9f91fa49e99871c0526d55b4fb\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f894819295bc4e70a3969e39ac14effd : SUCCESS in 4m 10s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5778cda178e4cd69e9bca1480e1f863 : SUCCESS in 3m 44s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/557c896f1bfd4e108dfc967014c052f3 : SUCCESS in 3m 47s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/671249194718404395ec1bdf48960b7d : SUCCESS in 17m 22s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a315ba4951814de880c60bba6b577666 : SUCCESS in 12m 59s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/064a158c162049c3a63e3fb55ac003c7 : SUCCESS in 12m 59s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a29fc84658e4be59e3532791ad21e02 : SUCCESS in 12m 57s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/407b2057fca94bd68e2c68007bb9051e : FAILURE in 1h 01m 56s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dafcb2cca2f3467593a75b957f347f76 : SUCCESS in 1h 15m 04s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75ba497095a246e196faf6e1c4066117 : FAILURE in 24m 18s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/02962a234f9f430496280cb5afabb394 : SUCCESS in 31m 16s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07e87c0ec8c94e75b6c3ab6589347aea : FAILURE in 23m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bb97ea5d6124572a3173414337debf0 : FAILURE in 23m 40s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0729ffed04f246988ad9d9f0bbd58b43 : SUCCESS in 1h 48m 02s","accounts_in_message":[],"_revision_number":6},{"id":"0ca4fd0dd392a8409cb450fd130cf4e032acb060","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-22 15:42:46.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"1dcad85ace4e42dd9883ec12d347bcbe685bbcc8","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-08-23 01:16:43.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"c8b0c5b65ba219af43e5fff07690696a70d4e073","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-25 05:59:19.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"d4f1a51d1b62762e3177b2178df503840a0e3c5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 06:00:41.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1d83fabeca233413f8f45559c68fa35b0072b468","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 07:39:06.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/604109502f0f49c49b150acf135700d0\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b81aafd598f541b2893f150a01b9490a : SUCCESS in 3m 59s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b77bc35e48714510aa080196ff408563 : SUCCESS in 5m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d91b082d70e4dd29bb33879544e9991 : SUCCESS in 5m 11s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/268e3cd0cb36450f803bb0b051e8719b : SUCCESS in 27m 17s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8243533aa1c445c3a84206f6812a1338 : SUCCESS in 21m 14s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f9767b89dd24d908b4acd956436b7aa : SUCCESS in 21m 01s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1eb9bc371004d388a8154114e180d43 : SUCCESS in 1h 16m 57s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc6c70b29c954d7fa31e19aa3c70f4d7 : SUCCESS in 39m 04s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4961da2628db48de8689c1d11b4ea900 : SUCCESS in 1h 17m 17s","accounts_in_message":[],"_revision_number":6},{"id":"e5fb25eb4815d3faedb3fa3728bbebea78b5eeb1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 07:39:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"11b193cab8f880f1b73c57208646a889979e6fd8","revisions":{"e0d0740880e9e860ca3d6cee77b784e55acf6a53":{"kind":"REWORK","_number":1,"created":"2022-08-22 08:15:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/853929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/29/853929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/1"}}},"commit":{"parents":[{"commit":"839ccc7f20ceeaa60965409ea59e6cc2c280e6ec","subject":"Merge \"Use ini provider to manage storage config files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/839ccc7f20ceeaa60965409ea59e6cc2c280e6ec"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-22 08:11:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-22 08:14:48.000000000","tz":540},"subject":"Fix type validation warning for storage policy rings","message":"Fix type validation warning for storage policy rings\n\nSince 63688a14e57282ae97072b406c81d0da2f61965c was merged, this module\nsupport configuring storage policies.\n\nWhen multiple storage policies are used, we need to create multiple\nobject rings like object.ring.gz, object-1.ring.gz. However the current\nregexp to validate the ring name does not consider this format and\nshows the type validation warning when creating a non-default policy\nrings.\n\nThis change updates the regexp used in validation to accept that\nobject-\u003cnumber\u003e format as well.\n\nCloses-Bug: #1987260\nChange-Id: Ibc7479e7defbfa4c49bf3c9f1574cdcf61b90ab9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e0d0740880e9e860ca3d6cee77b784e55acf6a53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e0d0740880e9e860ca3d6cee77b784e55acf6a53"}]},"branch":"refs/heads/master"},"d8c4b4d3b09a4e6e76529834a91417d4845938f6":{"kind":"REWORK","_number":2,"created":"2022-08-22 08:28:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/853929/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/29/853929/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/2"}}},"commit":{"parents":[{"commit":"839ccc7f20ceeaa60965409ea59e6cc2c280e6ec","subject":"Merge \"Use ini provider to manage storage config files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/839ccc7f20ceeaa60965409ea59e6cc2c280e6ec"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-22 08:11:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-22 08:27:54.000000000","tz":540},"subject":"Fix type validation warning for storage policy rings","message":"Fix type validation warning for storage policy rings\n\nSince 63688a14e57282ae97072b406c81d0da2f61965c was merged, this module\nsupport configuring storage policies.\n\nWhen multiple storage policies are used, we need to create multiple\nobject rings like object.ring.gz, object-1.ring.gz. However the current\nregexp to validate the ring name does not consider this format and\nshows the type validation warning when creating a non-default policy\nrings.\n\nThis change updates the regexp used in validation to accept that\nobject-\u003cnumber\u003e format as well.\n\nCloses-Bug: #1987260\nChange-Id: Ibc7479e7defbfa4c49bf3c9f1574cdcf61b90ab9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d8c4b4d3b09a4e6e76529834a91417d4845938f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d8c4b4d3b09a4e6e76529834a91417d4845938f6"}]},"branch":"refs/heads/master"},"43e242e5a79ef5e14348c7780f835c457f7c5a1b":{"kind":"REWORK","_number":3,"created":"2022-08-22 08:51:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/853929/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/29/853929/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/3"}}},"commit":{"parents":[{"commit":"839ccc7f20ceeaa60965409ea59e6cc2c280e6ec","subject":"Merge \"Use ini provider to manage storage config files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/839ccc7f20ceeaa60965409ea59e6cc2c280e6ec"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-22 08:11:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-22 08:51:10.000000000","tz":540},"subject":"Fix type validation warning for storage policy rings","message":"Fix type validation warning for storage policy rings\n\nSince 63688a14e57282ae97072b406c81d0da2f61965c was merged, this module\nsupport configuring storage policies.\n\nWhen multiple storage policies are used, we need to create multiple\nobject rings like object.ring.gz, object-1.ring.gz. However the current\nregexp to validate the ring name does not consider this format and\nshows the type validation warning when creating a non-default policy\nrings.\n\nThis change updates the regexp used in validation to accept that\nobject-\u003cnumber\u003e format as well.\n\nCloses-Bug: #1987260\nChange-Id: Ibc7479e7defbfa4c49bf3c9f1574cdcf61b90ab9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/43e242e5a79ef5e14348c7780f835c457f7c5a1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/43e242e5a79ef5e14348c7780f835c457f7c5a1b"}]},"branch":"refs/heads/master"},"eeef65af8fa3b2b8267cebe0c8e70a3e06446647":{"kind":"REWORK","_number":4,"created":"2022-08-22 08:56:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/853929/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/29/853929/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/4"}}},"commit":{"parents":[{"commit":"839ccc7f20ceeaa60965409ea59e6cc2c280e6ec","subject":"Merge \"Use ini provider to manage storage config files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/839ccc7f20ceeaa60965409ea59e6cc2c280e6ec"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-22 08:11:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-22 08:56:12.000000000","tz":540},"subject":"Fix type validation warning for storage policy rings","message":"Fix type validation warning for storage policy rings\n\nSince 63688a14e57282ae97072b406c81d0da2f61965c was merged, this module\nsupport configuring storage policies.\n\nWhen multiple storage policies are used, we need to create multiple\nobject rings like object.ring.gz, object-1.ring.gz. However the current\nregexp to validate the ring name does not consider this format and\nshows the type validation warning when creating a non-default policy\nrings.\n\nThis change updates the regexp used in validation to accept that\nobject-\u003cnumber\u003e format as well.\n\nCloses-Bug: #1987260\nChange-Id: Ibc7479e7defbfa4c49bf3c9f1574cdcf61b90ab9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/eeef65af8fa3b2b8267cebe0c8e70a3e06446647"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/eeef65af8fa3b2b8267cebe0c8e70a3e06446647"}]},"branch":"refs/heads/master"},"b8d097fd220a044375fd14309f4eb5733aa74ca2":{"kind":"REWORK","_number":5,"created":"2022-08-22 09:54:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/853929/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/29/853929/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/5"}}},"commit":{"parents":[{"commit":"839ccc7f20ceeaa60965409ea59e6cc2c280e6ec","subject":"Merge \"Use ini provider to manage storage config files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/839ccc7f20ceeaa60965409ea59e6cc2c280e6ec"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-22 08:11:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-22 09:54:27.000000000","tz":540},"subject":"Fix type validation warning for storage policy rings","message":"Fix type validation warning for storage policy rings\n\nSince 63688a14e57282ae97072b406c81d0da2f61965c was merged, this module\nsupport configuring storage policies.\n\nWhen multiple storage policies are used, we need to create multiple\nobject rings like object.ring.gz, object-1.ring.gz. However the current\nregexp to validate the ring name does not consider this format and\nshows the type validation warning when creating a non-default policy\nrings.\n\nThis change updates the regexp used in validation to accept that\nobject-\u003cnumber\u003e format as well.\n\nCloses-Bug: #1987260\nChange-Id: Ibc7479e7defbfa4c49bf3c9f1574cdcf61b90ab9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b8d097fd220a044375fd14309f4eb5733aa74ca2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b8d097fd220a044375fd14309f4eb5733aa74ca2"}]},"branch":"refs/heads/master"},"11b193cab8f880f1b73c57208646a889979e6fd8":{"kind":"REWORK","_number":6,"created":"2022-08-22 10:37:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/853929/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/29/853929/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/29/853929/6"}}},"commit":{"parents":[{"commit":"839ccc7f20ceeaa60965409ea59e6cc2c280e6ec","subject":"Merge \"Use ini provider to manage storage config files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/839ccc7f20ceeaa60965409ea59e6cc2c280e6ec"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-22 08:11:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-22 10:37:40.000000000","tz":540},"subject":"Fix type validation warning for storage policy rings","message":"Fix type validation warning for storage policy rings\n\nSince 63688a14e57282ae97072b406c81d0da2f61965c was merged, this module\nsupport configuring storage policies.\n\nWhen multiple storage policies are used, we need to create multiple\nobject rings like object.ring.gz, object-1.ring.gz. However the current\nregexp to validate the ring name does not consider this format and\nshows the type validation warning when creating a non-default policy\nrings.\n\nThis change updates the regexp used in validation to accept that\nobject-\u003cnumber\u003e format as well.\n\nCloses-Bug: #1987260\nChange-Id: Ibc7479e7defbfa4c49bf3c9f1574cdcf61b90ab9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/11b193cab8f880f1b73c57208646a889979e6fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/11b193cab8f880f1b73c57208646a889979e6fd8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
