)]}'
{"id":"openstack%2Fpuppet-swift~854131","triplet_id":"openstack%2Fpuppet-swift~stable%2Fwallaby~I200e3a4ff323a07778ec6c4d7dd96824c72a12cf","project":"openstack/puppet-swift","branch":"stable/wallaby","topic":"bug/1986599-stable/wallaby","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-25 07:39:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I200e3a4ff323a07778ec6c4d7dd96824c72a12cf","subject":"Support customizing internal-client.conf","status":"MERGED","created":"2022-08-23 10:04:04.000000000","updated":"2022-08-25 07:40:09.000000000","submitted":"2022-08-25 07:39:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854131-bug/1986599-stable/wallaby","cherry_pick_of_change":853766,"cherry_pick_of_patch_set":2,"meta_rev_id":"d97e4d021b412ee34db1a8b5c4a67712fd606ecf","_number":854131,"virtual_id_number":854131,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-25 07:39:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-08-23 18:41:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-08-25 05:55:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-08-25 05:55:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-23 11:54:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-23 18:41:56.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-08-25 05:55:18.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"b8dd00729b5169451e18ee06153ca2525083af51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-23 10:04:04.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"36f560b0dbc9334010e8463d3d1cac54c7dd5913","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-23 10:04:11.000000000","message":"Topic bug/1986599-stable/xena-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"12df68fb3aff3df35f52012d203b74074790699c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-23 10:04:18.000000000","message":"Topic set to bug/1986599-stable/wallaby","accounts_in_message":[],"_revision_number":1},{"id":"10fb36e40a5c01c2849ebd22667c8c235d7798e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 11:54:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44454583cfa946dd840d93bd67e6c30f\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/698ae3cbca9f452597bde2a7a8c7037e : SUCCESS in 3m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c24b963814a64d16ba7c4dbd94d27c08 : SUCCESS in 3m 26s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3906abbd59d941e29641323ea082fd6a : SUCCESS in 18m 45s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26cf777d136d4fb09734ce984f95cfd3 : SUCCESS in 18m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e247d2036eba4ad9ad68e386de1c4bcd : SUCCESS in 13m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afb58c6900a04a8dae81e8f3fc1cd6d5 : FAILURE in 1h 05m 30s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f12509fa122c4a93b16e2f5708b54a1b : SUCCESS in 1h 43m 33s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/818f4f1c8bed49c68d9076ec981f4bb8 : SUCCESS in 39m 42s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5be05b63882947f68d841c7bf72739f1 : FAILURE in 23m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39fdbc92eac74607b894c9a24941c135 : SUCCESS in 4m 17s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/243ab1a5001948a58020af65348ededf : SUCCESS in 1h 46m 51s\n- puppet-swift-tripleo-centos-9-standalone https://zuul.opendev.org/t/openstack/build/b1d84ad9c90e49ebacada68d45650461 : SUCCESS in 1h 40m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"829b112f7736917c78e2c217ca6f8fa4a4a31a87","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-23 18:41:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c6ff13a53a0ca5aa3be23a6e9841ea055b954ea3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-25 05:55:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ee94ee170d9476c4d371483139ce389c920fc305","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 05:56:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7df5725ee9247916c44bb06c422484f23180efc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 07:39:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6715e9685334440abe0cf9675a493e48\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/530f9398b8834b388abb211d5985d361 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/041b474c3aa543cc882b38bba1d44a99 : SUCCESS in 3m 20s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b7049f9c4ec348f8b1839f07a8a47979 : SUCCESS in 17m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10af1df8d30c4b2a824c59153c26186a : SUCCESS in 14m 43s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3883200a4cc8406bb7de2320b0cac190 : SUCCESS in 1h 13m 40s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c1fd0036d0a49b4823edfcd4e0f5b4a : SUCCESS in 45m 20s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a7903a7fa1941bd856f0050c13296f1 : FAILURE in 21m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d34b6253f400484d899d4b99e6e43601 : SUCCESS in 4m 25s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ecc0bbb827884934844ebb1c26a41a23 : SUCCESS in 1h 40m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f831d9ccd3bc1d0cdbfdf1fb69cad47cb1d1755f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 07:39:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d97e4d021b412ee34db1a8b5c4a67712fd606ecf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 07:40:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee83dbff8368451f956fdf0fadb97351\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/169157eac1e34d37b696ab9e4337b6d4 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4267671c250c0b980d97a23a3f59c1a7c184a07f","revisions":{"4267671c250c0b980d97a23a3f59c1a7c184a07f":{"kind":"REWORK","_number":1,"created":"2022-08-23 10:04:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/854131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/31/854131/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/31/854131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/31/854131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/31/854131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/31/854131/1"}}},"commit":{"parents":[{"commit":"ba44c317149d389c0fb0726b82e9a73dbdf7870e","subject":"Support more [oslo_messaging_rabbit] parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ba44c317149d389c0fb0726b82e9a73dbdf7870e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-18 02:47:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-23 10:04:04.000000000","tz":0},"subject":"Support customizing internal-client.conf","message":"Support customizing internal-client.conf\n\nThis change introduces capability to customize internal-client.conf,\nwhich is used by several daemon processes like container-sharder.\n\nBackport note:\nReplaced swift::params::user by the hard-coded user name because\nf781eb9e34501b651d3f2069fc8848e7cacef679 is present only in yoga and\nlater.\n\nCloses-Bug: #1986599\nRelated-Bug: #1921319\nChange-Id: I200e3a4ff323a07778ec6c4d7dd96824c72a12cf\n(cherry picked from commit efcef86de7ba859197aecd791b3b390a220a66d8)\n(cherry picked from commit ab3cf92d79615453b041b67233a38daf70467331)\n(cherry picked from commit b5cb280779127e3a7924573f280de7c6456aa311)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4267671c250c0b980d97a23a3f59c1a7c184a07f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4267671c250c0b980d97a23a3f59c1a7c184a07f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
