)]}'
{"id":"openstack%2Fpuppet-swift~855825","triplet_id":"openstack%2Fpuppet-swift~master~Ibed7f31e9ad472ac490ac6a2933ffa2c8329f376","project":"openstack/puppet-swift","branch":"master","topic":"ratelimit-opts","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-09-07 19:57:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibed7f31e9ad472ac490ac6a2933ffa2c8329f376","subject":"ratelimit: Support options for container access ratelimit","status":"MERGED","created":"2022-09-04 15:17:59.000000000","updated":"2022-09-07 19:57:33.000000000","submitted":"2022-09-07 19:57:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855825-ratelimit-opts","meta_rev_id":"42d1e9c745b824fca73ef2d486809a6f4231dca3","_number":855825,"virtual_id_number":855825,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-07 19:57:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-09-05 07:40:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-09-07 17:47:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-09-07 17:47:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-04 16:58:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-05 07:40:09.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-09-07 17:47:41.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"cc0c167c0c6b98fb1612d01b04544d877b05b522","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-04 15:17:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2ec6d4d1cd2857fb8f0e5b4eae8e98c86b697cf","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-04 15:18:16.000000000","message":"Topic set to ratelimit-opts","accounts_in_message":[],"_revision_number":1},{"id":"d01f7e395c6435f46c3402a2ba0236bbb1e4c983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-04 16:58:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f647a95d25434e4196dc14f54b717e59\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55dfd244d19641bcb9cea0b00067f905 : SUCCESS in 3m 32s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/845afd0285314c6a8e16ce80cee495ad : SUCCESS in 3m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4f81c2b78504f4c9e758de20067a7fd : SUCCESS in 3m 24s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a00854f4796841a893e2ada00461c207 : SUCCESS in 17m 18s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6ec4dc53c89488fa46adae737d242ca : SUCCESS in 12m 10s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7981d308baa441fca86ee9c505c580fc : SUCCESS in 13m 51s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77820fc478a145579a636abe460f6877 : SUCCESS in 13m 24s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36bfe96bea5041f89a09910de65b4907 : SUCCESS in 1h 02m 39s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/47da9ed474194550afb7be455d648331 : SUCCESS in 1h 11m 06s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25840ea1ba8947329a2fe811a4827ffa : FAILURE in 20m 17s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/02034af20e524c0eb6fee54c2e37380f : SUCCESS in 32m 01s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1d5391b688f49c5889b76ed9d22fb19 : FAILURE in 23m 36s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9549b1bcdbdd4b73a65dd9b296c0b31a : FAILURE in 18m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a393da7c13946c5aeeb3bdd6528e800 : SUCCESS in 4m 08s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ed17af85ee1949c98e3a131aff537a24 : SUCCESS in 1h 37m 09s","accounts_in_message":[],"_revision_number":1},{"id":"61d9e77d84c8951920534314f2f21f8746cad783","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-05 07:40:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7bc40215ca7875a6d00427dc0ec8ee97b0067728","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-09-07 17:47:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"73fd6258ebe2ab29cfcfd88b508569a6776f3962","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 17:49:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c0a0e8dbe8bbff7893a75db27e6fef917c461f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 19:57:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/068295b603444efea0a0efbda2fd7424\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d6e0abde3c24c1798085e4b745b6eb3 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e4f63f8dc8045c1bd81365f2293d855 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79c66f06635e4d598866951c951b89d6 : SUCCESS in 3m 39s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/695ae53a2a1a44ba9f38f28f33624a17 : SUCCESS in 14m 40s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a8b73853a2643c5803d08f2a558a6af : SUCCESS in 13m 25s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd8f7901956e4dd1b7f2d4ca8031b8f2 : SUCCESS in 12m 11s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6918f4ade7394523bfb2b1744ae36acc : SUCCESS in 1h 14m 12s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/4218029da1d749538cf2abca096ec799 : SUCCESS in 28m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af26e538664a4937b7ace41e387348a5 : SUCCESS in 4m 56s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/311587fa946a4c2eba0ae8195119bb25 : SUCCESS in 1h 31m 16s","accounts_in_message":[],"_revision_number":1},{"id":"42d1e9c745b824fca73ef2d486809a6f4231dca3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 19:57:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9686d3e41b660c2301ffc490e43237c72de226ff","revisions":{"9686d3e41b660c2301ffc490e43237c72de226ff":{"kind":"REWORK","_number":1,"created":"2022-09-04 15:17:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/855825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/25/855825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/25/855825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/25/855825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/25/855825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/25/855825/1"}}},"commit":{"parents":[{"commit":"12bb9f0cfe78528d8d52261d535cbfeee6888d0b","subject":"ratelimit: Remove redundant hard-coded defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/12bb9f0cfe78528d8d52261d535cbfeee6888d0b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-04 15:16:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-04 15:17:47.000000000","tz":540},"subject":"ratelimit: Support options for container access ratelimit","message":"ratelimit: Support options for container access ratelimit\n\nThis change introduces support for the following two options which\ncontrols rate limit for container access.\n - container_ratelimit_\u003csize\u003e \u003d \u003crate\u003e\n - container_listing_ratelimit_\u003csize\u003e \u003d \u003crate\u003e\n\nChange-Id: Ibed7f31e9ad472ac490ac6a2933ffa2c8329f376\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9686d3e41b660c2301ffc490e43237c72de226ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9686d3e41b660c2301ffc490e43237c72de226ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
