)]}'
{"id":"openstack%2Fpuppet-swift~855955","triplet_id":"openstack%2Fpuppet-swift~master~I6092105916e3c49ad0e0a45156b43cce99cbe3ce","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-09-07 19:57:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6092105916e3c49ad0e0a45156b43cce99cbe3ce","subject":"CentOS: Use consistent host parameters instead of hard-codes","status":"MERGED","created":"2022-09-05 15:42:29.000000000","updated":"2022-09-07 19:57:41.000000000","submitted":"2022-09-07 19:57:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855955","meta_rev_id":"fb89bd60bb097f4ff81ada860d142047c26f689d","_number":855955,"virtual_id_number":855955,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-07 19:57:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-09-06 18:10:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-09-07 17:48:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-09-07 17:48:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-05 16:10:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-06 18:10:00.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-09-07 17:48:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"7129455ec2afec994163e4cfc44e4ab8e02f0058","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-05 15:42:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89f565c2edf7a3aa5bc7fcf7fe3028ac122c09b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 16:10:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/314921eba7934827a85658f33b02b9a6\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c2c842f518d456396ec49353ec4ea5d : SUCCESS in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2edb4f6786d84684a09e67702465caa1 : SUCCESS in 5m 27s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afb26bd44a63490ebdebe1b66e99f449 : SUCCESS in 2m 55s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29ad4d467a9d4334943b9d7894d39ea8 : FAILURE in 18m 44s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/7bbad808c41e497d91c119e44cf352ab : FAILURE in 23m 38s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9eb9ed555b2f41f6b68dc453e7ed587c : FAILURE in 18m 17s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5728a1b42a445b7924ac5715cccf568 : FAILURE in 16m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f9d7088ca776671e1b2eefebf5fe1c6d45ceaf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-05 16:25:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7af60aee0d5253db4fc7d2df279ab7bee0d8a11f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 16:59:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3461a3f823854ba7ab2c08c93b8fbe04\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3694ed33dd294957a5a5f90fd4558a83 : SUCCESS in 3m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65ee28b067b8477aa75ceef099c165bd : SUCCESS in 3m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/278127e5f3b74136b183f5cd5a3a03d6 : SUCCESS in 3m 51s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eeca55ceaa23456abb9f3cfdc0c6e7a5 : FAILURE in 23m 52s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/82ae3653101b46c7a30082f517e2a0cf : SUCCESS in 32m 09s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37d3ec88887a452086d9dbaa5a3c3c6c : FAILURE in 24m 57s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae516ded00c64fd9bbff62ab936ec06a : FAILURE in 23m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0467dbbe3b3bf5ebdc3c3e213007a85db7702715","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-05 17:50:07.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c4b6e0acef1e60448ceae38841b911ffabcebbcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 18:17:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dae3d5d4a0374ac6932307e7410d0482\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20b2f17ceaf44da4988d6709a9e6d5fe : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0cc6a4620fc4bb199a111bf387657e6 : SUCCESS in 5m 27s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f18a759720314d35862efcd38b55f72a : SUCCESS in 3m 16s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77cd334ba2e54f4c8aa7a5292d8e4a77 : FAILURE in 19m 17s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2689f75a0e104b00a41c3176bef17ecf : SUCCESS in 25m 36s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29ce0e23811f49c8b31b538548c25dcf : FAILURE in 19m 57s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8dfe9a26f85f42fc92a9620402488c1c : FAILURE in 20m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1dc0e6e1991badb6ed2bd2190d3d4d422cdf52bc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-06 18:10:00.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"da5669b74b6db64f98578ad89b188d6f5b96dc5a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-09-07 17:48:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d8e42675403f834b6e265f470d8d1860532eb72b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 17:49:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"83bc97923ee3ab9decf6f3dcb2f577228d5db1db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 19:57:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/788398f3ef9149f7a4c81cb5f04d63f9\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffea8c3d0138446bb9f079223f0caabb : SUCCESS in 4m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/284abe5aa78c420899da6fa5a38ecb01 : SUCCESS in 4m 13s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6aa618edda946c3a33ba9efb326888b : SUCCESS in 4m 23s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5eb4a43ebc94263a3ffc6d2c3cb38ad : SUCCESS in 32m 37s","accounts_in_message":[],"_revision_number":3},{"id":"fb89bd60bb097f4ff81ada860d142047c26f689d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 19:57:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d3dce7e81acdb1aaf02daa2d7fc5f8d09f4695d8","revisions":{"57b9a029685bd7a868526f9a7dd4e3371cc14e23":{"kind":"REWORK","_number":1,"created":"2022-09-05 15:42:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/855955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/855955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/855955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/855955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/855955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/855955/1"}}},"commit":{"parents":[{"commit":"e6ceaf6d51d2cdd6edc1d51e979e29e8ba51baa4","subject":"Merge \"Remove remaining hard-code of username in each manifest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e6ceaf6d51d2cdd6edc1d51e979e29e8ba51baa4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-05 15:31:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-05 15:41:23.000000000","tz":540},"subject":"CentOS: Use IP v6 globally in acceptance tests","message":"CentOS: Use IP v6 globally in acceptance tests\n\nCurrently in acceptance jobs with CentOS, we use IPv6 for the dependent\nservices like memcached but use IPv4 for swift services.\n\nThis change ensures we use IPv6 globally.\n\nChange-Id: I6092105916e3c49ad0e0a45156b43cce99cbe3ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/57b9a029685bd7a868526f9a7dd4e3371cc14e23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/57b9a029685bd7a868526f9a7dd4e3371cc14e23"}]},"branch":"refs/heads/master"},"1be525f92bce2ce84dd3d76c89eb8a449e88747b":{"kind":"REWORK","_number":2,"created":"2022-09-05 16:25:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/855955/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/855955/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/855955/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/855955/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/855955/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/855955/2"}}},"commit":{"parents":[{"commit":"e6ceaf6d51d2cdd6edc1d51e979e29e8ba51baa4","subject":"Merge \"Remove remaining hard-code of username in each manifest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e6ceaf6d51d2cdd6edc1d51e979e29e8ba51baa4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-05 15:31:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-05 16:25:25.000000000","tz":540},"subject":"CentOS: Use IP v6 globally in acceptance tests","message":"CentOS: Use IP v6 globally in acceptance tests\n\nCurrently in acceptance jobs with CentOS, we use IPv6 for the dependent\nservices like memcached but use IPv4 for swift services.\n\nThis change ensures we use IPv6 globally.\n\nChange-Id: I6092105916e3c49ad0e0a45156b43cce99cbe3ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1be525f92bce2ce84dd3d76c89eb8a449e88747b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1be525f92bce2ce84dd3d76c89eb8a449e88747b"}]},"branch":"refs/heads/master"},"d3dce7e81acdb1aaf02daa2d7fc5f8d09f4695d8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-09-05 17:50:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/855955/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/855955/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/855955/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/855955/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/855955/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/855955/3"}}},"commit":{"parents":[{"commit":"e6ceaf6d51d2cdd6edc1d51e979e29e8ba51baa4","subject":"Merge \"Remove remaining hard-code of username in each manifest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e6ceaf6d51d2cdd6edc1d51e979e29e8ba51baa4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-05 15:31:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-05 17:49:47.000000000","tz":540},"subject":"CentOS: Use consistent host parameters instead of hard-codes","message":"CentOS: Use consistent host parameters instead of hard-codes\n\nThis change replaces hard-coded host address by the one from the base\nconfig class, to make sure we use the consistent value.\n\nChange-Id: I6092105916e3c49ad0e0a45156b43cce99cbe3ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d3dce7e81acdb1aaf02daa2d7fc5f8d09f4695d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d3dce7e81acdb1aaf02daa2d7fc5f8d09f4695d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
