)]}'
{"id":"openstack%2Fpuppet-swift~859738","triplet_id":"openstack%2Fpuppet-swift~master~Ibc261c2e2ba9fa9ba547ad5d0f736c1b7230c579","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-09-29 22:34:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibc261c2e2ba9fa9ba547ad5d0f736c1b7230c579","subject":"authtoken: Remove deprecated auth_plugin","status":"MERGED","created":"2022-09-29 00:10:38.000000000","updated":"2022-09-29 22:35:43.000000000","submitted":"2022-09-29 22:34:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859738","meta_rev_id":"3c579be4fab5e4be2d0e09f5197b38a861786d65","_number":859738,"virtual_id_number":859738,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-29 22:34:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-09-29 20:07:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-09-29 20:31:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-09-29 20:31:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-29 01:35:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-29 20:07:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-09-29 20:31:40.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"50d0b60ced6c04576bee02cc394870a6cd15c2a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-29 00:10:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75f9ce547b502566181df6c5b4bdc22c46c8c838","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 01:35:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81d10df456c247059dcd197fa5ed9d20\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27e9341a7afa44699a9dc48c929c8ac9 : SUCCESS in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86978726d3784af2ac3ef0b46446f2b3 : SUCCESS in 3m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8557a7b3b82f4d6a8b0a3566cd766822 : SUCCESS in 3m 55s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e2e70c37ac144fd9f0ad0e0570234be : SUCCESS in 17m 49s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bc93529d19a459db8eba5cfee073023 : SUCCESS in 13m 04s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f89969b30e5a45f58f93a75317b85596 : SUCCESS in 13m 41s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/439fceb6441345a1aaec59e6134bf8e0 : SUCCESS in 9m 47s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f3e93969d7e4cd2ae44820065308eae : SUCCESS in 57m 49s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d6e54fcdb974a5e9f25612c9d126d88 : SUCCESS in 1h 23m 12s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d1fe7cf3aac4513a0516c5baaa23e0f : FAILURE in 27m 53s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d681ce01326b4f42a0ba78745148a813 : SUCCESS in 23m 46s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f79137ecc4764b2d9e020522337613d2 : FAILURE in 28m 22s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/209dfcda81204566b6051e8c54ffd3a2 : FAILURE in 19m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab1bc4a0d2a947e8a4b7712e27629461 : SUCCESS in 4m 20s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cc960f5e7c2c4161b60a0728a91924f8 : SUCCESS in 1h 14m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0dbe68e28507c9902a2d70f9b73472a2f7df23a0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-09-29 20:07:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b919214c1fe9db194ab5d0aa7a4ae9abb96a1537","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-29 20:31:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c364ff46ea5bf8a118185eabca61b9ebf896ca61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 20:33:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"228822a408ed0b681c8fbdf1afb0aabef1a8ee2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 22:34:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38cdf6048421405291f714fe5fffae54\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c623829aae6b4d2482a8afd41960d996 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a860d9ea91744d7e90bf52c5b49446ef : SUCCESS in 3m 28s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0aa0c1a323904586973dac6eabc29460 : SUCCESS in 3m 12s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/b68fbf1f50ad4daf9b2f485ce6a314b3 : SUCCESS in 16m 51s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecba8c08569945c8b9ef8912b07b0a92 : SUCCESS in 12m 27s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c617c8ef8950422983fd3224198aed30 : SUCCESS in 10m 48s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b9e355f013649f88a5195262ded6b6e : SUCCESS in 1h 20m 27s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5a06301db0240baafd037d575a57186 : SUCCESS in 31m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6595a7a32a54a928c64381746fca5e0 : SUCCESS in 3m 17s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a921947dbe9c4539a89b6c49d1a509e8 : SUCCESS in 1h 57m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d8a56564ad92ab6bef313f705aec617093039dc3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 22:34:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3c579be4fab5e4be2d0e09f5197b38a861786d65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 22:35:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3021c831c7d24747bf7c7d7e5124fd87\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eab1f3b0bb4b41e181008fa9d65b675f : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e821025b2c171991eb93dd88a1e68ca2e66d74f","revisions":{"7e821025b2c171991eb93dd88a1e68ca2e66d74f":{"kind":"REWORK","_number":1,"created":"2022-09-29 00:10:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/859738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/38/859738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/859738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/859738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/859738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/38/859738/1"}}},"commit":{"parents":[{"commit":"febcd8fa4494d6dda73753fa4346f78a95644541","subject":"Merge \"Manage internal-client.conf in acceptance tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/febcd8fa4494d6dda73753fa4346f78a95644541"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-29 00:09:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-29 00:10:27.000000000","tz":540},"subject":"authtoken: Remove deprecated auth_plugin","message":"authtoken: Remove deprecated auth_plugin\n\nThe parameter was deprecated during Victoria cycle[1] in favor of\nthe auth_type parameter.\n\n[1] 5722f7149c678fcb79cf792a9eda53537e080cb8\n\nChange-Id: Ibc261c2e2ba9fa9ba547ad5d0f736c1b7230c579\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7e821025b2c171991eb93dd88a1e68ca2e66d74f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7e821025b2c171991eb93dd88a1e68ca2e66d74f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
