)]}'
{"id":"openstack%2Fpuppet-swift~861823","triplet_id":"openstack%2Fpuppet-swift~master~I2aaeb86424aea09d285d7134a4b3c74c3c76715b","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2024-02-21 02:17:28.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2aaeb86424aea09d285d7134a4b3c74c3c76715b","subject":"Fix unmounted drive ownership","status":"ABANDONED","created":"2022-10-19 07:11:18.000000000","updated":"2024-02-21 02:17:28.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0db26451a714e2c731b8fe74c7f08ef8dcf42c6e","_number":861823,"virtual_id_number":861823,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-10-19 11:03:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2022-10-20 02:02:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-19 08:32:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-20 02:02:24.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"324237b9885c049fa20717f5555f60b04adbc2fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-10-19 07:11:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d52eee8d73ecc8143be1ca32a5c9f0abe347117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 08:32:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7095e9bfe14444bda745487258b5190c\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/914cb3d85560429995b63e73ea01118e : FAILURE in 3m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/283e6f425ef34f279f0c866fb93f1222 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7844593a2bab4902ba01a1a515e28db8 : SUCCESS in 3m 36s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a546e3f2633f4bd0ac0531073db68b90 : FAILURE in 12m 45s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf81a4a0c66145f785faf3775c1dad56 : FAILURE in 9m 33s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9674fa3bc1a64464ab6e369876957043 : FAILURE in 9m 51s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b4216e8d99343be9b98e9bf33095097 : FAILURE in 10m 31s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2ee9c812041499981d58af004afd47a : SUCCESS in 1h 12m 35s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e4c791a353f4afcb474325a152bcf35 : SUCCESS in 1h 13m 58s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53cdba9557de4652b1f9d5f76806a23a : FAILURE in 24m 45s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/b21aee142c714bb38dedb595c9325e23 : SUCCESS in 31m 13s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94b2ec9399be468aafac428c5e27cbb4 : FAILURE in 25m 57s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27f1ac4d2b2c45f0add053c54c6f5243 : FAILURE in 26m 19s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/889c94e2aa1a41c39dde58c28d58ffa5 : FAILURE in 16m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1566a44e53fbd4d74f075e0319bfc6a48165323a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-10-19 09:43:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3eb9981701fc93e555b1e1a3b2607d91b6aceb0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 11:03:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4af07c072134b838009bda63aa150b7\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d643e02fd934f9bb87758ab5ee684e2 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b038d3d99048440e8572a7c4f9b36aaf : SUCCESS in 3m 00s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3c62772876d4606b42b290edc36ba40 : SUCCESS in 4m 08s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c6456ac0332456596e701eec7ff6a4e : FAILURE in 13m 16s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/091d97f0c28243e69febfccfd4e51908 : FAILURE in 8m 38s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/497e2e0707404a5c8f1ac8d8a2174c69 : FAILURE in 10m 26s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82f56a8ff7164473805fbfdc83129abb : FAILURE in 9m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80f584d6f12f4dca9a1b4583c55ba3d4 : SUCCESS in 1h 07m 18s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8cf914304cc47049db2f660ad827baa : SUCCESS in 1h 14m 07s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adab27ea4be64856b01dd74029bee14e : FAILURE in 25m 14s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/3bf6e7a01f554214b2f8ba77b117007d : SUCCESS in 27m 31s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e7fe801bf4a4753bed5c4897fa7353e : FAILURE in 21m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86f2a3f2ca624ff49201c36c0bc0db18 : FAILURE in 22m 42s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3c65648d72c64eaaaafeb0e249f4aba2 : FAILURE in 20m 27s","accounts_in_message":[],"_revision_number":2},{"id":"37b667a37ce2c976da5768e739e945dfc7915670","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-20 02:02:24.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0db26451a714e2c731b8fe74c7f08ef8dcf42c6e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-21 02:17:28.000000000","message":"Abandoned\n\nI\u0027ll abandon this because of inactivity for some time. Feel free to reopen this in case you resume your work.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4691cde8364e8d573898cc14d566bdb93a7dee41","revisions":{"948d896a6934ab90e8adbbcbe20b6291028daa37":{"kind":"REWORK","_number":1,"created":"2022-10-19 07:11:18.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/23/861823/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/23/861823/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/23/861823/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/23/861823/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/23/861823/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/23/861823/1"}}},"commit":{"parents":[{"commit":"b39a938a62e33dacc59f0ec16c77231fd5f306ad","subject":"Prepare Zed RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b39a938a62e33dacc59f0ec16c77231fd5f306ad"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-10-19 07:03:10.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-10-19 07:11:16.000000000","tz":120},"subject":"Fix unmounted drive ownership","message":"Fix unmounted drive ownership\n\nCurrently, puppet-swift creates /srv/node/\u003cdrive\u003e with a swift:swift\nownership. As a result, when the swift-drive-auditor unmounts a broken\ndisk, it\u0027s ownership allows swift to write files, potentially filling\nup the system partition.\n\nThis patch adds a new fact \"swift_fstab_entries\" that is used to tell\nif the mount class was called in a previous run, in which case the\n/srv/node/\u003cdrive\u003e can be modified as swift:swift. If the fstab entry\nis commented, the /srv/node/\u003cdrive\u003e directory will be set to root:root,\ncorrecting the previous behavior of this puppet module.\n\nChange-Id: I2aaeb86424aea09d285d7134a4b3c74c3c76715b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/948d896a6934ab90e8adbbcbe20b6291028daa37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/948d896a6934ab90e8adbbcbe20b6291028daa37"}]},"branch":"refs/heads/master"},"4691cde8364e8d573898cc14d566bdb93a7dee41":{"kind":"REWORK","_number":2,"created":"2022-10-19 09:43:28.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/23/861823/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/23/861823/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/23/861823/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/23/861823/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/23/861823/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/23/861823/2"}}},"commit":{"parents":[{"commit":"b39a938a62e33dacc59f0ec16c77231fd5f306ad","subject":"Prepare Zed RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b39a938a62e33dacc59f0ec16c77231fd5f306ad"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-10-19 07:03:10.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-10-19 09:43:19.000000000","tz":120},"subject":"Fix unmounted drive ownership","message":"Fix unmounted drive ownership\n\nCurrently, puppet-swift creates /srv/node/\u003cdrive\u003e with a swift:swift\nownership. As a result, when the swift-drive-auditor unmounts a broken\ndisk, it\u0027s ownership allows swift to write files, potentially filling\nup the system partition.\n\nThis patch adds a new fact \"swift_fstab_entries\" that is used to tell\nif the mount class was called in a previous run, in which case the\n/srv/node/\u003cdrive\u003e can be modified as swift:swift. If the fstab entry\nis commented, the /srv/node/\u003cdrive\u003e directory will be set to root:root,\ncorrecting the previous behavior of this puppet module.\n\nChange-Id: I2aaeb86424aea09d285d7134a4b3c74c3c76715b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4691cde8364e8d573898cc14d566bdb93a7dee41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4691cde8364e8d573898cc14d566bdb93a7dee41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
