)]}'
{"id":"openstack%2Fpuppet-swift~861932","triplet_id":"openstack%2Fpuppet-swift~master~I6da122c569e7a66a1933b7483fb8cf68a40dd94f","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-11-08 08:27:52.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I6da122c569e7a66a1933b7483fb8cf68a40dd94f","subject":"Restrict access to the mount base directory","status":"MERGED","created":"2022-10-20 02:11:16.000000000","updated":"2022-11-08 12:07:23.000000000","submitted":"2022-11-08 12:07:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861932","meta_rev_id":"7d3b61ce067ae23349027f8d97dd6c1504af57f4","_number":861932,"virtual_id_number":861932,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-08 12:07:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-10-21 21:35:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-11-04 09:10:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-22 06:18:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-11-04 09:10:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-20 04:05:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-21 21:35:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-10-22 06:18:10.000000000","updated_by":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"reviewer":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"state":"REVIEWER"},{"updated":"2022-11-04 09:10:11.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"0eba3630aa6366976dfcae45323a88493befb0f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-20 02:11:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"652923579c067b7a3f19067c84f429eb0c3acaa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 04:05:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/440d29d001604ecf88a56675eaa3da05\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76a77c6190a040079541568db75d5ccb : SUCCESS in 2m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6391d13193f14d88bd4a022a5d0e457e : FAILURE in 2m 32s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4dd887c8be3640edae708300dd76456a : FAILURE in 3m 29s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6b16307a9834cf2ad5a55e1db62ce29 : FAILURE in 12m 59s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c99f7a07e7646229a8b48f0a7075c78 : FAILURE in 10m 17s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f76481e488344961af953e6937526217 : FAILURE in 12m 16s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59aff3f993f34ca9973255bab696e814 : FAILURE in 9m 00s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edd7a28553e44f228416d49bbd8c6c51 : SUCCESS in 1h 16m 10s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/38e5fe59591a4f069408dc8ba0f7f1c9 : SUCCESS in 1h 27m 54s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcda71178ba74ecbb8e79f7398b26eb5 : FAILURE in 6m 43s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/e49496a0447e42cc939fbf02898bfc97 : FAILURE in 10m 18s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5eac937f0368472a9ea7397ea9bd956a : FAILURE in 7m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/143a2aafa9624ab1b702a6f5061bb823 : FAILURE in 6m 57s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/079a3f86d3694844aa1d61b14c6f2c01 : SUCCESS in 1h 44m 54s","accounts_in_message":[],"_revision_number":1},{"id":"c04aaf283862924f6de3410b45f1be7043418873","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-20 04:09:11.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ee82630fc9a0f22b159b3991020fa12eea50c25d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 05:53:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d59a4e4a407f47a9b500113d0e6d7723\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a00f8bd5abea485487d03a619f1eb2b7 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68674aef5daa4ee2ae66fb4519f6ec18 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96165b2f32194a07935c626fc0f48eaf : SUCCESS in 2m 48s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5bb31ce727945bd924f0e8d2d0e6395 : SUCCESS in 13m 25s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09d055920e4747af9398558a0dfef624 : SUCCESS in 10m 47s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c0273ddc6f3477d8845361895bc2530 : SUCCESS in 9m 38s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54fbff5d599e4e559e71551a77d2a208 : SUCCESS in 9m 47s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc6a56443af94771a92359ad0e462d42 : SUCCESS in 54m 58s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c0d5c20f4a540b69d1320bacf906969 : SUCCESS in 1h 11m 55s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b60947a9b3cb4e5e857e2986d589b549 : FAILURE in 24m 49s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/1cf16d33b28b4d978dd893b7434b6433 : SUCCESS in 30m 47s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3c39e40e9d14b7995fc299d2ce73671 : FAILURE in 23m 34s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23d460f4359f4f699361c0eb2ac851de : FAILURE in 19m 20s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5426b7ba7f4e42a5b295fe3a38a2956b : SUCCESS in 1h 41m 40s","accounts_in_message":[],"_revision_number":2},{"id":"3dce503eba9b660ee34a5e0bd46cf28b74b562a7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-10-21 21:35:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d17e0dabf92d373760a35411d5b6f90d5600830a","author":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"date":"2022-10-22 06:18:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"284e445c749a44cfd488a1f725838b4cf316f52c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-11-04 09:10:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"679fdc33f452250e620c623797f6fa54d248e827","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 09:11:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fd1d5d92bc6de1acf567bcf707b20f24c635ab04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 10:53:10.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba8e194974654350a1457327d5ef537c\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a479340f5b4d4cd79e61fb929c8fce70 : SUCCESS in 2m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee262e2abd6c4098bd0a3c76015b1522 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0dc2b91db054b65aada5e868bcc2929 : SUCCESS in 2m 44s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8991868f37b49b185742a33d9c7127c : SUCCESS in 12m 50s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2a7991bf5c341048784875cad2f9a31 : SUCCESS in 9m 49s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcc70d20524242fd989b7cb8793d52a2 : SUCCESS in 9m 13s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/314291c5eb114c398d6dc0a62decc5b8 : SUCCESS in 1h 08m 55s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2dbd545c11d147a99da8e2226f7dd93b : SUCCESS in 27m 58s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a1587d7ef0cf4407b16c43a7a76f19ec : FAILURE in 1h 40m 46s","accounts_in_message":[],"_revision_number":2},{"id":"3f965ab8083403d1d90aa2eba5658540cbaa1fa4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-08 08:27:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9e515c3e918af6410311bb8419f6b520334526c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 10:44:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e577e354a2754f28ba1a3bce368e669c\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcf4e2d68ad04b45b7ac19a2520c7aee : SUCCESS in 2m 55s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/274f5a2b671c42f0bbd2fef1459c8aa1 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8cce07b5439402f82d33e53bfddb12d : SUCCESS in 2m 45s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a1c56003b0b472f91abe31c2389211a : SUCCESS in 14m 02s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/018b8d75eb58461b9f45c0b0a5c2ac53 : SUCCESS in 9m 42s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47c403703f73474ea12bbd21d9b3dc14 : SUCCESS in 10m 06s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d37cffbce5fe4addbdac8c75f5dca34a : SUCCESS in 10m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81ce6ec1878f4731a307fecc7f45b900 : SUCCESS in 58m 18s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/20ef680ebdfb4418885638836eabc50f : SUCCESS in 1h 16m 47s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b45aaf829d445ad8930e881776f7625 : FAILURE in 24m 40s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2879ef81d91466f8d150107da3f28c9 : SUCCESS in 28m 46s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9252d9ea0124abe990613524fdbcdf2 : FAILURE in 24m 57s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1713405ef874462e9b151786cb0db75c : FAILURE in 25m 11s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2aabe645779b404aa076b56ec3753ee4 : SUCCESS in 2h 06m 16s","accounts_in_message":[],"_revision_number":2},{"id":"e4f6d50cb5ced4099c0770f512cc3ca65c373d84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 10:45:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9023efd5c237b1d9ecf77d132bf6a216c88e0d15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 12:07:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9241719e44694511a5274d4a7b385e74\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0a7d18ce3204a59a0722e9f704b7981 : SUCCESS in 2m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/baf6730b43e342f9a61ee04a27baa26d : SUCCESS in 2m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07726a88a48f43be9daec64df0a72d7f : SUCCESS in 2m 31s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/328a5117d68d4a79a5863c986272cc72 : SUCCESS in 12m 05s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e889e2122f0410393d09b46c91bff7e : SUCCESS in 9m 55s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdbcb53cb29a4f949f49710cce3228f8 : SUCCESS in 9m 37s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/05b6a590582c4f189bb0f3eabcf2db38 : SUCCESS in 1h 09m 25s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a31316cd0b354b1cab07c2775f026357 : SUCCESS in 28m 55s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4b0b407f31834e72adecfdb558943c47 : SUCCESS in 1h 19m 21s","accounts_in_message":[],"_revision_number":2},{"id":"7d3b61ce067ae23349027f8d97dd6c1504af57f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 12:07:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"911919e5a5fca9bd1f8fe67f66aad51cf899822b","revisions":{"52fc70da378b309e8c18f4a4dea369e09ff7ebb2":{"kind":"REWORK","_number":1,"created":"2022-10-20 02:11:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/861932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/32/861932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/861932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/861932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/861932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/32/861932/1"}}},"commit":{"parents":[{"commit":"b39a938a62e33dacc59f0ec16c77231fd5f306ad","subject":"Prepare Zed RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b39a938a62e33dacc59f0ec16c77231fd5f306ad"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-20 02:08:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-20 02:08:56.000000000","tz":540},"subject":"Restrict access to the mount base directory","message":"Restrict access to the mount base directory\n\nThe swift processes does not require write access to the mount base\ndirectory (which is usually /srv/node), thus the directory can be owned\nby root. This is more consistent with the current installation guide\nof Swift.\n\nChange-Id: I6da122c569e7a66a1933b7483fb8cf68a40dd94f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/52fc70da378b309e8c18f4a4dea369e09ff7ebb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/52fc70da378b309e8c18f4a4dea369e09ff7ebb2"}]},"branch":"refs/heads/master"},"911919e5a5fca9bd1f8fe67f66aad51cf899822b":{"kind":"REWORK","_number":2,"created":"2022-10-20 04:09:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/861932/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/32/861932/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/861932/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/861932/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/861932/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/32/861932/2"}}},"commit":{"parents":[{"commit":"b39a938a62e33dacc59f0ec16c77231fd5f306ad","subject":"Prepare Zed RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b39a938a62e33dacc59f0ec16c77231fd5f306ad"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-20 02:08:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-20 04:09:02.000000000","tz":0},"subject":"Restrict access to the mount base directory","message":"Restrict access to the mount base directory\n\nThe swift processes does not require write access to the mount base\ndirectory (which is usually /srv/node), thus the directory can be owned\nby root. This is more consistent with the current installation guide\nof Swift.\n\nChange-Id: I6da122c569e7a66a1933b7483fb8cf68a40dd94f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/911919e5a5fca9bd1f8fe67f66aad51cf899822b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/911919e5a5fca9bd1f8fe67f66aad51cf899822b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
