)]}'
{"id":"openstack%2Fpuppet-swift~862300","triplet_id":"openstack%2Fpuppet-swift~master~Ie32d7cd12680ffb6e350a1f25309e92b7df698b2","project":"openstack/puppet-swift","branch":"master","topic":"ceilometermiddleware","attention_set":{},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2023-10-22 05:25:16.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie32d7cd12680ffb6e350a1f25309e92b7df698b2","subject":"Fix python-ceilometermiddleware -\u003e python3-ceilometermiddleware","status":"ABANDONED","created":"2022-10-21 12:39:30.000000000","updated":"2023-10-22 05:25:16.000000000","total_comment_count":4,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"98c4cbca1c9f527a0ffb3810c010bcf74733586b","_number":862300,"virtual_id_number":862300,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-11-03 15:37:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2022-11-08 08:24:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-21 14:00:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-21 18:23:58.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-11-08 08:24:16.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"20045d61360e859e29bc04acaf32c404c26c5dfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-10-21 12:39:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16460d5ec5932a5343b9a2c487c85ee2491abc58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 14:00:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/42a7b52f42d4428fbc567d83f7aefde0\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edf4a168decc41b886ba60215ec6f2ac : SUCCESS in 3m 47s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a46842a8837a40f9a48473e751ffe82d : SUCCESS in 3m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/893dc874c86f471bbad50221449ed1a0 : SUCCESS in 3m 55s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1146a7063954c0aa7f71b78588ff825 : SUCCESS in 9m 57s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f92e1c938774b22b984ccfaae9c387f : SUCCESS in 11m 05s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b14be6accf17498594f51b72b5915ea8 : SUCCESS in 13m 48s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7aa71861d4c24b4288ba7d1b9695e35e : SUCCESS in 8m 14s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/190d387f511243c6b76b6f21344df684 : SUCCESS in 1h 09m 36s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/393da14e28394f559b286dfb08d9421b : SUCCESS in 1h 10m 46s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ffacecc3d2840f0ba6b9671907f7e2a : FAILURE in 24m 44s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/b0dc40fa91aa42a5b5874d2958f666df : SUCCESS in 26m 17s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bb491587d3541ea9babfb579f22dfdc : FAILURE in 22m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ef6f02de726455eac9baad7c92e4563 : FAILURE in 18m 24s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6e978f9339ea4a9a87af024313aec36e : FAILURE in 20m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a5bb2527645bd2edc902d3eee28375d821b0a414","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-21 18:23:58.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c13b422e3f0477da0e04aa184d4710f669180dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-11-03 13:55:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2f393ab7c8a24d9922c4bb3b34b19311bc72ffd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 15:37:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3239f5598cc42118fa545f4fe39f799\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4bbfbfb81a8427081e69ddfd49a429a : SUCCESS in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90e495be207640a7b916b3f95148dc3e : SUCCESS in 3m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/905f95391ec34608bdf07eb6fea1401a : SUCCESS in 3m 30s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/b0bd4bf929fc4d309753901f178ef09b : FAILURE in 12m 37s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8392234518194307b95f2978ee5833ba : FAILURE in 10m 08s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/249a3559ff8b4ddaa680107611fbe6cd : FAILURE in 9m 21s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05a72a81481c4b81b13a4cfb1d777467 : FAILURE in 10m 14s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad5ff04d38ef43569d7a755068090a6a : SUCCESS in 1h 09m 35s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/07351be2aa03493c8143c4cd01233253 : SUCCESS in 1h 09m 04s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90646e42211142ce88c61ee448386c79 : FAILURE in 25m 08s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/66a1cf158c29455e99682b500c80d009 : SUCCESS in 26m 41s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25c37ced2043407c8c88e42405268611 : FAILURE in 23m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d8a419ab2194163b42da4a9743f5830 : FAILURE in 22m 17s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a306ca8e26514b0590e5a304ed8d854d : SUCCESS in 1h 39m 13s","accounts_in_message":[],"_revision_number":2},{"id":"99de1d13a7bcae808004ce2dcff43f89e4f96198","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-08 08:24:16.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"209129f819fb92ce11fcbcd13a83b2b87b071183","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-08 08:25:18.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"98c4cbca1c9f527a0ffb3810c010bcf74733586b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-22 05:25:16.000000000","message":"Abandoned\n\nI\u0027ll abandon this because of no feedback to the given review comments. As is described in a few comments we are not certain that the proposed change really resolves the problem. Feel free to reopen this in case the problem still persists in your env.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"71bc4e1d2c4965886ecf9b15846849d901a59d0b","revisions":{"65dc2bcab18d6122089c10065481942de97ee28e":{"kind":"REWORK","_number":1,"created":"2022-10-21 12:39:30.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/00/862300/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/00/862300/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/00/862300/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/00/862300/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/00/862300/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/00/862300/1"}}},"commit":{"parents":[{"commit":"2161ec46e9cff440baaca2182f155450caba02ea","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2161ec46e9cff440baaca2182f155450caba02ea"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-10-21 12:38:38.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-10-21 12:38:38.000000000","tz":120},"subject":"Fix python-ceilometermiddleware -\u003e python3-ceilometermiddleware","message":"Fix python-ceilometermiddleware -\u003e python3-ceilometermiddleware\n\nThere\u0027s a ::swift::params for it, but it\u0027s not in use. As a result,\nthis breaks the install of ceilometermiddleware. Let\u0027s use it.\n\nChange-Id: Ie32d7cd12680ffb6e350a1f25309e92b7df698b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/65dc2bcab18d6122089c10065481942de97ee28e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/65dc2bcab18d6122089c10065481942de97ee28e"}]},"branch":"refs/heads/master"},"71bc4e1d2c4965886ecf9b15846849d901a59d0b":{"kind":"REWORK","_number":2,"created":"2022-11-03 13:55:01.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/00/862300/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/00/862300/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/00/862300/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/00/862300/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/00/862300/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/00/862300/2"}}},"commit":{"parents":[{"commit":"2161ec46e9cff440baaca2182f155450caba02ea","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2161ec46e9cff440baaca2182f155450caba02ea"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-10-21 12:38:38.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-11-03 13:54:52.000000000","tz":60},"subject":"Fix python-ceilometermiddleware -\u003e python3-ceilometermiddleware","message":"Fix python-ceilometermiddleware -\u003e python3-ceilometermiddleware\n\nThere\u0027s a ::swift::params for it, but it\u0027s not in use. As a result,\nthis breaks the install of ceilometermiddleware. Let\u0027s use it.\n\nChange-Id: Ie32d7cd12680ffb6e350a1f25309e92b7df698b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/71bc4e1d2c4965886ecf9b15846849d901a59d0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/71bc4e1d2c4965886ecf9b15846849d901a59d0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
