)]}'
{"id":"openstack%2Fpuppet-swift~865094","triplet_id":"openstack%2Fpuppet-swift~master~I762f0780ba25536aa616482a49d59c7db69924d6","project":"openstack/puppet-swift","branch":"master","topic":"bug/1997313","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-11-25 15:58:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I762f0780ba25536aa616482a49d59c7db69924d6","subject":"Ensure consistent IPv6 address format","status":"MERGED","created":"2022-11-21 05:03:08.000000000","updated":"2022-11-25 15:58:57.000000000","submitted":"2022-11-25 15:58:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865094-bug/1997313","meta_rev_id":"613af64f86575d458e86c3cbb5df8a8eff8a4c0d","_number":865094,"virtual_id_number":865094,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-25 15:58:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-11-23 15:25:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-11-25 13:53:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-11-25 13:53:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-21 07:46:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-23 15:25:05.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-11-25 13:53:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"240e570113719bab124ee0c24c3053c458316371","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-21 05:03:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d83864ea66624d0203ef224cd009e2ef438fbb48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-21 05:56:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2126f265d544edadb7a809a742fc81de024218b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 07:46:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4880330d78044c07adcff5b12a8bf658\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05c36f78e83040e3a5f91004a272d643 : SUCCESS in 2m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d63c7a1c0f44f21b6223c89c28a456e : SUCCESS in 2m 50s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a30a233c97e441b8bbf7657e8c414a9e : SUCCESS in 2m 54s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/b34b00c4ebf14a3f83b72193ec19042b : FAILURE in 12m 40s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39c46ab97f90430fa2c1446eff596d45 : FAILURE in 10m 00s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/087ee8fda9344a188662a87bccdfd553 : FAILURE in 10m 01s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e67513db441841c783a4ffdfb09e7d27 : FAILURE in 9m 40s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28e8ce7768694ba6a0eabe54cea7bd8d : FAILURE in 6m 31s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e73954532ee74473a951aafe44c6c498 : FAILURE in 10m 49s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e53a4615b4f44d99b6825325ba8a2021 : FAILURE in 6m 41s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/85ff0248f2b845698cd208287c95fc57 : FAILURE in 9m 31s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71e4d10a4ad84a3da7a0689542013d18 : FAILURE in 6m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1722a95db9c94d8daec857973e8a0414 : FAILURE in 6m 44s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a2400b16a3e345e193778530bb455a35 : FAILURE in 1h 42m 37s","accounts_in_message":[],"_revision_number":2},{"id":"8e6fb6f79514d40505b88e97a6ddd1f9cc8fd08d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-21 08:01:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"795c76199c3aa18825c40941cd74cec766d12af2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-21 08:05:23.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"77c434e3482fd38da199c685826a117092c7acf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-21 09:04:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"77fc0ee07dac947c4074956a3a4155cf44b99124","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 10:35:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4ffa6478fcd4eed94ef5bb1fb333193\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ead369d54fb496e9d27edfd3638361f : SUCCESS in 3m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76a869aa089e4dc7aa1fd5d2a3d469a8 : SUCCESS in 2m 33s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e847583e0c134bd28d2f3b9208dd6bec : SUCCESS in 2m 42s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/4048336266b34a53a45d9c7defe607bc : FAILURE in 12m 05s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8b10b4b0c06400da8028c10c25fc740 : FAILURE in 10m 06s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db69b62237024ef0a6c803e8086ff4a5 : FAILURE in 10m 08s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dda202ee5ed48a2b46da56fbc382137 : FAILURE in 9m 19s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6074daaae14b41f0ba2d7de0459ff275 : FAILURE in 5m 45s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/32896425a8bf4108bed631bc94c1c2f6 : FAILURE in 11m 28s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d96eac0c4dd04c20bae9f9376785f5d4 : FAILURE in 22m 54s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea351c7617c94a9c863d1c0a3e0b8e03 : SUCCESS in 28m 20s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5a0ff8e99984f0d9bd4bfc61d6b5008 : FAILURE in 23m 37s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffd362228f4a4c6aa1ff91f10f485b28 : FAILURE in 24m 54s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/71b7312791ed4c9da7a8889d540f9386 : FAILURE in 1h 27m 40s","accounts_in_message":[],"_revision_number":5},{"id":"1a5f4a90833c309d703b25b48768f4f5ed0894a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-21 14:16:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"92d189ab2b51d533f75c29d921533366c53e7534","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 16:19:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0693cf2f763b4387a7886c29950e593f\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c112805311c14bdfa9176da7aab44943 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a13c9a0efadf40c6805e17dbcd995362 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c992d8ee8990497985da5a555f8aa467 : SUCCESS in 2m 32s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/58894eb73268416f8961a307b2a32efc : SUCCESS in 13m 37s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0886d601045045a6a6d6a38b09e0e9bf : SUCCESS in 9m 29s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78128ab5f47f4a5db9294231b38fe278 : SUCCESS in 11m 17s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c524bf47d784eafba75fb7e5048b3ba : SUCCESS in 10m 32s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f721e77ea2f4c268256437cec58f945 : SUCCESS in 58m 30s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b55a86c65ce7492da22bc66979e7e525 : FAILURE in 49m 14s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28f2e9837bfc4c8faec5b0a001588827 : FAILURE in 24m 38s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb03a32daf114860b578ee0704887bbd : SUCCESS in 27m 55s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37091a23634a446ca05ff9990485a03c : FAILURE in 24m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fa85960b8354ea8afbb945cb555caed : FAILURE in 23m 47s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ab9e843145bc42948d2ec7df6468fc6b : SUCCESS in 2h 01m 00s","accounts_in_message":[],"_revision_number":6},{"id":"21a7a76237aae1984f7c68f90d776a7c30e806d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-21 23:45:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"99c768734e5280f0b4124efb80f326d28ca6594a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 01:30:29.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/025f2dab63d2465ea29b6ded10334d27\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e089f3c7c21b47d1bbb702e703889844 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7056db14ffe94e3fb4497f8ce698c758 : SUCCESS in 2m 30s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/779fdb7105d64fb6ab8344b704464829 : SUCCESS in 2m 29s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/16957238afb14cf2b714b17c4316d74b : SUCCESS in 11m 33s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef9223b5fcf944d286a2a5b19c095885 : SUCCESS in 8m 09s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f289f388c4d4ff19c2d9f9356f4ef64 : SUCCESS in 7m 42s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db81980a5566459088e1dd46cc737fc8 : SUCCESS in 8m 45s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcc7f5906d71453cbe6cd09668fb588b : SUCCESS in 53m 38s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/29f39300747e4898a6065f3d2f2d21e5 : SUCCESS in 1h 02m 32s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24e0270a6dd24231a2c576c4697d6ccb : FAILURE in 22m 33s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/3fb1075892cb40889d3b100faea92951 : SUCCESS in 28m 05s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9529aefe9c84dcbbf7f66c1217cd0fc : FAILURE in 26m 11s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d4c90af87eb47e78ade54db64f946f5 : FAILURE in 26m 03s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8d1e2f806c384ddda0d599bcbe3bf0b2 : SUCCESS in 1h 33m 54s","accounts_in_message":[],"_revision_number":7},{"id":"8dec6e9bb0b3151821c0157de8e869e979320e8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-22 05:45:07.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"e589e5e96db3077c75bc805a4dfe4b0a38fd6e82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 07:56:57.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1440b49f301845b9b869dedf12f85f47\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afcdd65866294878899a7ac1f41d871c : SUCCESS in 3m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/087eaf9ac6a34d8580f701e0532a6ac5 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c1937538cca42e5b06cd4b6f94cea34 : SUCCESS in 3m 01s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a970d10d4a2e463cb00a4363c1cc9d84 : SUCCESS in 12m 22s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94b9722cca17474facfaf88ac142af3e : SUCCESS in 10m 01s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ab26e094a7f446282935ee7d9fef077 : SUCCESS in 9m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69e124c5891648828c232f44f6ae4c50 : SUCCESS in 10m 11s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cd009dcab30478984ccdcddb14345ab : SUCCESS in 59m 36s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/18b96c5e62ca4d8daf21ca6786f44ad6 : SUCCESS in 1h 12m 24s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c15c1bc6ffa44b05b68c0e22bfaa3e93 : FAILURE in 24m 18s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/72d15739921845369baef199b52474db : SUCCESS in 26m 51s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73c6cfc54210464aab3871f1e29ae3f0 : FAILURE in 24m 57s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62c6409bcf78401ea9da4e7192ac1c29 : FAILURE in 26m 29s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2ac842d06892485ca8030fc89912cfa3 : SUCCESS in 2h 06m 46s","accounts_in_message":[],"_revision_number":8},{"id":"02f82158d3fdb4050217fb19069ae91e045d52d8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-11-23 15:25:05.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"ee6ead62f72543fd33dd0590f2a594ff85431fc5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-11-25 13:53:23.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"2b75c4b8f8f483ba3335239b24c80c7bc4d68894","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 13:54:40.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"05863c5e5ea3e2cbcfc4b414eaa833acffdaa156","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 15:58:57.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c67c764f486749f8b7bff1c6e5fe40fd\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f30d801e8aaf4534bee66df80730f877 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea30da5072bb45c6a90a2b291d304129 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9691eb0a0048471ca728ed8587a516a4 : SUCCESS in 3m 49s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf5a49617d1141d482d57e89e07038b6 : SUCCESS in 14m 28s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f34daa91862a4970bd6e81efff824177 : SUCCESS in 11m 18s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84faf80db5ce4de98c4f1ba0d9ae8ed5 : SUCCESS in 9m 45s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0670f2815cfa4e5caf8d9e87f0f5b201 : SUCCESS in 1h 21m 53s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/46a62629050146a1a9375a3fd6cd3637 : SUCCESS in 27m 21s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/96dcd0e3e65441c98e831f8adf19c374 : SUCCESS in 1h 32m 44s","accounts_in_message":[],"_revision_number":8},{"id":"613af64f86575d458e86c3cbb5df8a8eff8a4c0d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 15:58:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"249f466f94441ae554478467949378d81b034f89","revisions":{"dd01a0a4d6066c37fc309808e38855f10405ab42":{"kind":"REWORK","_number":1,"created":"2022-11-21 05:03:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/865094/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/94/865094/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/1"}}},"commit":{"parents":[{"commit":"2161ec46e9cff440baaca2182f155450caba02ea","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2161ec46e9cff440baaca2182f155450caba02ea"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 04:55:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 05:02:02.000000000","tz":540},"subject":"WIP: Ensure consistent IPv6 format","message":"WIP: Ensure consistent IPv6 format\n\nThis change ensures the shortened address representations are always\nused for ring devices to avoid broken idempotency caused by different\nrepresentations (shortened vs canonical).\n\nChange-Id: I762f0780ba25536aa616482a49d59c7db69924d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/dd01a0a4d6066c37fc309808e38855f10405ab42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/dd01a0a4d6066c37fc309808e38855f10405ab42"}]},"branch":"refs/heads/master"},"1e221ce1e5b9e00a02cb0e34305fc172be98faa3":{"kind":"REWORK","_number":2,"created":"2022-11-21 05:56:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/865094/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/94/865094/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/2"}}},"commit":{"parents":[{"commit":"2161ec46e9cff440baaca2182f155450caba02ea","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2161ec46e9cff440baaca2182f155450caba02ea"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 04:55:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 05:56:47.000000000","tz":540},"subject":"WIP: Ensure consistent IPv6 format","message":"WIP: Ensure consistent IPv6 format\n\nThis change ensures the shortened address representations are always\nused for ring devices to avoid broken idempotency caused by different\nrepresentations (shortened vs canonical).\n\nChange-Id: I762f0780ba25536aa616482a49d59c7db69924d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1e221ce1e5b9e00a02cb0e34305fc172be98faa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1e221ce1e5b9e00a02cb0e34305fc172be98faa3"}]},"branch":"refs/heads/master"},"87716d9624466d51fbf7e4404b350ae5b09a48b1":{"kind":"REWORK","_number":3,"created":"2022-11-21 08:01:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/865094/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/94/865094/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/3"}}},"commit":{"parents":[{"commit":"2161ec46e9cff440baaca2182f155450caba02ea","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2161ec46e9cff440baaca2182f155450caba02ea"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 04:55:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 08:01:45.000000000","tz":540},"subject":"WIP: Ensure consistent IPv6 format","message":"WIP: Ensure consistent IPv6 format\n\nThis change ensures the shortened address representations are always\nused for ring devices to avoid broken idempotency caused by different\nrepresentations (shortened vs canonical).\n\nChange-Id: I762f0780ba25536aa616482a49d59c7db69924d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/87716d9624466d51fbf7e4404b350ae5b09a48b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/87716d9624466d51fbf7e4404b350ae5b09a48b1"}]},"branch":"refs/heads/master"},"4c13991ca8d439314d9c6f2a478cf54773dd4c5e":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-11-21 08:05:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/865094/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/94/865094/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/4"}}},"commit":{"parents":[{"commit":"2161ec46e9cff440baaca2182f155450caba02ea","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2161ec46e9cff440baaca2182f155450caba02ea"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 04:55:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 08:02:59.000000000","tz":540},"subject":"Ensure consistent IPv6 format","message":"Ensure consistent IPv6 format\n\nThis change ensures the shortened address representations are always\nused for ring devices to avoid broken idempotency caused by different\nrepresentations (shortened[1] vs canonical[2]).\n\n[1] ::1\n[2] 0000:0000:0000:0000:0000:0000:0000:0001\n\nChange-Id: I762f0780ba25536aa616482a49d59c7db69924d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4c13991ca8d439314d9c6f2a478cf54773dd4c5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4c13991ca8d439314d9c6f2a478cf54773dd4c5e"}]},"branch":"refs/heads/master"},"facb76cffa253987672da018567d876243444308":{"kind":"REWORK","_number":5,"created":"2022-11-21 09:04:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/865094/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/94/865094/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/5"}}},"commit":{"parents":[{"commit":"2161ec46e9cff440baaca2182f155450caba02ea","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2161ec46e9cff440baaca2182f155450caba02ea"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 04:55:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 09:03:23.000000000","tz":540},"subject":"Ensure consistent IPv6 format","message":"Ensure consistent IPv6 format\n\nThis change ensures the shortened address representations are always\nused for ring devices to avoid broken idempotency caused by different\nrepresentations (shortened[1] vs canonical[2]).\n\n[1] ::1\n[2] 0000:0000:0000:0000:0000:0000:0000:0001\n\nChange-Id: I762f0780ba25536aa616482a49d59c7db69924d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/facb76cffa253987672da018567d876243444308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/facb76cffa253987672da018567d876243444308"}]},"branch":"refs/heads/master"},"7d1f43239db1f5f2d01877c44607ac333024ce5c":{"kind":"REWORK","_number":6,"created":"2022-11-21 14:16:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/865094/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/94/865094/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/6"}}},"commit":{"parents":[{"commit":"2161ec46e9cff440baaca2182f155450caba02ea","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2161ec46e9cff440baaca2182f155450caba02ea"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 04:55:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 14:16:15.000000000","tz":540},"subject":"Ensure consistent IPv6 format","message":"Ensure consistent IPv6 format\n\nThis change ensures the shortened address representations are always\nused for ring devices to avoid broken idempotency caused by different\nrepresentations (shortened[1] vs canonical[2]).\n\n[1] ::1\n[2] 0000:0000:0000:0000:0000:0000:0000:0001\n\nChange-Id: I762f0780ba25536aa616482a49d59c7db69924d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7d1f43239db1f5f2d01877c44607ac333024ce5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7d1f43239db1f5f2d01877c44607ac333024ce5c"}]},"branch":"refs/heads/master"},"697362c0dd07f13386676e25a839f12f1f9e6caa":{"kind":"REWORK","_number":7,"created":"2022-11-21 23:45:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/865094/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/94/865094/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/7"}}},"commit":{"parents":[{"commit":"2161ec46e9cff440baaca2182f155450caba02ea","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2161ec46e9cff440baaca2182f155450caba02ea"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 04:55:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 23:45:35.000000000","tz":540},"subject":"Ensure consistent IPv6 format","message":"Ensure consistent IPv6 format\n\nThis change ensures the shortened address representations are always\nused for ring devices to avoid broken idempotency caused by different\nrepresentations (shortened[1] vs canonical[2]).\n\n[1] ::1\n[2] 0000:0000:0000:0000:0000:0000:0000:0001\n\nChange-Id: I762f0780ba25536aa616482a49d59c7db69924d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/697362c0dd07f13386676e25a839f12f1f9e6caa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/697362c0dd07f13386676e25a839f12f1f9e6caa"}]},"branch":"refs/heads/master"},"249f466f94441ae554478467949378d81b034f89":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2022-11-22 05:45:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/865094/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/94/865094/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/94/865094/8"}}},"commit":{"parents":[{"commit":"2161ec46e9cff440baaca2182f155450caba02ea","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2161ec46e9cff440baaca2182f155450caba02ea"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 04:55:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-22 05:44:48.000000000","tz":540},"subject":"Ensure consistent IPv6 address format","message":"Ensure consistent IPv6 address format\n\nThis change ensures the shortened address representations are always\nused for ring devices to avoid broken idempotency caused by different\nrepresentations (shortened[1] vs canonical[2]).\n\n[1] ::1\n[2] 0000:0000:0000:0000:0000:0000:0000:0001\n\nCloses-Bug: #1997313\nChange-Id: I762f0780ba25536aa616482a49d59c7db69924d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/249f466f94441ae554478467949378d81b034f89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/249f466f94441ae554478467949378d81b034f89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
