)]}'
{"id":"openstack%2Fpuppet-swift~865215","triplet_id":"openstack%2Fpuppet-swift~master~I390f3c7bbfbbbc7217f81e8a0312e5db201ef409","project":"openstack/puppet-swift","branch":"master","topic":"bug/1997295","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-11-25 15:58:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I390f3c7bbfbbbc7217f81e8a0312e5db201ef409","subject":"Format IPv6 address for ring device","status":"MERGED","created":"2022-11-22 02:14:33.000000000","updated":"2022-11-25 15:58:56.000000000","submitted":"2022-11-25 15:58:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865215-bug/1997295","meta_rev_id":"31a5bb9ebb00385f2b7ae29cde78daef3f36bdd2","_number":865215,"virtual_id_number":865215,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-25 15:58:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-11-23 15:25:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-11-25 13:52:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-11-25 13:52:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-22 05:09:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-23 15:25:14.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-11-25 13:52:54.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"c3ee8134a4e9f3f1772be224111d43e447f62b0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-22 02:14:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89ff173ad611cd25f63e3d79b6445892c91d7472","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-22 02:24:44.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"03db5de6dab396fac140eddbd1b5e2db9d595d3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-22 02:33:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"82f71bc585f5cba31396c6750c8b7645c3c522d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 05:09:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0df8a33221f240709586ec6f75eec175\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8aa1360c7bf34746850107ec37f2341c : SUCCESS in 3m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27256978c07c4553b8f580117f3a1483 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef4d5b47d9f747baa02d155fedb998b6 : SUCCESS in 3m 48s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/cad6025e5f1044868871af4fde36733c : FAILURE in 10m 30s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/418fe52cf6924d0b8ce9568ceeed2da8 : FAILURE in 11m 07s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e4801f905b04afa8ed557f4ab79d226 : FAILURE in 13m 06s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16ffa01206a548689f23ba2e57e8dbd0 : FAILURE in 10m 15s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90d1aedfb6ac49448652fa9d1fa9a3ec : SUCCESS in 1h 07m 00s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b0fae96a0de14210a55ab68bece522b7 : SUCCESS in 1h 15m 40s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad291c61a0474ad996588e44c07aef3d : FAILURE in 22m 57s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/feeeaff7928e4e43a378dbcb83591385 : SUCCESS in 27m 12s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03feec00783645dabbac418122807f8f : FAILURE in 22m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f67f058aeb3f47fdab387dbc73536f0c : FAILURE in 22m 57s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8c778a91da0543198ca6fdc599eda415 : SUCCESS in 2h 27m 18s","accounts_in_message":[],"_revision_number":3},{"id":"0687c26de2addd2778e3b2fbe02a4167aacafc6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-22 05:52:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4ebc2afcf098818c494564db0578ebea50a3b4e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 07:32:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da009af0e75e47048f008886134dbf98\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d734c390976a4862ac4b31820eb6a0cf : SUCCESS in 3m 17s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d48f948353594ee986eb094a90ce2071 : SUCCESS in 3m 01s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf38d463b2284a6e951a4b93a581b029 : SUCCESS in 2m 46s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/b15aa02bdc9d4f0a99a6f2c23f854e54 : SUCCESS in 14m 13s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb5e2d854d1b4eddb44cb22f870d192d : SUCCESS in 9m 41s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8dc341a3771749f4a856f9dd11158ab7 : SUCCESS in 8m 23s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e364a1f6666422c96d159e33407cb72 : SUCCESS in 11m 02s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a58a0c0af74471784b6b2dda55c7ed8 : SUCCESS in 57m 42s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/44188294c62a4148acbbd4f0426b9bb5 : SUCCESS in 1h 11m 30s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26d0da821da043d5b56f92a30e503dba : FAILURE in 20m 13s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ab3fda133f045a792d298faea1a2ccc : SUCCESS in 27m 23s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a04497bcae434028a4f2bc33ecce9e63 : FAILURE in 24m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abe7fc27ea6d4b6989d99b53aec38959 : FAILURE in 23m 39s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3f873f3b1015408181ac4dbec05e0177 : SUCCESS in 1h 37m 42s","accounts_in_message":[],"_revision_number":4},{"id":"6f49e82e5b87315c3e8878f828f803bc6768ce61","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-11-23 15:25:14.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"41d705b4e3f596ed2f5e87d3e5b7aa66c2c4c9c6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-11-25 13:52:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"18d986bd03c02d38242e1e5803cf55c4689330df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 13:54:09.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d4a52fe4f574641c29d23c135231c719e95cc06c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 15:58:56.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64ceeeb37bb94a0c891d8b9360e29a02\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc0f95b6d9ed4270b0910aa1b55262bf : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61dfe918490344f8ad9e7c20f9cc6d6f : SUCCESS in 3m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f54386b4c5947c6ba85c342a7118d4d : SUCCESS in 3m 58s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9575d8e390f4680ae654953b546bf1e : SUCCESS in 11m 22s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06bdfd1d8e58433e8b3b46538fb8d1a3 : SUCCESS in 11m 07s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85439820c0d5410da9bf99f1d89db5b7 : SUCCESS in 9m 32s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9c3025a9e12443fb63d9796ab5a3aed : SUCCESS in 1h 18m 34s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/46b469504cf544a1adfaaaa421ec3f5e : SUCCESS in 26m 14s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/13c2a362352c451188ca01cf758e8933 : SUCCESS in 1h 58m 09s","accounts_in_message":[],"_revision_number":4},{"id":"31a5bb9ebb00385f2b7ae29cde78daef3f36bdd2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 15:58:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a8ef5f69060adda86433ff2de4c761a0f86f68c3","revisions":{"4738e82b5af5079a0040f7366d5d22e0b3397937":{"kind":"REWORK","_number":1,"created":"2022-11-22 02:14:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/865215/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/15/865215/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/15/865215/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/15/865215/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/15/865215/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/15/865215/1"}}},"commit":{"parents":[{"commit":"0dcd3afa091821d608f6a9f5aa8ae17157085e12","subject":"Merge \"Restrict access to the mount base directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0dcd3afa091821d608f6a9f5aa8ae17157085e12"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-22 02:13:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-22 02:13:16.000000000","tz":540},"subject":"Format IPv6 address for ring device","message":"Format IPv6 address for ring device\n\nThe IPv6 addresses in ring_*_device resource names should be surrounded\nby []. This ensures the address is properly formatted.\n\nChange-Id: I390f3c7bbfbbbc7217f81e8a0312e5db201ef409\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4738e82b5af5079a0040f7366d5d22e0b3397937"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4738e82b5af5079a0040f7366d5d22e0b3397937"}]},"branch":"refs/heads/master"},"5e5acfb37080714211afcc0ad687c534a0fe375d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-22 02:24:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/865215/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/15/865215/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/15/865215/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/15/865215/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/15/865215/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/15/865215/2"}}},"commit":{"parents":[{"commit":"0dcd3afa091821d608f6a9f5aa8ae17157085e12","subject":"Merge \"Restrict access to the mount base directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0dcd3afa091821d608f6a9f5aa8ae17157085e12"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-22 02:13:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-22 02:24:25.000000000","tz":540},"subject":"Format IPv6 address for ring device","message":"Format IPv6 address for ring device\n\nThe IPv6 addresses in ring_*_device resource names should be surrounded\nby []. This ensures the address is properly formatted when the resource\nis define in swift::storage::node.\n\nCloses-Bug: #1997295\nChange-Id: I390f3c7bbfbbbc7217f81e8a0312e5db201ef409\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5e5acfb37080714211afcc0ad687c534a0fe375d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5e5acfb37080714211afcc0ad687c534a0fe375d"}]},"branch":"refs/heads/master"},"1779f6cc547b53733627734ab04f1f87c3b973ae":{"kind":"REWORK","_number":3,"created":"2022-11-22 02:33:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/865215/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/15/865215/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/15/865215/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/15/865215/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/15/865215/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/15/865215/3"}}},"commit":{"parents":[{"commit":"0dcd3afa091821d608f6a9f5aa8ae17157085e12","subject":"Merge \"Restrict access to the mount base directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0dcd3afa091821d608f6a9f5aa8ae17157085e12"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-22 02:13:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-22 02:32:28.000000000","tz":540},"subject":"Format IPv6 address for ring device","message":"Format IPv6 address for ring device\n\nThe IPv6 addresses in ring_*_device resource names should be surrounded\nby []. This ensures the address is properly formatted when the resource\nis define in swift::storage::node.\n\nCloses-Bug: #1997295\nChange-Id: I390f3c7bbfbbbc7217f81e8a0312e5db201ef409\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1779f6cc547b53733627734ab04f1f87c3b973ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1779f6cc547b53733627734ab04f1f87c3b973ae"}]},"branch":"refs/heads/master"},"a8ef5f69060adda86433ff2de4c761a0f86f68c3":{"kind":"REWORK","_number":4,"created":"2022-11-22 05:52:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/865215/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/15/865215/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/15/865215/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/15/865215/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/15/865215/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/15/865215/4"}}},"commit":{"parents":[{"commit":"0dcd3afa091821d608f6a9f5aa8ae17157085e12","subject":"Merge \"Restrict access to the mount base directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0dcd3afa091821d608f6a9f5aa8ae17157085e12"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-22 02:13:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-22 05:52:42.000000000","tz":540},"subject":"Format IPv6 address for ring device","message":"Format IPv6 address for ring device\n\nThe IPv6 addresses in ring_*_device resource names should be surrounded\nby []. This ensures the address is properly formatted when the resource\nis define in swift::storage::node.\n\nCloses-Bug: #1997295\nChange-Id: I390f3c7bbfbbbc7217f81e8a0312e5db201ef409\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a8ef5f69060adda86433ff2de4c761a0f86f68c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a8ef5f69060adda86433ff2de4c761a0f86f68c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
