)]}'
{"id":"openstack%2Fpuppet-swift~865451","triplet_id":"openstack%2Fpuppet-swift~master~I694ca4c046f490e4ffecf2101f3f147a69045162","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-11-29 18:52:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I694ca4c046f490e4ffecf2101f3f147a69045162","subject":"Remove logic for old swift-ring-builder","status":"MERGED","created":"2022-11-23 17:18:39.000000000","updated":"2022-11-29 18:52:11.000000000","submitted":"2022-11-29 18:52:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865451","meta_rev_id":"f7623d5f0fb2cdf7910c80ed185cbc147e90b9c3","_number":865451,"virtual_id_number":865451,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-29 18:52:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-11-25 13:52:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-11-29 17:04:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-11-29 17:04:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-23 19:08:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-25 13:52:27.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-11-29 17:04:57.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"e29a1bf2141539f86429785468ab7458627e8e20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-23 17:18:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd2d56110c86f0ce03375c3624148102628c5387","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-23 17:20:44.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1026496f11be139bb92d599b0f5db76267ca45b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-23 17:32:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3d572c1743dcdd7dd4c6737f91a5dbd607d58d9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 19:08:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47671c2b08274636a1ea7aba1870327d\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7af0ed26e234e079453c254ceb706ff : SUCCESS in 4m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb94a1e800ed43f6b3bcdc55e8c4fbe6 : SUCCESS in 3m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f36fffb49c114f5c9cb43701ac925d04 : SUCCESS in 3m 51s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd74b458fdfa4999b1f9ad643501e168 : FAILURE in 13m 10s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ab908a50b8c4fd59e3aa920798b6b70 : FAILURE in 10m 52s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0c082560c0b4d228676daa40e64e55f : FAILURE in 10m 41s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6b545e48ccb4c1bbeb7a096694794a3 : FAILURE in 11m 00s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb7c371838cb4073b778bbd1505f7244 : SUCCESS in 58m 15s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0153cdd83234fcd8cb588889a213a1e : SUCCESS in 1h 09m 05s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7bc22d33435430cb7fcfad47374209a : FAILURE in 25m 24s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/73a1b651ef8744bea624d117ce2c607d : SUCCESS in 28m 57s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87d3867f752e4a7d891123c580eef27d : FAILURE in 24m 06s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb1e77933b6841fe80b93d5bea0f3cb4 : FAILURE in 24m 31s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a779f58d187745148c081d93c4c4bac0 : SUCCESS in 1h 32m 10s","accounts_in_message":[],"_revision_number":3},{"id":"f0cfaaa0d9294f7bb1c9af46a563208bb2e2b3db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-23 19:17:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d7b4602292d6db1ed0fa814cea358c29382baf09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-23 19:19:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"944c969d6e8ff0bd2626701d083d0a8d1c0b9cad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-23 19:20:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e9ad5e26496fa8288ecac7d79d7afe25697f0d83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-23 19:20:39.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"fe70838ff376b072fb541a421a231b80b029da8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-23 19:23:34.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"6e77089987798a6016ad36a98bfff58e9cae2035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 21:08:54.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/489e245666af4fad83aac39e3b52a3ee\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ad774ba44df4ddc91d16037bc73f83f : SUCCESS in 2m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e42c72364f734b1eb0839fcf66ee8346 : SUCCESS in 3m 32s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb58a04c11094787a4cddce06fd2bab3 : SUCCESS in 3m 06s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e0daf62b2424e63b1cc2c01474bdd92 : SUCCESS in 12m 55s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fde7d7883e414d91b0d90d143cdfbd9e : SUCCESS in 9m 26s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dce95981a8dc47c3984f5963f1eb39bc : SUCCESS in 10m 13s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35f52739073741cbbc1d734d9f7a4da5 : SUCCESS in 9m 20s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e215ab17724c4a9f8759ccb36c3146e2 : FAILURE in 17m 55s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b013a80e9164a88abe7a5c6ee14a59e : SUCCESS in 1h 14m 20s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/705f0800f7ce4bb5846025ca684b764e : FAILURE in 9m 57s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1145bedcbc9469b9bf450964a8491b0 : SUCCESS in 24m 19s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d65bc1d59dc449799ff17728b261427b : FAILURE in 50m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9334c10f4e684847a33c1ff94dd93a2d : FAILURE in 47m 15s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7d55450b33014c6bb8d809f400c78e28 : SUCCESS in 1h 42m 28s","accounts_in_message":[],"_revision_number":8},{"id":"5959e1c75c1953916847d0d12c31c71467efd5e5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-11-25 13:52:27.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"69b9ea5823281619c08d9eb0b3f4b4ba05881eaa","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-11-29 17:04:57.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"3942267f9e27c722bc2f1c45e3180a32d3b14784","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 17:06:18.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"5e2880bce785c48eb7130be9dd8aecf78086f51c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 18:52:11.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8089672d71994db9ac9207f81c9924a0\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/925cc8560b6f405d8e32d58e3e01889d : SUCCESS in 2m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b5125f4ebf74fcc8f87e12e03283673 : SUCCESS in 3m 33s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b9800d5955f42319c6fb9d143896747 : SUCCESS in 2m 43s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3acdbc1753d4c828258e32b1fff54c2 : SUCCESS in 12m 14s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc16a890ac9f441eb72551951e55b96c : SUCCESS in 10m 56s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/345f19e39a5149b28031ddc50665aa93 : SUCCESS in 10m 28s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/99377413cb7049538740aeb41d556168 : SUCCESS in 1h 12m 47s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b72b198ba0d4a0a941a2e570ca8d486 : SUCCESS in 27m 36s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c124d0d53cf441d9ac2f38026ea60670 : SUCCESS in 1h 40m 32s","accounts_in_message":[],"_revision_number":8},{"id":"f7623d5f0fb2cdf7910c80ed185cbc147e90b9c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 18:52:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"8ac0db49280b7c4451bff6a3e66c13d43a167b12","revisions":{"71c4ca62d115d4b16a0ca5ae2b970090ce93aa28":{"kind":"REWORK","_number":1,"created":"2022-11-23 17:18:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/865451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/865451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/1"}}},"commit":{"parents":[{"commit":"0dcd3afa091821d608f6a9f5aa8ae17157085e12","subject":"Merge \"Restrict access to the mount base directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0dcd3afa091821d608f6a9f5aa8ae17157085e12"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 17:11:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 17:16:49.000000000","tz":540},"subject":"Remove logic for old swift releases","message":"Remove logic for old swift releases\n\nswift 2.10.0 was released on 2016. We no longer need to support even\nolder versions.\n\nChange-Id: I694ca4c046f490e4ffecf2101f3f147a69045162\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/71c4ca62d115d4b16a0ca5ae2b970090ce93aa28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/71c4ca62d115d4b16a0ca5ae2b970090ce93aa28"}]},"branch":"refs/heads/master"},"a8ffacd2d22a7d9ac165a2979356b9e22930d3ea":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-23 17:20:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/865451/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/865451/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/2"}}},"commit":{"parents":[{"commit":"0dcd3afa091821d608f6a9f5aa8ae17157085e12","subject":"Merge \"Restrict access to the mount base directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0dcd3afa091821d608f6a9f5aa8ae17157085e12"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 17:11:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 17:20:28.000000000","tz":540},"subject":"Remove logic for old swift-ring-builder","message":"Remove logic for old swift-ring-builder\n\nswift 2.10.0 was released on 2016. We no longer need to support even\nolder versions.\n\nChange-Id: I694ca4c046f490e4ffecf2101f3f147a69045162\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a8ffacd2d22a7d9ac165a2979356b9e22930d3ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a8ffacd2d22a7d9ac165a2979356b9e22930d3ea"}]},"branch":"refs/heads/master"},"fb584cfa4c5eb7eb11ce2a495992112f81b89507":{"kind":"REWORK","_number":3,"created":"2022-11-23 17:32:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/865451/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/865451/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/3"}}},"commit":{"parents":[{"commit":"0dcd3afa091821d608f6a9f5aa8ae17157085e12","subject":"Merge \"Restrict access to the mount base directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0dcd3afa091821d608f6a9f5aa8ae17157085e12"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 17:11:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 17:32:07.000000000","tz":540},"subject":"Remove logic for old swift-ring-builder","message":"Remove logic for old swift-ring-builder\n\nswift 2.10.0 was released in 2016. We no longer need to support even\nolder versions.\n\nChange-Id: I694ca4c046f490e4ffecf2101f3f147a69045162\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fb584cfa4c5eb7eb11ce2a495992112f81b89507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fb584cfa4c5eb7eb11ce2a495992112f81b89507"}]},"branch":"refs/heads/master"},"cdd8b28f0ce672859bbfabae140325a7002a5b38":{"kind":"REWORK","_number":4,"created":"2022-11-23 19:17:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/865451/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/865451/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/4"}}},"commit":{"parents":[{"commit":"0dcd3afa091821d608f6a9f5aa8ae17157085e12","subject":"Merge \"Restrict access to the mount base directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0dcd3afa091821d608f6a9f5aa8ae17157085e12"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 17:11:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 19:17:16.000000000","tz":540},"subject":"Remove logic for old swift-ring-builder","message":"Remove logic for old swift-ring-builder\n\nswift 2.10.0 was released in 2016. We no longer need to support even\nolder versions.\n\nChange-Id: I694ca4c046f490e4ffecf2101f3f147a69045162\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cdd8b28f0ce672859bbfabae140325a7002a5b38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cdd8b28f0ce672859bbfabae140325a7002a5b38"}]},"branch":"refs/heads/master"},"179ecf4020412b86cb73cd84ce1d22af18fa1035":{"kind":"REWORK","_number":5,"created":"2022-11-23 19:19:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/865451/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/865451/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/5"}}},"commit":{"parents":[{"commit":"0dcd3afa091821d608f6a9f5aa8ae17157085e12","subject":"Merge \"Restrict access to the mount base directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0dcd3afa091821d608f6a9f5aa8ae17157085e12"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 17:11:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 19:19:06.000000000","tz":540},"subject":"Remove logic for old swift-ring-builder","message":"Remove logic for old swift-ring-builder\n\nswift 2.10.0 was released in 2016. We no longer need to support even\nolder versions.\n\nThis also updates the logic to skip the header lines, to make it more\nrobust. The new logic works even when any additional lines are added.\n\nChange-Id: I694ca4c046f490e4ffecf2101f3f147a69045162\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/179ecf4020412b86cb73cd84ce1d22af18fa1035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/179ecf4020412b86cb73cd84ce1d22af18fa1035"}]},"branch":"refs/heads/master"},"7bde933eb50e2053b477c15d2229a6703d8c3a84":{"kind":"REWORK","_number":6,"created":"2022-11-23 19:20:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/865451/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/865451/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/6"}}},"commit":{"parents":[{"commit":"0dcd3afa091821d608f6a9f5aa8ae17157085e12","subject":"Merge \"Restrict access to the mount base directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0dcd3afa091821d608f6a9f5aa8ae17157085e12"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 17:11:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 19:20:08.000000000","tz":540},"subject":"Remove logic for old swift-ring-builder","message":"Remove logic for old swift-ring-builder\n\nswift 2.10.0 was released in 2016. We no longer need to support even\nolder versions.\n\nThis also updates the logic to skip the header lines, to make it more\nrobust. The new logic works even when any additional lines are added.\n\nChange-Id: I694ca4c046f490e4ffecf2101f3f147a69045162\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7bde933eb50e2053b477c15d2229a6703d8c3a84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7bde933eb50e2053b477c15d2229a6703d8c3a84"}]},"branch":"refs/heads/master"},"94717f9b4405d67be15b2c1e5195c663a079b22e":{"kind":"REWORK","_number":7,"created":"2022-11-23 19:20:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/865451/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/865451/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/7"}}},"commit":{"parents":[{"commit":"0dcd3afa091821d608f6a9f5aa8ae17157085e12","subject":"Merge \"Restrict access to the mount base directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0dcd3afa091821d608f6a9f5aa8ae17157085e12"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 17:11:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 19:20:31.000000000","tz":540},"subject":"Remove logic for old swift-ring-builder","message":"Remove logic for old swift-ring-builder\n\nswift 2.10.0 was released in 2016. We no longer need to support even\nolder versions.\n\nThis also updates the logic to skip the header lines, to make it more\nrobust. The new logic works even when any additional lines are added.\n\nChange-Id: I694ca4c046f490e4ffecf2101f3f147a69045162\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/94717f9b4405d67be15b2c1e5195c663a079b22e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/94717f9b4405d67be15b2c1e5195c663a079b22e"}]},"branch":"refs/heads/master"},"8ac0db49280b7c4451bff6a3e66c13d43a167b12":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2022-11-23 19:23:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/865451/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/865451/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/865451/8"}}},"commit":{"parents":[{"commit":"0dcd3afa091821d608f6a9f5aa8ae17157085e12","subject":"Merge \"Restrict access to the mount base directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0dcd3afa091821d608f6a9f5aa8ae17157085e12"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 17:11:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-23 19:23:03.000000000","tz":540},"subject":"Remove logic for old swift-ring-builder","message":"Remove logic for old swift-ring-builder\n\nswift 2.10.0 was released in 2016. We no longer need to support even\nolder versions. The output format depends on the version currently\ninstalled, thus the latest logic should work as long as the same\nrelease (or even a bit older release) is installed.\n\nThis also updates the logic to skip the header lines. The new logic\nis considered to be more robust, because it works even when additional\nlines are added.\n\nChange-Id: I694ca4c046f490e4ffecf2101f3f147a69045162\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8ac0db49280b7c4451bff6a3e66c13d43a167b12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8ac0db49280b7c4451bff6a3e66c13d43a167b12"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
