)]}'
{"id":"openstack%2Fpuppet-swift~866000","triplet_id":"openstack%2Fpuppet-swift~stable%2Fxena~I762f0780ba25536aa616482a49d59c7db69924d6","project":"openstack/puppet-swift","branch":"stable/xena","topic":"bug/1997313","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-12-12 01:59:52.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I762f0780ba25536aa616482a49d59c7db69924d6","subject":"Ensure consistent IPv6 address format","status":"MERGED","created":"2022-11-30 09:26:00.000000000","updated":"2022-12-12 04:30:46.000000000","submitted":"2022-12-12 04:30:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866000-bug/1997313","cherry_pick_of_change":866001,"cherry_pick_of_patch_set":1,"meta_rev_id":"067307f700e11a696607bfb4fbceff1a0ed6db1a","_number":866000,"virtual_id_number":866000,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-12 04:30:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-12-01 16:22:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-12-02 08:25:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-12-02 08:25:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-30 10:31:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-01 16:22:59.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-12-02 08:25:58.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"d7f35ed8cb7d6e7d8673dab2a1907a85464e6458","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-30 09:26:00.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"087c6cd41853f65506e21449b5f0d91058b43a1d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-30 09:26:05.000000000","message":"Topic bug/1997313-stable/xena removed","accounts_in_message":[],"_revision_number":1},{"id":"1e85464571a1f1481b4e532729040c2b30b121e3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-30 09:26:07.000000000","message":"Topic set to bug/1997313","accounts_in_message":[],"_revision_number":1},{"id":"a466b050edfa5afa1ff101605f968d52d6abd368","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-30 09:26:21.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36555b07c46011fb73566dd6a31319883f2f87a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-30 09:26:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a365f9daf1ea2f8ad313787d9c4848623af62b62","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-30 09:27:27.000000000","message":"Topic bug/1997313-stable/xena removed","accounts_in_message":[],"_revision_number":2},{"id":"6facaf718421b36168e4b145b3ec19bd3c0105b5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-30 09:27:30.000000000","message":"Topic set to bug/1997313","accounts_in_message":[],"_revision_number":2},{"id":"ab46747f5e39d0e687a0e221bd2f834a8a55f1f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 10:31:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d00d555b49bb4bfd984dc3dab2352054\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0f61167c9ca48c8a15245b4718c87c4 : SUCCESS in 3m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd3a1171cfff4772aba18083cf610e4d : SUCCESS in 3m 33s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c90d43035094d3990756ecd1925085a : SUCCESS in 3m 16s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d55ea3cdacf14bf489cf15d976786e68 : SUCCESS in 13m 57s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/343bcb07db5b440c9f1451b7aaa95aa1 : SUCCESS in 14m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81a40033bf7b440b99f35cc1b0c9215d : SUCCESS in 13m 06s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6e0fa028c244171b44e0b8707bd2006 : SUCCESS in 13m 43s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/046494c6f4c04b42a98d3fefa802ac1a : SUCCESS in 13m 01s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86e5ddcea3eb4e25a232c50b79c0675c : SUCCESS in 50m 01s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8058d3547424445c81f82eaa97f68a31 : SUCCESS in 1h 01m 50s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/348c5ff5328d41abbe783c0e3e486114 : FAILURE in 23m 06s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/d22e72f2b0e04ac6b37a32e49b94202c : SUCCESS in 30m 07s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85d0e29be1d44144a8a109a9a910ef55 : FAILURE in 23m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5613e55dfc6447da1010d395727cf09 : FAILURE in 21m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2a4177e7416d08cc6e6563b4324af28ff1ff3d54","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-12-01 16:22:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a446c616c7d2f415669e02430df94d13c55140f5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-12-02 08:25:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"600b6ee66d091bd3d559a60bbd4808d36121dd8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 08:27:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aeb7f4900cffed309b896ea01c424234e30a2a4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 09:43:37.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/738f90b9b7214e85a94f9513f4282eb9\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff08a37421b24acdaebf4ab735403334 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57d1f1b262f74e809dfff6d0fea91cc6 : SUCCESS in 3m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45f12840aabb48aabdabee6549884124 : SUCCESS in 2m 21s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4063f0c3a8554f7999f7968709e8e300 : SUCCESS in 13m 12s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa49a3e4e2994d8687a6492e7d78b629 : SUCCESS in 14m 37s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5e0abe553df4755954deffc2edb25a6 : SUCCESS in 9m 38s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6147c46cc5540288a8ce1d88f3725d5 : SUCCESS in 11m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/85d0b91ea12d4cfb916a0d8e2fbb4fa3 : FAILURE in 10m 36s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/d886fa5242ea4b6585c396c1025b95c9 : SUCCESS in 25m 45s","accounts_in_message":[],"_revision_number":2},{"id":"c98ca651608d5675604c69670474ee6dbde1f5f0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-12 01:59:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5194cbdba461c8320f3c815e4b410074ea0f773f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 03:16:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f975f2e8d10d4306885f9840eca5023d\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff2bee4862f14e70bf5e82fc6d6c63f3 : SUCCESS in 2m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bf6515fb82a4af8a5d7c7f0630119b4 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da785d1fdab049cc8f04d134f5f2a02e : SUCCESS in 3m 14s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6874317ff9bb4c2792a3ccd7bc435858 : SUCCESS in 11m 18s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/97ad5096e2b941f99a273099ec25ea75 : SUCCESS in 13m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b99af303c83b43239f735bd0047f81a1 : SUCCESS in 13m 12s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4daaf4042d5748f3b3055848d99cb621 : SUCCESS in 11m 25s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65476744425c4d32abe3ba5722f4d12d : SUCCESS in 10m 52s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92d86c9734754e448de02e13dbac0408 : FAILURE in 34m 49s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc154435e0b542859decd9fe55ad010a : SUCCESS in 1h 10m 18s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c296a1892804ff3bb9321906c56064d : FAILURE in 21m 23s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/fd03f8d4d4e648fb918a2b649f505118 : SUCCESS in 26m 54s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3de043e647134eec880ed3b232ed5da9 : FAILURE in 18m 36s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d4144c6cc554fb3bdb3f605d10dc30f : FAILURE in 21m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6eb80ad28a59cadcff6ebee8752f078a6fdc8090","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 03:17:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4313d1b3f09ee54a8caf43933edf739364dcc62c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 04:30:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ecfb173674a4316af9a60772c38688f\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cff957208ae4248a67f70eb8d89f381 : SUCCESS in 3m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68c4a08fb9c5477d85706152fd0a2cb9 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49ccfb6c60f142d98134c48ec3e2d935 : SUCCESS in 3m 00s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc9a0f621e1a46028f1b04624ecb1a9d : SUCCESS in 14m 30s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/53e0ac5598574defbebc2cf4670adde7 : SUCCESS in 12m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e38c78a93b24dae931b6ea75978afb9 : SUCCESS in 12m 32s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e2d6f7dc2a64685a0c4f27e92924089 : SUCCESS in 12m 05s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/126b88d250d14a9daaef676a927097ba : SUCCESS in 1h 08m 07s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/34efc16db3014c29a153adac5b152095 : SUCCESS in 29m 01s","accounts_in_message":[],"_revision_number":2},{"id":"067307f700e11a696607bfb4fbceff1a0ed6db1a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 04:30:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e4a28d280671ef2452b6a0857a505f0c8891fbc3","revisions":{"fef4cec1e0dc6af481a69ebb31f81afd068b85a2":{"kind":"REWORK","_number":1,"created":"2022-11-30 09:26:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/866000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/00/866000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/00/866000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/00/866000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/00/866000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/00/866000/1"}}},"commit":{"parents":[{"commit":"e7d8c64f9907b5bddfa10406c86b4870faac3b93","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e7d8c64f9907b5bddfa10406c86b4870faac3b93"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 04:55:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-30 09:26:00.000000000","tz":0},"subject":"Ensure consistent IPv6 address format","message":"Ensure consistent IPv6 address format\n\nThis change ensures the shortened address representations are always\nused for ring devices to avoid broken idempotency caused by different\nrepresentations (shortened[1] vs canonical[2]).\n\n[1] ::1\n[2] 0000:0000:0000:0000:0000:0000:0000:0001\n\nCloses-Bug: #1997313\nChange-Id: I762f0780ba25536aa616482a49d59c7db69924d6\n(cherry picked from commit 249f466f94441ae554478467949378d81b034f89)\n(cherry picked from commit 4b37f4c4df42c3069b502ee9f539fdb5d0d4c897)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fef4cec1e0dc6af481a69ebb31f81afd068b85a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fef4cec1e0dc6af481a69ebb31f81afd068b85a2"}]},"branch":"refs/heads/stable/xena"},"e4a28d280671ef2452b6a0857a505f0c8891fbc3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-30 09:26:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/866000/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/00/866000/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/00/866000/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/00/866000/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/00/866000/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/00/866000/2"}}},"commit":{"parents":[{"commit":"e7d8c64f9907b5bddfa10406c86b4870faac3b93","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e7d8c64f9907b5bddfa10406c86b4870faac3b93"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 04:55:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-30 09:26:59.000000000","tz":0},"subject":"Ensure consistent IPv6 address format","message":"Ensure consistent IPv6 address format\n\nThis change ensures the shortened address representations are always\nused for ring devices to avoid broken idempotency caused by different\nrepresentations (shortened[1] vs canonical[2]).\n\n[1] ::1\n[2] 0000:0000:0000:0000:0000:0000:0000:0001\n\nCloses-Bug: #1997313\nChange-Id: I762f0780ba25536aa616482a49d59c7db69924d6\n(cherry picked from commit 249f466f94441ae554478467949378d81b034f89)\n(cherry picked from commit 4b37f4c4df42c3069b502ee9f539fdb5d0d4c897)\n(cherry picked from commit 81f930723b694260558f396b077a6fec4f9223f1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e4a28d280671ef2452b6a0857a505f0c8891fbc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e4a28d280671ef2452b6a0857a505f0c8891fbc3"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
