)]}'
{"id":"openstack%2Fpuppet-swift~866957","triplet_id":"openstack%2Fpuppet-swift~master~I37164a2fda28e63604545db46fc925ecf04f9204","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-12-12 00:11:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I37164a2fda28e63604545db46fc925ecf04f9204","subject":"Support customizing fallocate options","status":"MERGED","created":"2022-12-08 06:03:06.000000000","updated":"2022-12-12 00:12:05.000000000","submitted":"2022-12-12 00:11:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866957","meta_rev_id":"568f3643e6ea5adda91ed30d0fabb18ba2340f21","_number":866957,"virtual_id_number":866957,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-12 00:11:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-12-08 15:36:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-12-11 22:37:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-12-11 22:37:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-08 07:32:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-08 15:36:55.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-12-11 22:37:29.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"175eb344d1888eba21772ed40a4a97cf00eeb95c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-08 06:03:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82c1c8965d02816e6feb464dbbb0db12125dc020","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 07:32:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b25d0e32e298490eb3522b462ea35da8\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/309f3b9043c04e87bd02cecd99ce3fcf : SUCCESS in 2m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b4b1ebd815c4983ad0d005b990cc9ae : SUCCESS in 2m 45s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c437914cfd845b8a743fd1a67a776a6 : SUCCESS in 2m 41s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/3deb9d35e8484e2095cbc21b8c7dab0b : SUCCESS in 12m 06s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34a8455ae7794074b062aa28a15fb95d : SUCCESS in 7m 38s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99eb78874b77496aaa9490ca82b60883 : SUCCESS in 9m 30s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa0417fef1d84752bd763c5aedc0849e : SUCCESS in 8m 24s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bdd5a876d1e477aba87ae69033bc29b : FAILURE in 17m 40s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb2eb042f4024456b30d2f970681ca74 : SUCCESS in 1h 16m 11s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/663bde98e3ca40a28ae16cac73036a72 : FAILURE in 10m 50s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1efeeb078994d28a95787265fc91e69 : SUCCESS in 22m 50s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d39eb73247de40e09a40ff0560270aef : FAILURE in 33m 11s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef3eb8f53a9c45629ae29a489275832d : FAILURE in 36m 00s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/906eae005e6144e4a7384bd5a8307329 : SUCCESS in 2m 46s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a7c265fba55e40d685982560a2226c76 : SUCCESS in 1h 27m 37s","accounts_in_message":[],"_revision_number":1},{"id":"3698d85adcab8f0d075871df63f650760e51013f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-12-08 15:36:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5e35bd5a18ff872d559b6813d49c384651c376ae","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-12-11 22:37:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3f2e69fb926a2e85f1b3630baec4355b955c867b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-11 22:38:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e716628b6943d35ea5dcf8dcb832cd61ac446558","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 00:11:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53e94d514b404b95a30eb5872e89ed9f\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4a99a0215f94b1481d36d55d048f8b5 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a896bf1258fd4284a83d550a23013072 : SUCCESS in 2m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09d2791547f84c9c95caed14738bc6dd : SUCCESS in 3m 04s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a19c156d3bf64b92814a223ffaaf989e : SUCCESS in 12m 17s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0cffbd0a6344222a9129b1cd6d20821 : SUCCESS in 8m 50s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0c041f341b34565acd3a3ce63e5d437 : SUCCESS in 8m 36s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0edba93436ae4e2c9936c2fb9d9297a9 : SUCCESS in 1h 12m 18s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/cda6dd35d6884ae0a73fd96829dc9e25 : SUCCESS in 22m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2bd872ba5b74c178ecbc701a1eba393 : SUCCESS in 3m 08s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ee42e5f35e374ebfadd07d34e2fe26e6 : SUCCESS in 1h 31m 31s","accounts_in_message":[],"_revision_number":1},{"id":"3543d8364d3506beb0d9efee7346bbfc2b9d5e88","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 00:11:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"568f3643e6ea5adda91ed30d0fabb18ba2340f21","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 00:12:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f098f1a7edee4deaa37c8ac30aaf603a\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cfa101c6b484c399fe8ec1f00ea2fc0 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a05cf5840c9976c89e859af63e72e6eb94fca39","revisions":{"7a05cf5840c9976c89e859af63e72e6eb94fca39":{"kind":"REWORK","_number":1,"created":"2022-12-08 06:03:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/866957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/57/866957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/57/866957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/57/866957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/57/866957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/57/866957/1"}}},"commit":{"parents":[{"commit":"c17989e647120e15afec2730fa216b36e51de348","subject":"Merge \"Remove logic for old swift-ring-builder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c17989e647120e15afec2730fa216b36e51de348"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-08 06:01:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-08 06:02:55.000000000","tz":540},"subject":"Support customizing fallocate options","message":"Support customizing fallocate options\n\nThis change introduces the new parameters for the following options so\nthat users can customize fallocate calls in storage server processes.\n\n - [DEFAULT] disable_fallocate\n - [DEFAULT] fallocate_reserve\n - [app:\u003ctype\u003e-server] fallocate_reserve\n\nChange-Id: I37164a2fda28e63604545db46fc925ecf04f9204\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a05cf5840c9976c89e859af63e72e6eb94fca39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a05cf5840c9976c89e859af63e72e6eb94fca39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
