)]}'
{"id":"openstack%2Fpuppet-swift~867079","triplet_id":"openstack%2Fpuppet-swift~master~I8aded9b59939e5a2eab04424277d9082cbda0a3a","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-01-09 17:47:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8aded9b59939e5a2eab04424277d9082cbda0a3a","subject":"Add support for swift-drive-audit cron job","status":"MERGED","created":"2022-12-09 06:22:43.000000000","updated":"2023-01-09 17:48:36.000000000","submitted":"2023-01-09 17:47:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867079","meta_rev_id":"09ef67ea78cfd450bd8a46b255520ada6de7f4c5","_number":867079,"virtual_id_number":867079,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-09 17:47:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-12-20 16:26:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-01-09 16:11:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-01-09 16:11:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-09 08:36:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-20 16:26:28.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-01-09 16:11:55.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"b59922e450ebcecf0e0902562327b4579d6527ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-09 06:22:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"008d93b18c57c4bb47ed2b586c015c2b89c64190","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 08:36:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a729b76051774cec88f27a11a5b2b66b\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c00f8851b794e14a0a4bcaaf01bb9d9 : FAILURE in 2m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/365157da84414c3886af8d824c0bad63 : FAILURE in 3m 55s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f25b1e25c9e843fcb3749ed75fd2c1f7 : FAILURE in 3m 07s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/3dd0514d81b249509c0bea459894f74d : SUCCESS in 10m 40s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecd201463cc9446fbc5d16218ab4c20a : SUCCESS in 9m 23s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a92a20061504250b85f1c40fe0c19a2 : SUCCESS in 10m 08s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e70f5292aac3498ba1846a9cd23c1474 : SUCCESS in 10m 25s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e46c39d3de0546728152343478b5820b : FAILURE in 18m 26s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c8ece6f3c8c4cd98f39fc97d759f5c4 : SUCCESS in 1h 14m 12s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8b56e5af77849a887e437bc6dd12295 : FAILURE in 7m 09s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c66ae9fa7284a4abead9004987da922 : SUCCESS in 29m 18s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d0b9f0f96ff4be28a58467611e5c094 : FAILURE in 36m 03s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc3202ee49974180855e7f4b43aab0e6 : FAILURE in 36m 13s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6a5d8486365442329430ccc057c38c1e : SUCCESS in 2h 05m 59s","accounts_in_message":[],"_revision_number":1},{"id":"671c8e1b6b34a0ad86f995081964b5999f2e7a51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-12 05:27:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4a15297d1ffa86a59f30088b7bf33e23686a1ae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 07:31:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb2500aefd3d43faa48fdc56c87af543\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75e2917f60444b17b0f205c8c8905c29 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80c3d4afaa384613a46f96eab5e4431e : SUCCESS in 3m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0cae575bac548c5942ba44046dcfb7e : SUCCESS in 3m 35s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e3aff148c3941659157fdb980b48289 : SUCCESS in 13m 51s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3786084e667d490f93be82a91a9692e5 : SUCCESS in 7m 47s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/886ab39395174d41adf964eb3a9a615d : SUCCESS in 7m 42s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e08b7cf7be854daf9554b956377f0eae : SUCCESS in 10m 50s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af49a1db38ed4ec48b8c9b2294c12e74 : FAILURE in 16m 51s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5395c865fedf4158b997e1a1bc444c94 : SUCCESS in 51m 41s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f202c256dfef4a31bb14db834cef5d63 : FAILURE in 11m 12s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/6adb826a5c3b4a5e8f05823e0c438980 : SUCCESS in 28m 51s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ce11bfc8ddc40f383098c0142718409 : FAILURE in 33m 00s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3eacdf9143ab40cf8cef896951987ea2 : FAILURE in 36m 18s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8fd106d58a1b48ae93476dbe195ef51c : SUCCESS in 1h 59m 00s","accounts_in_message":[],"_revision_number":2},{"id":"1f17b7d3bc8c4225751cc4ec807f014bb2b57773","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-14 03:14:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"af93dff077c494228bb9bcd624cb1355f1752928","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 04:38:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e449aaec48a40b495c38afab2dd5918\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86813d13381c4b4dac46a187db978cc9 : SUCCESS in 3m 37s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff9f400f55924f4fb042aaedae9f7e48 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee1cd4ed1aea4453866d9f7199e1f2aa : SUCCESS in 3m 34s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/f824bd106a084fceb0131a8bf5d07c68 : FAILURE in 11m 05s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cc49e5be8c247f1af92a0a6bb6fcd3c : FAILURE in 13m 33s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/868592b58da64ac0ad33b947255dd8f2 : FAILURE in 10m 41s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88a72de8667542e2a56fd70bb15ecc93 : FAILURE in 10m 56s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fb064c69a1e4bdcaa4bb7f2e12e91ac : FAILURE in 14m 07s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1a6174ca8934bfa90d4a5b64dd0cc03 : SUCCESS in 1h 10m 32s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2198e7e7d3d4ee291835b93b8fd25cb : FAILURE in 11m 41s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c5bf552f9774b69951b376f4af489c5 : SUCCESS in 23m 52s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5a9140ac9c243ff86433bc635eff03d : FAILURE in 36m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ac4a05a9f96489ca28a08313b3b5ef7 : FAILURE in 34m 18s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3152d5df7f144451b74450498b86d4f0 : SUCCESS in 1h 14m 49s","accounts_in_message":[],"_revision_number":3},{"id":"d85097d77ef5cd3e0956677285119615a9b90d5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-14 06:30:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9a367bf4e277a95ce44e813d3edbb77db6afe0c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 08:09:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d318a4529134b918daa4b76b7de126c\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd181acc5ee442eeae8895898eb12515 : SUCCESS in 3m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03473b8e0b3742b3a40b5ce04dccd2aa : SUCCESS in 3m 02s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97d0839f99774bf99bbcc7b722c40133 : SUCCESS in 2m 53s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/20a4d36b96c94b0bb6e0291effabb4d1 : FAILURE in 13m 20s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9414e045f074f149369ffbfa07d0bd6 : FAILURE in 9m 20s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f44c77faf4964e57a0b75148422ca6c2 : FAILURE in 10m 20s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07ec6d5b49ea491e86f45f7f3a0a2132 : FAILURE in 9m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2a1bb7fca5c4ec49503ccaba40bfda0 : FAILURE in 17m 14s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4d3b3b7f19d49f699071b63facf9539 : SUCCESS in 1h 11m 42s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b382ecacffd84600a3d3f306c9869514 : FAILURE in 11m 06s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c7093b7948944d8a14e7e675ff96975 : SUCCESS in 28m 19s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e03196b9b3844f69c3a58bc31ebd065 : FAILURE in 36m 04s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11d4bc802f5a4972875b63cd794808fc : FAILURE in 36m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/380fbd47413f44b0bfa81402f533e8cf : SUCCESS in 3m 12s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4e6c1f990f6543b3829c1297bbf9a7b3 : SUCCESS in 1h 37m 13s","accounts_in_message":[],"_revision_number":4},{"id":"cf30b6be72e7958cecbf98d0ea24eafcffd5a2fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-19 01:45:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b5a1c12da469d13be061762893a59244e9dd7f62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-19 02:40:52.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c28a41a6d5bb2750c5b7cf64c17b5ed57d554b1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 04:17:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5b3be58fe7343428824cc3f3076f7eb\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1571228d9a2c46d79c624ceb4bdf1b67 : SUCCESS in 3m 44s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7978c13b9ea4d22869fc82ef94b382e : SUCCESS in 3m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/254a5b194ce94dc88b109d23552908be : SUCCESS in 3m 35s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/827d05e506994b8492db1fbc4718dbd3 : SUCCESS in 12m 42s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6061e1e2e4404807b27b20bf01de22b0 : SUCCESS in 13m 15s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b12a706cd6ae48f39e12ed96a8bc23b7 : SUCCESS in 13m 29s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14357d8aad614528896ac95c728f8779 : SUCCESS in 13m 09s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/413c5f7d189c4d998015bc3709f50db6 : FAILURE in 16m 54s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/32aea49892c64f35ba1e0aa38e89dc96 : SUCCESS in 1h 13m 28s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/844a1b32b1d14394ac63be66cb200d8b : FAILURE in 9m 59s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e373866915f4c7498bbf5b9eac7ad89 : SUCCESS in 29m 42s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/581d1b11cef44aaba43b9619c641e650 : FAILURE in 34m 11s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5961c4cc38a54eaf988b376af40ee97d : FAILURE in 38m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faf3bff6038e49368de44679bbd25d73 : SUCCESS in 3m 39s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9b6c9c8ab48945a59a780ace4448a57d : SUCCESS in 1h 33m 56s","accounts_in_message":[],"_revision_number":6},{"id":"f4c223457cbc67fae7f7db0064127688acb2f440","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-12-20 16:26:28.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"b20bafadf0478bdced51627c67075d41076944d4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-09 16:11:55.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"6e9ad9fe06cb46bc2bc918b6700511a55b97f915","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 16:13:15.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4eeb260f8654e64b3c6006b339c8a48bca6f48bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 17:47:22.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b06a0026c1f48458c1a2a4d81882fbd\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3234a4c7a9a346da9062e326a15d67e3 : SUCCESS in 3m 06s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d070ea09320b49f8bdc0324da535cf4c : SUCCESS in 3m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a216c984ff44be69eba9bfa9ac4f89d : SUCCESS in 2m 52s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/11e3a3bbe57145c4a7b525c126ec5ebb : SUCCESS in 14m 02s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e42d2778b6554a2b81e2157d99c4ad1c : SUCCESS in 9m 21s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d33fb4b2a72c4a109f23acece5ac911c : SUCCESS in 9m 32s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae01914a0903439e86d9151eb77aafa8 : SUCCESS in 1h 13m 08s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/f268e1c2e7134f28ab019fd1a957c729 : SUCCESS in 30m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ce267bb2aa8465d8b9909274ffcee81 : SUCCESS in 4m 00s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/29ca7599e05e4cb5952861c236ddf508 : SUCCESS in 1h 33m 00s","accounts_in_message":[],"_revision_number":6},{"id":"5c3862120c375880b8a6ef1a0a1d49679c96d543","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 17:47:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"09ef67ea78cfd450bd8a46b255520ada6de7f4c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 17:48:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7448843d7f64d4a993bf9117ffd72c2\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2006d8fae78d458ea50798af8daa85c4 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"fd38b9cbfa44cae350358a5feb01c555df1602fb","revisions":{"9130fc8c9cca51f6ed006e037b2a12e5c833fdae":{"kind":"REWORK","_number":1,"created":"2022-12-09 06:22:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/867079/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/79/867079/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/1"}}},"commit":{"parents":[{"commit":"c17989e647120e15afec2730fa216b36e51de348","subject":"Merge \"Remove logic for old swift-ring-builder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c17989e647120e15afec2730fa216b36e51de348"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-09 06:22:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-09 06:22:32.000000000","tz":540},"subject":"WIP: Add support for swift-drive-audit cron job","message":"WIP: Add support for swift-drive-audit cron job\n\nChange-Id: I8aded9b59939e5a2eab04424277d9082cbda0a3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9130fc8c9cca51f6ed006e037b2a12e5c833fdae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9130fc8c9cca51f6ed006e037b2a12e5c833fdae"}]},"branch":"refs/heads/master"},"f9829069a05ae247b6b433fb3b01f718bcc03195":{"kind":"REWORK","_number":2,"created":"2022-12-12 05:27:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/867079/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/79/867079/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/2"}}},"commit":{"parents":[{"commit":"c17989e647120e15afec2730fa216b36e51de348","subject":"Merge \"Remove logic for old swift-ring-builder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c17989e647120e15afec2730fa216b36e51de348"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-09 06:22:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-12 05:26:54.000000000","tz":540},"subject":"WIP: Add support for swift-drive-audit cron job","message":"WIP: Add support for swift-drive-audit cron job\n\nChange-Id: I8aded9b59939e5a2eab04424277d9082cbda0a3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f9829069a05ae247b6b433fb3b01f718bcc03195"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f9829069a05ae247b6b433fb3b01f718bcc03195"}]},"branch":"refs/heads/master"},"23e1443112db9f1bd0edb6741ea51e8d21aaaf34":{"kind":"REWORK","_number":3,"created":"2022-12-14 03:14:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/867079/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/79/867079/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/3"}}},"commit":{"parents":[{"commit":"c17989e647120e15afec2730fa216b36e51de348","subject":"Merge \"Remove logic for old swift-ring-builder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c17989e647120e15afec2730fa216b36e51de348"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-09 06:22:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-14 03:14:01.000000000","tz":540},"subject":"WIP: Add support for swift-drive-audit cron job","message":"WIP: Add support for swift-drive-audit cron job\n\nChange-Id: I8aded9b59939e5a2eab04424277d9082cbda0a3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/23e1443112db9f1bd0edb6741ea51e8d21aaaf34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/23e1443112db9f1bd0edb6741ea51e8d21aaaf34"}]},"branch":"refs/heads/master"},"edf6aad206b8e6689ca9030bc1599f7246e454b7":{"kind":"REWORK","_number":4,"created":"2022-12-14 06:30:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/867079/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/79/867079/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/4"}}},"commit":{"parents":[{"commit":"928a0e467b47f7aac7ef0524949ee8e8b95bbc55","subject":"Merge \"Support customizing fallocate options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/928a0e467b47f7aac7ef0524949ee8e8b95bbc55"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-09 06:22:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-14 06:30:33.000000000","tz":540},"subject":"Add support for swift-drive-audit cron job","message":"Add support for swift-drive-audit cron job\n\nThis change introduces the new class to configure cron job to run\nthe swift-drive-audit CLI tool periodically. The CLI tool checks\nthe log files to detect any device failure and umount faulty disks.\n\nChange-Id: I8aded9b59939e5a2eab04424277d9082cbda0a3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/edf6aad206b8e6689ca9030bc1599f7246e454b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/edf6aad206b8e6689ca9030bc1599f7246e454b7"}]},"branch":"refs/heads/master"},"84634139a2d1d794b7e351676cf9c6a24261d560":{"kind":"REWORK","_number":5,"created":"2022-12-19 01:45:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/867079/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/79/867079/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/5"}}},"commit":{"parents":[{"commit":"928a0e467b47f7aac7ef0524949ee8e8b95bbc55","subject":"Merge \"Support customizing fallocate options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/928a0e467b47f7aac7ef0524949ee8e8b95bbc55"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-09 06:22:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-19 01:44:51.000000000","tz":540},"subject":"Add support for swift-drive-audit cron job","message":"Add support for swift-drive-audit cron job\n\nThis change introduces the new class to configure cron job to run\nthe swift-drive-audit CLI tool periodically. The CLI tool checks\nthe log files to detect any device failure and umount faulty disks.\n\nChange-Id: I8aded9b59939e5a2eab04424277d9082cbda0a3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/84634139a2d1d794b7e351676cf9c6a24261d560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/84634139a2d1d794b7e351676cf9c6a24261d560"}]},"branch":"refs/heads/master"},"fd38b9cbfa44cae350358a5feb01c555df1602fb":{"kind":"REWORK","_number":6,"created":"2022-12-19 02:40:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/867079/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/79/867079/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/79/867079/6"}}},"commit":{"parents":[{"commit":"928a0e467b47f7aac7ef0524949ee8e8b95bbc55","subject":"Merge \"Support customizing fallocate options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/928a0e467b47f7aac7ef0524949ee8e8b95bbc55"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-09 06:22:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-19 02:40:25.000000000","tz":540},"subject":"Add support for swift-drive-audit cron job","message":"Add support for swift-drive-audit cron job\n\nThis change introduces the new class to configure cron job to run\nthe swift-drive-audit CLI tool periodically. The CLI tool checks\nthe log files to detect any device failure and umount faulty disks.\n\nChange-Id: I8aded9b59939e5a2eab04424277d9082cbda0a3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fd38b9cbfa44cae350358a5feb01c555df1602fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fd38b9cbfa44cae350358a5feb01c555df1602fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
