)]}'
{"id":"openstack%2Fpuppet-swift~870461","triplet_id":"openstack%2Fpuppet-swift~master~Ibe67a9516078e09e1c30727eb024de2a6e284ebf","project":"openstack/puppet-swift","branch":"master","topic":"puppet-jammy","hashtags":[],"change_id":"Ibe67a9516078e09e1c30727eb024de2a6e284ebf","subject":"Switch to Ubuntu Jammy (22.04)","status":"MERGED","created":"2023-01-15 13:32:18.000000000","updated":"2023-01-16 00:36:28.000000000","submitted":"2023-01-16 00:35:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870461-puppet-jammy","meta_rev_id":"5d4543ae56f7caed3e30bf18e43be7e5d7f7774c","_number":870461,"virtual_id_number":870461,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-16 00:35:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-01-16 00:23:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2023-01-16 00:23:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-15 13:36:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2de3b696ae33bef3555799d15baf514250626e61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-15 13:32:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1dd8d482f37a8dacdd61e65094a2ab22e9c8159","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-15 13:36:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de99f9f7522542ff843cd2dc055cb723\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/4f7fa1e9264940d1ba9fea00ee387025 : SUCCESS in 1m 51s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29ce8bd4dcc4410bb2ea11fd6fc551b7 : SUCCESS in 2m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0001e1eeed374ba18559e55dbfa9f747 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4c590d64fb8627a320a991409b27691115644ea9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-16 00:23:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a64063f9e802ec64242ea8231f97a341a5434a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 00:24:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d407bf73e84ef213d494302f50df43797e019e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 00:35:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/821e5e2217144ad39bcf921b0fd8068d\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/c3c9267c7a1946208f460993b40be5c3 : SUCCESS in 2m 02s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd0bd334000649959297576fd0650ead : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61920d1189fb4bcda2221e89fe06d9d8 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d36e781414c942fd2f6d6846807a0f6f6dc4f632","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 00:35:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5d4543ae56f7caed3e30bf18e43be7e5d7f7774c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 00:36:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/722438c58cb94380b93d2b5c61b267ed\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f23b8a3762904ef686d89b20803b7799 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4418d8a0c1db8745246fa1bf89fba3b1456b1752","revisions":{"4418d8a0c1db8745246fa1bf89fba3b1456b1752":{"kind":"REWORK","_number":1,"created":"2023-01-15 13:32:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/870461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/61/870461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/61/870461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/61/870461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/61/870461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/61/870461/1"}}},"commit":{"parents":[{"commit":"fd38b9cbfa44cae350358a5feb01c555df1602fb","subject":"Add support for swift-drive-audit cron job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fd38b9cbfa44cae350358a5feb01c555df1602fb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-15 13:32:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-15 13:32:10.000000000","tz":540},"subject":"Switch to Ubuntu Jammy (22.04)","message":"Switch to Ubuntu Jammy (22.04)\n\n... because Focal no longer supports the recent releases such as Zed.\n\nChange-Id: Ibe67a9516078e09e1c30727eb024de2a6e284ebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4418d8a0c1db8745246fa1bf89fba3b1456b1752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4418d8a0c1db8745246fa1bf89fba3b1456b1752"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
