)]}'
{"id":"openstack%2Fpuppet-swift~875309","triplet_id":"openstack%2Fpuppet-swift~master~I72b9b920a0fc9ea9628a4fca8c941399344deab1","project":"openstack/puppet-swift","branch":"master","topic":"puppet-unit-constraint","hashtags":[],"change_id":"I72b9b920a0fc9ea9628a4fca8c941399344deab1","subject":"DNM: Testing unit tests","status":"ABANDONED","created":"2023-02-26 04:21:15.000000000","updated":"2023-02-26 05:00:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"20eb5aa5002eac70e399fe59b7fd8f1156039708","_number":875309,"virtual_id_number":875309,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-26 04:46:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-26 04:46:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce53b0abbe72c88c0277558307ddc60ad80a992a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-26 04:21:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"010f21cb3f6d526ba618c1fa14eac269035c0c00","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-26 04:22:13.000000000","message":"Topic dnm removed","accounts_in_message":[],"_revision_number":1},{"id":"100c773bfb737a5f44221a0a0b37b37d1a7e4a76","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-26 04:22:16.000000000","message":"Topic set to puppet-unit-constraint","accounts_in_message":[],"_revision_number":1},{"id":"2638301dd50ff3af151ddea1acccf8beb51aaef6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-26 04:46:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5af19246e73b4d91bb9638d914da5f10\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e519bbdd8f0d43e9adf6e648e169833e : SUCCESS in 3m 23s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c5813c5e50649b9b11d982dd9acba33 : SUCCESS in 3m 25s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/01fe99dc5c3a47d0804481a0b9ba7ef5 : SUCCESS in 12m 33s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9117a930cb844acb79910323d8e8651 : SUCCESS in 8m 49s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab9f2472174b40f5842b0c9bd7522777 : SUCCESS in 8m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"20eb5aa5002eac70e399fe59b7fd8f1156039708","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-26 05:00:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1505606dbe2cc31d577853a1071e5a426bd2b06","revisions":{"d1505606dbe2cc31d577853a1071e5a426bd2b06":{"kind":"REWORK","_number":1,"created":"2023-02-26 04:21:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/875309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/09/875309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/09/875309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/09/875309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/09/875309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/09/875309/1"}}},"commit":{"parents":[{"commit":"b6544ca2a0dacb7c948d63c7261af87328705736","subject":"Compact multi-line warning message into a single line","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b6544ca2a0dacb7c948d63c7261af87328705736"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-26 04:20:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-26 04:20:56.000000000","tz":540},"subject":"DNM: Testing unit tests","message":"DNM: Testing unit tests\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-openstack-integration/+/875300\nChange-Id: I72b9b920a0fc9ea9628a4fca8c941399344deab1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d1505606dbe2cc31d577853a1071e5a426bd2b06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d1505606dbe2cc31d577853a1071e5a426bd2b06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
