)]}'
{"id":"openstack%2Fpuppet-swift~876026","triplet_id":"openstack%2Fpuppet-swift~master~I4c0a9357fcc4184df852024e1f6f36c0ed2e1440","project":"openstack/puppet-swift","branch":"master","topic":"fact-lint","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-03 11:26:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4c0a9357fcc4184df852024e1f6f36c0ed2e1440","subject":"Replace legacy facts and use fact hash","status":"MERGED","created":"2023-03-02 03:50:26.000000000","updated":"2023-03-03 11:26:40.000000000","submitted":"2023-03-03 11:26:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876026-fact-lint","meta_rev_id":"333e708ebfc286bd43757689b1bc6d308e1adc91","_number":876026,"virtual_id_number":876026,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-03 11:26:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-03-02 16:51:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-03-03 08:19:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-03-03 08:19:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-02 07:56:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-02 16:51:13.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-03-03 08:19:50.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"f3491d18c226cf3029d1fa782788ad9307bc861c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-02 03:50:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"064c39a2f2f36de1e6a883e424d3241767bc299b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-02 04:31:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ee59fc8090f745df3b598e2163f4a74cbfa41121","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-02 04:55:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6d9536c8f89dc23fadf0e0139936bb60e53b21f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-02 04:55:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b4be4bf201c38bca21177c1c6d681d3acf78c040","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-02 05:47:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7137a5bbb5558bf93c1cbd2a4db9519f003e7ef2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 07:56:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b589aeb227534721967c8aafa7abf2c4\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2a426d3c28a4b3e9ab1032ea7f7bceb : SUCCESS in 2m 31s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f1bc564f8304180a3c6503d43135c52 : SUCCESS in 2m 36s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee72bafdae374237ac9a711b041b66ff : SUCCESS in 13m 27s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79d819f927c0411da7381261cd17c3aa : SUCCESS in 9m 18s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1bb8f83facbb42aa8b869cdfa92ecbcf : SUCCESS in 8m 11s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eae5831c448b4b838f8bff147b481640 : FAILURE in 43m 14s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee3850d039ae4a46b70ddb3529f4fe85 : SUCCESS in 1h 11m 15s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ce1a2e763c24b3386931ae2f56ece71 : SUCCESS in 27m 08s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ef4ae9b4bb34fe6876d6d4ca3420b87 : SUCCESS in 17m 30s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e92d2ab3cbb249c39bd47806e35e5518 : SUCCESS in 17m 41s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9f904bd65ed14bdeb6a767e162c01d5d : SUCCESS in 2h 00m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"06dd3a093961766ac30e5549adafaa2722f9df78","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-03-02 16:51:13.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"54bbe8a42c38f057a5c5ac9cf768f2da9aa016af","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-03 08:19:50.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a30dedd2476078dd2e44703f186a7ebad1635317","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 08:20:57.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e0ea1025f5572a9c96418a4386f754f767f24e86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 11:26:40.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4bdd8489ba84008aa1f780c6f24d014\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8e56e66b15f4418b6f469fbf12fb2ba : SUCCESS in 3m 15s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca5b74e23fe44e7db3548c46f17e1673 : SUCCESS in 2m 28s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/b82d7515b8e0473da1d0a3a6a7c4acc2 : SUCCESS in 15m 19s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9548a9b4af04f4188f1396e76378c7b : SUCCESS in 9m 16s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b4fb7da2dbd4b67abc5e7c024f938eb : SUCCESS in 1h 18m 18s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6486d5a6c794042888bd42a1aac0a4a : SUCCESS in 32m 59s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/94a7bbd988dc429db68c14ec1384dae0 : SUCCESS in 2h 59m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"333e708ebfc286bd43757689b1bc6d308e1adc91","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 11:26:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"109cd356de1573fa86e1290a78e9be26975ac270","revisions":{"92d456dc793c04d7022fbcb0afcae181f01afa1e":{"kind":"REWORK","_number":1,"created":"2023-03-02 03:50:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/876026/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/26/876026/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/1"}}},"commit":{"parents":[{"commit":"76d421cc0755c4d0d3945324d720abf59ed65425","subject":"Merge \"Merge (more) common options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/76d421cc0755c4d0d3945324d720abf59ed65425"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-02 03:49:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-02 03:50:17.000000000","tz":540},"subject":"Replace legacy facts and use fact hash","message":"Replace legacy facts and use fact hash\n\n... because the latest lint no longer allows usage of legacy facts and\ntop scope fact.\n\nChange-Id: I4c0a9357fcc4184df852024e1f6f36c0ed2e1440\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/92d456dc793c04d7022fbcb0afcae181f01afa1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/92d456dc793c04d7022fbcb0afcae181f01afa1e"}]},"branch":"refs/heads/master"},"341a97132ecb19a0969a7a8989aab213e039bf8c":{"kind":"REWORK","_number":2,"created":"2023-03-02 04:31:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/876026/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/26/876026/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/2"}}},"commit":{"parents":[{"commit":"76d421cc0755c4d0d3945324d720abf59ed65425","subject":"Merge \"Merge (more) common options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/76d421cc0755c4d0d3945324d720abf59ed65425"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-02 03:49:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-02 04:31:07.000000000","tz":540},"subject":"Replace legacy facts and use fact hash","message":"Replace legacy facts and use fact hash\n\n... because the latest lint no longer allows usage of legacy facts and\ntop scope fact.\n\nChange-Id: I4c0a9357fcc4184df852024e1f6f36c0ed2e1440\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/341a97132ecb19a0969a7a8989aab213e039bf8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/341a97132ecb19a0969a7a8989aab213e039bf8c"}]},"branch":"refs/heads/master"},"3194a4df7e310f093530ae02413ea268e5a95bd0":{"kind":"REWORK","_number":3,"created":"2023-03-02 04:55:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/876026/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/26/876026/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/3"}}},"commit":{"parents":[{"commit":"76d421cc0755c4d0d3945324d720abf59ed65425","subject":"Merge \"Merge (more) common options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/76d421cc0755c4d0d3945324d720abf59ed65425"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-02 03:49:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-02 04:55:03.000000000","tz":540},"subject":"Replace legacy facts and use fact hash","message":"Replace legacy facts and use fact hash\n\n... because the latest lint no longer allows usage of legacy facts and\ntop scope fact.\n\nChange-Id: I4c0a9357fcc4184df852024e1f6f36c0ed2e1440\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3194a4df7e310f093530ae02413ea268e5a95bd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3194a4df7e310f093530ae02413ea268e5a95bd0"}]},"branch":"refs/heads/master"},"286ee4e8f152acebfa23c9b30188b4a3e6f16a91":{"kind":"REWORK","_number":4,"created":"2023-03-02 04:55:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/876026/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/26/876026/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/4"}}},"commit":{"parents":[{"commit":"76d421cc0755c4d0d3945324d720abf59ed65425","subject":"Merge \"Merge (more) common options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/76d421cc0755c4d0d3945324d720abf59ed65425"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-02 03:49:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-02 04:55:30.000000000","tz":540},"subject":"Replace legacy facts and use fact hash","message":"Replace legacy facts and use fact hash\n\n... because the latest lint no longer allows usage of legacy facts and\ntop scope fact.\n\nChange-Id: I4c0a9357fcc4184df852024e1f6f36c0ed2e1440\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/286ee4e8f152acebfa23c9b30188b4a3e6f16a91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/286ee4e8f152acebfa23c9b30188b4a3e6f16a91"}]},"branch":"refs/heads/master"},"109cd356de1573fa86e1290a78e9be26975ac270":{"kind":"REWORK","_number":5,"created":"2023-03-02 05:47:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/876026/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/26/876026/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/26/876026/5"}}},"commit":{"parents":[{"commit":"76d421cc0755c4d0d3945324d720abf59ed65425","subject":"Merge \"Merge (more) common options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/76d421cc0755c4d0d3945324d720abf59ed65425"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-02 03:49:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-02 05:47:43.000000000","tz":540},"subject":"Replace legacy facts and use fact hash","message":"Replace legacy facts and use fact hash\n\n... because the latest lint no longer allows usage of legacy facts and\ntop scope fact.\n\nChange-Id: I4c0a9357fcc4184df852024e1f6f36c0ed2e1440\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/109cd356de1573fa86e1290a78e9be26975ac270"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/109cd356de1573fa86e1290a78e9be26975ac270"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
