)]}'
{"id":"openstack%2Fpuppet-swift~885280","triplet_id":"openstack%2Fpuppet-swift~master~I2bf7240f8c52aef8b4dd74ce5af12056377ab227","project":"openstack/puppet-swift","branch":"master","topic":"rsync-hosts","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-06-26 16:16:10.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2023-06-26 13:55:53.000000000","reason":"removed on reply"},"7130":{"account":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"last_update":"2023-06-26 16:16:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2bf7240f8c52aef8b4dd74ce5af12056377ab227","subject":"Expose hosts_allow/deny for rsyncd configuration.","status":"MERGED","created":"2023-06-05 12:10:46.000000000","updated":"2023-06-26 16:17:22.000000000","submitted":"2023-06-26 16:16:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885280-rsync-hosts","meta_rev_id":"43b76ea8f02d1f9ba4a9c88a176af41c210d952c","_number":885280,"virtual_id_number":885280,"owner":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-26 16:16:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-06-26 11:17:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-06-26 13:55:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-06-26 13:55:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-05 13:48:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-05 14:31:28.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-06-06 01:46:24.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-06-26 11:17:33.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"95e8a9eaceb7cf866cb7fee8ea3cb7c5eb8c3025","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2023-06-05 12:10:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a6df81a683c08f24260c1ad128e34aac1628abc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2023-06-05 12:21:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0dff98ac3262c1c8322ec5d0cda1a58bcd322f4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 13:48:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/704a7c084fdd406682233d310e8f3ecb\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f352ed855f1466e895423b35939010b : SUCCESS in 2m 33s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8675664372264badb4eb847bc5d1f2c2 : SUCCESS in 2m 25s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a7079854211433ebea9fabce8603e5b : SUCCESS in 5m 18s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a2ae1ed7a6e4e27acc984418881996f : SUCCESS in 11m 57s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b84148a1dacf42068b862ee94f14c0d3 : SUCCESS in 8m 09s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/01d6f0bd697040c5bb5db4627c2df03c : FAILURE in 13m 41s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7ddb4235ccd407893f78cc01ba025d8 : SUCCESS in 8m 08s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20d95ff603f14e44939cfdeb6163eb94 : SUCCESS in 55m 59s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/19a1b6ae05214c17bf5c0859e28b9755 : SUCCESS in 1h 22m 02s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/b62cd994f5964784ad24e7f996915dcc : SUCCESS in 24m 17s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9fa8b7fc13e4caeb3bf60b0594e012d : SUCCESS in 16m 22s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd9ff4f9134845c5bb684b808a96e210 : SUCCESS in 15m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2922a5f94456e20efcba17fb11716590210627f6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-06-05 14:31:28.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0316d6d7d64dd7aa7ee893c413a50d32c02996bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2023-06-05 14:36:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"171ee66bd0ded0c9b6b2b2ef1a009a669d22ad26","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2023-06-05 15:29:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4fd29a3ace916fea25fc32ae7077e74c6c42a192","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 15:52:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36bdb11026b04851b10317f991a7d739\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/105df74c52e44e70b2abdb0ffb2761da : SUCCESS in 2m 51s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7924005d3c724834a29055d5257a874a : SUCCESS in 3m 14s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1cd907c04344b03a7e22d4f72e351bb : SUCCESS in 5m 38s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d376ae1c5d84dc1acef83821102197a : SUCCESS in 10m 32s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23c34b92064a44b1aa563fece0789bb7 : SUCCESS in 8m 35s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/87a6897a04b4473988063846b3aa3829 : FAILURE in 11m 53s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/768bbcbd8c004b8ab40575f4e71bd23f : SUCCESS in 8m 24s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bae1bacf38f4498a1b7f8862e5e0fab : SUCCESS in 1h 04m 37s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2456c4b984f6493c9cfbf8e1dab10885 : SUCCESS in 1h 14m 44s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/98266d4683054fe392c89fd3f0877b22 : SUCCESS in 22m 10s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ded759222e25477bb39b1e06623c94e5 : SUCCESS in 21m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2abeb1ee877d47b184520243a9afa0c8 : SUCCESS in 15m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"41f2d982f5c1dee80e58f2408e1d18270d08379a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-06 01:46:24.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"43f704dd8ef241ef13330f2352de6082b4e37441","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-06 01:46:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1b50f2b03e75529d94520114be18e0491cdfe9ad","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-06 01:49:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c5961d9e0b1abc361db61d97799ab3814e5594e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-26 02:15:08.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"0c27a602f1c4563d4654f63dfa2258e196c3b7b7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-26 02:15:35.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"35a04781015b6aacb51e9c24ac8c67c2bc605111","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 04:46:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e15974da3cf4833add0325df476fa32\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eab2d63ca42945a99907c69b83072f5c : SUCCESS in 2m 45s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e30ac6ae74d4b29b5d51a1be8fe9070 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3593c6e83084ce6b642bd40dc5a7531 : SUCCESS in 5m 36s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c5971f08b554ad9b2c9aec895497dd6 : SUCCESS in 14m 17s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41e89ce02ffa46f7ac8aa801fc797485 : SUCCESS in 10m 11s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/51bf1e5e9ef54b76b025f29908fec28f : SUCCESS in 14m 33s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80ab65411b0b4f9cac5147f9a066b06b : SUCCESS in 9m 44s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f99d649f8cbe4e29aa9715ade1b78182 : SUCCESS in 1h 11m 26s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/501fef5e122147d8a162a25acb979515 : SUCCESS in 1h 32m 19s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/c87ebadf65434fcdb86725ef01015dbc : SUCCESS in 26m 18s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea3a7932cb724ba9b0cea793b683204b : SUCCESS in 21m 03s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5aa9c5b17eca4420841bfb0ad189258f : SUCCESS in 20m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41ee5f4f87c64fa6ac203c1d4f0f5f00 : FAILURE in 3m 02s","accounts_in_message":[],"_revision_number":4},{"id":"41373e6ceaf9e12dfe079da46e4ecc778b38980b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-26 08:48:54.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"5edea9f84ba795fa06a75b1aabbd97e5b7d1545c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-26 08:49:02.000000000","message":"Topic hosts_allow removed","accounts_in_message":[],"_revision_number":5},{"id":"dc7c459108761cd5b725a039d07c0ceb0765b10a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-26 08:49:13.000000000","message":"Topic set to rsynchosts-allow","accounts_in_message":[],"_revision_number":5},{"id":"76448bab46b9c9efa46774fc6eff603b3ae700b8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-26 08:49:16.000000000","message":"Topic rsynchosts-allow removed","accounts_in_message":[],"_revision_number":5},{"id":"206f066916cb37fa8e1abe3dc56e6804540f975f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-26 08:49:26.000000000","message":"Topic set to rsync-hosts","accounts_in_message":[],"_revision_number":5},{"id":"b21126cbd393af5a099a5e4bac3e1d5af78bb001","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 10:44:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27f227589dc843329497f50146495799\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73891897f4fe4e9bad71242367a805e0 : SUCCESS in 2m 50s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11589eee123c48e6bafd1739b196453c : SUCCESS in 2m 49s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/8194c47ccea241e6a9599b00c962b05a : SUCCESS in 6m 45s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6bc0d5cec914c74ad75bb6f8598b386 : SUCCESS in 14m 10s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de597763ec4e4fa2a2f20fe3b511ec6a : SUCCESS in 9m 54s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b2fbfd42d174483b6c5ac3f7a7be0a9 : SUCCESS in 13m 32s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d62f36140007464e9ce5a16bada9571e : SUCCESS in 10m 11s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffea067a8d7a4d93ae58f77304d1a55c : SUCCESS in 54m 49s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee2093b5686946d88f5f582809efbe96 : SUCCESS in 1h 44m 04s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/28c68bced77c405284e951e774840f18 : SUCCESS in 30m 32s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af96d15cf49d4fbea77ef6e213f8e5db : SUCCESS in 19m 11s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52df52d0c92d4ca39a9eaabdd6aae84e : SUCCESS in 17m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d5cec005ef640b9ba004d9bef175fec : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":5},{"id":"0f3d087ad07ac4b973a1a64043388d825c146176","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-06-26 11:17:33.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e9fe424328775c69e607856a1b30cbd93e962d59","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-06-26 13:55:53.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8852c01391b742818633f33d951fe47dac9e55d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 13:56:55.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d0583442a009e0c367fd56e090e3fc54765b97dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 16:16:10.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60acd6b5e29947828fd705b880ceac90\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64e56628d924471390440e26b20fdcd6 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/daf813298dc345f5a3ef85cce1b0ad92 : SUCCESS in 2m 41s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/67a997ac1c774b669f89e6ec0d71cc5f : SUCCESS in 12m 31s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d3f4aedc79a49f0af3d9ea8cabd7bbb : SUCCESS in 9m 53s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/788a432cf0ce44aab47c746d488c5be6 : SUCCESS in 12m 34s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca85230aa0f0444e8e0897409fa72e0c : SUCCESS in 1h 29m 03s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/2df437eb43124395bb0c62700eec2149 : SUCCESS in 24m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44e4a24d21f94b0cb79918d72cf3c374 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":5},{"id":"ccd76ec00e6860adccaf0799573adde13b62bb14","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 16:16:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"43b76ea8f02d1f9ba4a9c88a176af41c210d952c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 16:17:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95e5cf3267634b008b53e3b5cace7564\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/994346388f5644ca9a742ef0f3de9266 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"de05151b9c8f058d164294b6951d2e30fd27e7a4","revisions":{"fdcea61d3ad87aa87bb11e82bd9d25541774d452":{"kind":"REWORK","_number":1,"created":"2023-06-05 12:10:46.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/80/885280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/80/885280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/1"}}},"commit":{"parents":[{"commit":"ee412357135b11206acd631ea49335f78fbbd577","subject":"Merge \"Remove support for Puppet 6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ee412357135b11206acd631ea49335f78fbbd577"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2023-06-05 12:10:28.000000000","tz":-240},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2023-06-05 12:10:28.000000000","tz":-240},"subject":"Expose hosts_allow for rsyncd configuration.","message":"Expose hosts_allow for rsyncd configuration.\n\nExpose hosts_allow for rsync configuration.\n\nChange-Id: I2bf7240f8c52aef8b4dd74ce5af12056377ab227\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fdcea61d3ad87aa87bb11e82bd9d25541774d452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fdcea61d3ad87aa87bb11e82bd9d25541774d452"}]},"branch":"refs/heads/master"},"c436db7ca5f461a0dcd6b3e08621729b2cd39afa":{"kind":"REWORK","_number":2,"created":"2023-06-05 12:21:03.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/80/885280/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/80/885280/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/2"}}},"commit":{"parents":[{"commit":"ee412357135b11206acd631ea49335f78fbbd577","subject":"Merge \"Remove support for Puppet 6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ee412357135b11206acd631ea49335f78fbbd577"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2023-06-05 12:10:28.000000000","tz":-240},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2023-06-05 12:21:00.000000000","tz":-240},"subject":"Expose hosts_allow for rsyncd configuration.","message":"Expose hosts_allow for rsyncd configuration.\n\nExpose hosts_allow for rsync configuration.\n\nChange-Id: I2bf7240f8c52aef8b4dd74ce5af12056377ab227\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c436db7ca5f461a0dcd6b3e08621729b2cd39afa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c436db7ca5f461a0dcd6b3e08621729b2cd39afa"}]},"branch":"refs/heads/master"},"67e9a16d6fe51182564f1f782bf4e8f4995ccdf1":{"kind":"REWORK","_number":3,"created":"2023-06-05 14:36:10.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/80/885280/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/80/885280/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/3"}}},"commit":{"parents":[{"commit":"ee412357135b11206acd631ea49335f78fbbd577","subject":"Merge \"Remove support for Puppet 6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ee412357135b11206acd631ea49335f78fbbd577"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2023-06-05 12:10:28.000000000","tz":-240},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2023-06-05 14:35:53.000000000","tz":-240},"subject":"Expose hosts_allow for rsyncd configuration.","message":"Expose hosts_allow for rsyncd configuration.\n\nExpose hosts_allow for rsync configuration.\n\nChange-Id: I2bf7240f8c52aef8b4dd74ce5af12056377ab227\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/67e9a16d6fe51182564f1f782bf4e8f4995ccdf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/67e9a16d6fe51182564f1f782bf4e8f4995ccdf1"}]},"branch":"refs/heads/master"},"8e0f0f3d3d4841b9303c4a7c274840c4a7aedd5a":{"kind":"REWORK","_number":4,"created":"2023-06-26 02:15:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/885280/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/80/885280/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/4"}}},"commit":{"parents":[{"commit":"209a71a0163ad881d3169d4f3767bac1db3e2a86","subject":"Refactor unit tests of swift::storage::all","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/209a71a0163ad881d3169d4f3767bac1db3e2a86"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2023-06-05 12:10:28.000000000","tz":-240},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-26 02:13:26.000000000","tz":540},"subject":"Expose hosts_allow/deny for rsyncd configuration.","message":"Expose hosts_allow/deny for rsyncd configuration.\n\nThis introduces the hosts_allow parameter and the hosts_deny parameter\nso that users can limit access to the rsync server.\n\nChange-Id: I2bf7240f8c52aef8b4dd74ce5af12056377ab227\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8e0f0f3d3d4841b9303c4a7c274840c4a7aedd5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8e0f0f3d3d4841b9303c4a7c274840c4a7aedd5a"}]},"branch":"refs/heads/master"},"de05151b9c8f058d164294b6951d2e30fd27e7a4":{"kind":"REWORK","_number":5,"created":"2023-06-26 08:48:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/885280/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/80/885280/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/80/885280/5"}}},"commit":{"parents":[{"commit":"209a71a0163ad881d3169d4f3767bac1db3e2a86","subject":"Refactor unit tests of swift::storage::all","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/209a71a0163ad881d3169d4f3767bac1db3e2a86"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2023-06-05 12:10:28.000000000","tz":-240},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-26 08:48:53.000000000","tz":0},"subject":"Expose hosts_allow/deny for rsyncd configuration.","message":"Expose hosts_allow/deny for rsyncd configuration.\n\nThis introduces the hosts_allow parameter and the hosts_deny parameter\nso that users can limit access to the rsync server.\n\nChange-Id: I2bf7240f8c52aef8b4dd74ce5af12056377ab227\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/de05151b9c8f058d164294b6951d2e30fd27e7a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/de05151b9c8f058d164294b6951d2e30fd27e7a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
