)]}'
{"id":"openstack%2Fpuppet-swift~885334","triplet_id":"openstack%2Fpuppet-swift~master~I7c291035e4eef39333150629b65abd915976ca79","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-06-08 14:11:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7c291035e4eef39333150629b65abd915976ca79","subject":"Remove deprecated swift::max_header_size","status":"MERGED","created":"2023-06-06 03:26:04.000000000","updated":"2023-06-08 14:12:21.000000000","submitted":"2023-06-08 14:11:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885334","meta_rev_id":"640ae0bbab8f484ae0a98d983aa37e1bf33af40f","_number":885334,"virtual_id_number":885334,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-08 14:11:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-06-06 15:51:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-06-08 12:40:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-06-08 12:40:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-06 03:46:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-06 15:51:28.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-06-08 12:40:28.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"d99e626205021dff78bf340ec73ce8d352679d3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-06 03:26:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7cf5355776a712eddf94479f13f293a44a8d2256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 03:46:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/715a698b03dd4020852ae57f9150dedf\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/991e3c47e25c40b2b869c8c776cc2871 : FAILURE in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6476b6d572794dc9bce49720ac30152b : FAILURE in 2m 45s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc70522d14a94aad89df4829fbd5c6ff : FAILURE in 4m 50s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f684ad8405145c4b035844ed2f33c21 : FAILURE in 12m 03s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45a24e40fd0549d1a6cf8b78a01bdff7 : FAILURE in 9m 07s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/64bf1d7ef2b0496a8788482c7476efcb : FAILURE in 10m 51s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3491dd2c50e426995246c05a415d23e : FAILURE in 10m 36s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8285f3b729eb45e4b37f748d4c48331f : FAILURE in 8m 13s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/98cde703b30a4adc81361e1a2d8d8043 : FAILURE in 11m 51s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc0c6085363247dbbec98bc4f1997e5b : FAILURE in 10m 52s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/055df10cea8b47a8b316599fcb24bd40 : FAILURE in 8m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7b5649a83884fd7bb96a929ee19317b : FAILURE in 8m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a48e5536b6ec43778d509201b1ca0e4f : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"9177cea80f12993bc864a827b929248caa7aa6a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-06 03:53:54.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"941b9e126853259bc96d07286eeec7df6ef8abb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 04:58:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcc2bef471dd4eda8973b7b42fc98f3f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b5af964687e4181adda6e7a34e4909a : SUCCESS in 2m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cfe68f3633b46c18660c9ee019f827b : SUCCESS in 2m 43s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc7866822352413281ae23b242ac0898 : SUCCESS in 5m 00s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4757c688a794646844b23eb88ed456b : SUCCESS in 12m 11s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/296e2322af3543d7a2a2a76135ade610 : SUCCESS in 10m 04s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/719ca10212b9489397900315be2e0fbd : SUCCESS in 12m 34s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82c04216638d4af8851a847d197d2464 : SUCCESS in 8m 05s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1104230a0f44c56a57d3edc5936ac5e : SUCCESS in 59m 17s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c8b6512a70b4abfac6b3d422203c551 : SUCCESS in 59m 43s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/11610c102d284bf9b953504efdf4ec38 : SUCCESS in 26m 06s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/877b922b53b9497f9e71781fc535c273 : SUCCESS in 17m 47s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eef5a6cbd3ce4927bafcdb2932292de2 : SUCCESS in 17m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d7075bbd33246b1bf3406eeaaf0b6eb : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":2},{"id":"5f6cc5f35ec7d2277645e725a8836f105ec8cdbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-06 05:14:51.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"208c1f4cf6c597009a25ea409baa6f989eee370d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 06:54:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb07b115f8fc4e72bcae554e9b4c0575\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/558b180aba88431e909219af3965bac7 : SUCCESS in 3m 09s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05ac90feb5f3441ca3fa66d6134d7918 : SUCCESS in 2m 59s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e4f00f6e15c4fe1aa1f3acaffb28886 : SUCCESS in 4m 49s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/53e947d0b3484d14a29c7f699a8d65ba : SUCCESS in 11m 54s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61225dfff978497eafe3d6e000fd6c9d : SUCCESS in 9m 41s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/18aafcf77d8b41cea1c883f3547de959 : SUCCESS in 11m 53s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbe3c52dcf4d4579a730fd0bd0eddb66 : SUCCESS in 9m 14s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55f8baceb28a49399b966a0c4951bec6 : SUCCESS in 54m 53s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b7cbb628f8641ab9b42c44262b8b038 : SUCCESS in 1h 34m 09s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/50c1c92b1fd64a98a958285766602307 : SUCCESS in 24m 03s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd464a58c5624b23aa92ea7fb00dd2bb : SUCCESS in 19m 31s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8bd18e5c18974178bf2de2a6aaaf260b : SUCCESS in 15m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90369d90454e462c9f4c5b9db29b5064 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":3},{"id":"07a342ed4f58eff23f2a23abcf73e6cdb3095e23","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-06-06 15:51:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1997ad6d1dc779e206efd27d686f25c781a7ce77","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-06-08 12:40:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b93debe0cfe36f11242fd47e970ef6f1d3dc7b54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 12:41:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"da7f8d7651435ba14b5d873a6d108c5e2991ce0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 14:11:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6aa46f8563ea4ac3b2b94260da9705c0\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e96565c9f0f243b3ba716a6e055ef0df : SUCCESS in 2m 57s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad6797e387f9461abc0be65d5d892aee : SUCCESS in 2m 38s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/c32f472cc4f94650a12281102cc3f8d6 : SUCCESS in 13m 13s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7dec89430d7742db9a95ff4616dd01d0 : SUCCESS in 12m 02s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/a33fd05759d54729ac66056a59838d91 : SUCCESS in 12m 52s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/48d6486470674bfb8959c638e178b5eb : SUCCESS in 1h 03m 45s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/1882ad5b6fa4494186efca5cbdf7f621 : SUCCESS in 25m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46f46c8ce86c423da83ecbc5e96ff8dd : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":3},{"id":"e77a056d814a2d5b17eaf55a7f44be9ff0403cf7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 14:11:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"640ae0bbab8f484ae0a98d983aa37e1bf33af40f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 14:12:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dea13e54211442a99d9b15be3e91aa80\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77a2280c2daf4caeabcb3d7cb421482e : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f2bb7abca8020970288d13377c2bfd354baa6ea5","revisions":{"54109801d5e9b90577b3c47ddd3b6ddb0d4af92a":{"kind":"REWORK","_number":1,"created":"2023-06-06 03:26:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/885334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/34/885334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/34/885334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/34/885334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/34/885334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/34/885334/1"}}},"commit":{"parents":[{"commit":"a90f486b34fe59bb5db404fd4c6b19811ce4d529","subject":"Fix undefined reseller_prefix_upcase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a90f486b34fe59bb5db404fd4c6b19811ce4d529"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-06 03:25:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-06 03:25:51.000000000","tz":540},"subject":"Remove deprecated swift::max_header_size","message":"Remove deprecated swift::max_header_size\n\nThe parameter was deprecated during the past cycle in favor of the new\nconstraints class[1].\n\n[1] 7d220dd93468815f88c95c67b2d45d95152f8cae\n\nChange-Id: I7c291035e4eef39333150629b65abd915976ca79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/54109801d5e9b90577b3c47ddd3b6ddb0d4af92a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/54109801d5e9b90577b3c47ddd3b6ddb0d4af92a"}]},"branch":"refs/heads/master"},"6782bcc7e7b3984c3edde68f9dfcdfc97a077fa7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-06 03:53:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/885334/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/34/885334/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/34/885334/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/34/885334/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/34/885334/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/34/885334/2"}}},"commit":{"parents":[{"commit":"c971de515d17ee0a633e680047ceebb401dcc423","subject":"Fix undefined reseller_prefix_upcase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c971de515d17ee0a633e680047ceebb401dcc423"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-06 03:25:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-06 03:53:46.000000000","tz":540},"subject":"Remove deprecated swift::max_header_size","message":"Remove deprecated swift::max_header_size\n\nThe parameter was deprecated during the past cycle in favor of the new\nconstraints class[1].\n\n[1] 7d220dd93468815f88c95c67b2d45d95152f8cae\n\nChange-Id: I7c291035e4eef39333150629b65abd915976ca79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/6782bcc7e7b3984c3edde68f9dfcdfc97a077fa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/6782bcc7e7b3984c3edde68f9dfcdfc97a077fa7"}]},"branch":"refs/heads/master"},"f2bb7abca8020970288d13377c2bfd354baa6ea5":{"kind":"NO_CHANGE","_number":3,"created":"2023-06-06 05:14:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/885334/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/34/885334/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/34/885334/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/34/885334/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/34/885334/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/34/885334/3"}}},"commit":{"parents":[{"commit":"908bd01717e1c8e33274a436cfc763ae55b27447","subject":"Fix undefined reseller_prefix_upcase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/908bd01717e1c8e33274a436cfc763ae55b27447"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-06 03:25:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-06 05:14:43.000000000","tz":540},"subject":"Remove deprecated swift::max_header_size","message":"Remove deprecated swift::max_header_size\n\nThe parameter was deprecated during the past cycle in favor of the new\nconstraints class[1].\n\n[1] 7d220dd93468815f88c95c67b2d45d95152f8cae\n\nChange-Id: I7c291035e4eef39333150629b65abd915976ca79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f2bb7abca8020970288d13377c2bfd354baa6ea5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f2bb7abca8020970288d13377c2bfd354baa6ea5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
