)]}'
{"id":"openstack%2Fpuppet-swift~886970","triplet_id":"openstack%2Fpuppet-swift~master~I920294342c9c2c0567796f345cbfa9e39bb1f1d3","project":"openstack/puppet-swift","branch":"master","topic":"replace-validate_legacy","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-09-20 17:01:40.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-09-20 17:01:40.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2023-09-20 15:49:28.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I920294342c9c2c0567796f345cbfa9e39bb1f1d3","subject":"replace validate_legacy","status":"MERGED","created":"2023-06-26 09:15:33.000000000","updated":"2023-09-20 17:01:40.000000000","submitted":"2023-09-20 17:01:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886970-replace-validate_legacy","meta_rev_id":"ee5cd87a3c72b7b2a698c2f96313c60b601d8395","_number":886970,"virtual_id_number":886970,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-20 17:01:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-09-19 13:33:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-09-20 15:49:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-09-20 15:49:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-26 10:25:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-12 18:04:13.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-09-04 04:23:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"c3897da4ba7ccc4acf7d366c4692b691b988a51c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-26 09:15:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"849497a8eef2282bfaa022c232e4ca607ab2edde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 10:25:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/84462234f40c431eaea1292e08dcdc10\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4062f213ee4e4607aaaf2f7830dff2f1 : FAILURE in 2m 35s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/965daaef1ef3408097785f200e206bff : FAILURE in 2m 44s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/84e1e472672545548eb368b5bec8b94c : FAILURE in 5m 56s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/85124c938e3f45bcb133a916f309b1bc : FAILURE in 16m 03s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c983f95c3ad45708ef37fb659d51ca5 : FAILURE in 13m 06s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a58aad0ebaf4e4bbdbff55c03396591 : FAILURE in 15m 43s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b763b582d014317a874e850f16634c6 : TIMED_OUT in 1h 01m 25s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f4dac7058fb487b912de7e9da932e30 : FAILURE in 7m 53s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a42af33ff8b344f8894767c2252d46fc : FAILURE in 11m 15s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb361b42eb9c45a8a95bce5e6076c187 : FAILURE in 9m 53s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/476c3b0e460845f9adfb9d90df0b23d1 : FAILURE in 6m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7490b6cd0f5490198dc09ca496b9dff : FAILURE in 6m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c52fb8ff08608d8476f22dabc8b9ded0d4fdd16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-26 13:28:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8898f5828103649b9a58843ddd10383470aa7fbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 14:48:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/16ca36dbac224f86936a9dc9ee5a2785\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38ae672564cd408092c0f3824ad44c41 : FAILURE in 2m 41s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/870af67da3344219a511a519c18028e8 : FAILURE in 4m 20s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c216606c26b44a08275aec7c3f2f04e : FAILURE in 4m 54s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e5981c1dcb24426ae1e68c18310e069 : FAILURE in 12m 35s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ddb10f200564d2b9428be3b3ed30c2b : FAILURE in 9m 33s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/134cd51f5db14577b81ebbe69bef2cbd : FAILURE in 14m 33s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ef7b89aa3d0481da42314ef7bf109be : FAILURE in 12m 59s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c49a535bb7b4880af728fd7bc5572c6 : FAILURE in 8m 15s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/190b2954a84f404b8873b8c7af211a82 : FAILURE in 9m 32s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab3a3ec176d44198a2d78379e98919b0 : FAILURE in 12m 44s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0de10f4673143a6bea5232e0384b039 : FAILURE in 12m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11b860bbddb54464b9a1b40a5672f199 : FAILURE in 6m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"923e464c72bfae4b24d33ada0005f9fe8efe2cd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-27 06:12:58.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c7694fc5e166e9b62dbf118aec42bce474a36031","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 06:41:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/54ae5ee89d784ec1a347f417d143dc43\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/024f4554623a4bcea3d28501700993cb : FAILURE in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8326cb5075a74139a6e377a1d9ee4572 : FAILURE in 2m 52s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fd753beae22400b9e6394e332cf37c4 : FAILURE in 5m 45s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c78eb65227d4117b3e8e454c0e02949 : FAILURE in 10m 03s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/578c0b55952f48e1a6439f0d415128ee : FAILURE in 6m 36s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/1921b02afe814096a386067b5c58c931 : FAILURE in 8m 45s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8711a1a83924f27813407f9005dc67a : FAILURE in 5m 44s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01b4d21922eb40018ae5045e64e60ba8 : FAILURE in 6m 13s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8feeab4baccb4809abb2dc711a2509be : FAILURE in 12m 40s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f7310bf35224137a7f3773722d73101 : FAILURE in 12m 53s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72a83c7fe3a240428da32bf29f7052e7 : FAILURE in 7m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c51e910f0bf04b15a9fc62cf6a374ad0 : FAILURE in 6m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cfd8b6232daa564f25cd561c645fd619a98ba134","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-27 06:50:23.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1717f739b510775fbab80008c49c644aaf381858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 07:25:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5be946a3c72413a8ac100b9239e9b30\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5cc32afb29434860a90b52ffb9f5667d : SUCCESS in 3m 05s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/456d0c06308348b58ad38e88d8b2f0fe : SUCCESS in 2m 38s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/506f78f5f7f84c9eaf8253f924a98f8e : SUCCESS in 5m 58s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c6e569c5bf24bcb9ae6160c050f5440 : FAILURE in 8m 37s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e39416de7074914b9a921ce35f6a11b : FAILURE in 5m 59s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/8579705141b949e4aa9e03ae79e912f8 : FAILURE in 8m 38s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd7ff6218de1465490b756b17594b468 : FAILURE in 5m 58s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d2c850280a24d7d88395e5a06e8975b : FAILURE in 7m 09s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2cef14d7c2d047b39c460b742d0cacb1 : FAILURE in 10m 52s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/0777526c114f4b5ca314a913a1c2bc45 : SUCCESS in 23m 31s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b31cdea9d4fa4853832fdb272e710eec : SUCCESS in 20m 21s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41ac8ef5b9e74179b7015dd25c467b39 : SUCCESS in 15m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"be3f771840b0baf66d3b3264f16b68eecfb821fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-05 03:08:41.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"6c5bf3a4b90a108fdb68ac88734cde0dd416eb98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 03:49:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1422ed7a0ba74b1184bb0deb43159cf1\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30f829b5ff4440ada2d096b68b73851d : POST_FAILURE in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98778f07e8f3480ea2266833b86ac9dc : SUCCESS in 2m 52s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4ab7b76bff146d59578f68e5f9b26ee : POST_FAILURE in 6m 58s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb762780087e497291096731b7b8166c : POST_FAILURE in 15m 44s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbfa57f6217746c3ae1e3724406eb566 : POST_FAILURE in 11m 08s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ea1f8148c2c4b0f97c85204d2c11a0e : FAILURE in 12m 27s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9c94445b72343589972e8a5363e46ae : POST_FAILURE in 13m 27s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6940c09b0cf54fd4ad55217ee402b260 : POST_FAILURE in 7m 45s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d53e73d5dd874be795467740499a600e : FAILURE in 21m 07s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/26796db0685649eaba96691f593fbf04 : POST_FAILURE in 30m 43s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b2d801e0a654181a710003428decc39 : POST_FAILURE in 19m 42s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16635c02e4d74c7b8edf0fb7f040de69 : SUCCESS in 19m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2ee6576975b2a158f1f4df347b8b3b419898f429","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-05 07:00:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"436a43afcba586cfffdb40b256dc79679e42f04d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 07:34:40.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa6cfdf513784b0ca8b3d4f6396afef6\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02f4505838124ac18c8108e15fcb9916 : SUCCESS in 2m 16s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e7e415fdbc4447aa42fa36c56623757 : SUCCESS in 2m 14s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/faa27cc0e7a44dce9c83d1294b633cb0 : SUCCESS in 4m 39s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5c7bf384fca44bcae2be7f470f01c22 : POST_FAILURE in 13m 09s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f971dd317dbd4c488f7780f3f0d42320 : FAILURE in 8m 15s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/12c6ff07473d49d5834478d54ac69630 : POST_FAILURE in 10m 10s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a51d89c1e5964422a286022ffdcf2794 : FAILURE in 7m 40s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25b678268617461287b72379e6da10fc : FAILURE in 5m 59s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1ebdb93962246a2ae424ad9a023b3c3 : FAILURE in 9m 06s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b46ae34168e4f639e140d846b7e3171 : SUCCESS in 28m 24s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b39dad3d5cd48128e816fbb2579cc38 : POST_FAILURE in 18m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37cd8d6f2d0d4c59a9ad97f047158982 : POST_FAILURE in 20m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bc222a1cb57b0b31b7f5d5b87e9c0a261fc8077a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-12 08:05:59.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"d63c0b116b249ffe5182d6471cb9471cc2c880e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-12 08:07:12.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ad29ccfa9982c4d09747239d440c039ee353f439","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 09:35:55.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5faf45e073944ec0a8fa1469d52e9485\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81f2333a6ccf421c88340e955a30086b : SUCCESS in 2m 58s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46c739142cb4428090175215b4579d20 : SUCCESS in 2m 35s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2633beba9374303b899bd359ce167c0 : SUCCESS in 5m 37s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/62cfecae4d6e4900a801e89d4206a886 : FAILURE in 14m 09s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2adcbc4041641c0b2edd73ce2812719 : FAILURE in 9m 11s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f79ebb916434faea15bba61e2327301 : FAILURE in 12m 37s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24b19ba4e0bf4627ae44374f5cc8adb9 : FAILURE in 7m 54s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a286af09dbc64cc59e0fe488b18873bb : SUCCESS in 48m 22s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/369e9b28804c46f3a66530f6dd152604 : SUCCESS in 1h 21m 29s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/62bb73aa39454287b72c17c1c9608bfe : SUCCESS in 20m 54s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c38a80d700c492c8d1016688f581e61 : SUCCESS in 18m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e54d6e375904d7bab99658d7141afc5 : SUCCESS in 15m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"53ed778f29dd1ae4d681e20d290fc2cbfade31b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-12 10:55:17.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"905368aa13ab58cb5ed8bcdaed22afdb32afd061","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 13:03:05.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c18ea522e78425b9f3020a19ae4a552\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89addf56b5c3436a98b4356850bdd396 : SUCCESS in 2m 41s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0042ad68e48548258e4dcb63c94031af : SUCCESS in 2m 45s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/f77a706af069487ba2d82cc7a8abd88e : SUCCESS in 5m 43s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/7fcaec67b64b4cc7ae8d95f323478c57 : FAILURE in 12m 05s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ffdb9b2d5fe4a6fb24282cfac63bf4d : FAILURE in 9m 24s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/1bc0867e916a43d6965e8150ca798c9c : FAILURE in 11m 39s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15fb5a6f028f4c8cb6b3216656fe8c16 : FAILURE in 8m 19s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b42d65db3db42ceb85733f5b93e6b61 : SUCCESS in 1h 09m 53s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/26b382eeb9f74158bd632adb3d0eff2a : TIMED_OUT in 2h 06m 46s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb066908a2bf4aab8c2c86d4550a6daf : SUCCESS in 25m 17s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94a90f0b1f344e35b34ea814e95cb980 : SUCCESS in 16m 49s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b68004f4f354c8086da511f75076f9d : SUCCESS in 15m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7fc908135b4cfc7767d5cd08b416043d2bc0412b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-12 13:10:28.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"42b238ea1722ecbba446167883b78369ec2d1ab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 14:42:40.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d26f401c578f4e31a7200af5df5fe891\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3bae2a6a8234dcb81592d8bb36cda98 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/513567a9bfdf4a1cafef3ce8f5dafbeb : SUCCESS in 2m 58s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/77db0dbbca984212a94f4a807ccd100c : SUCCESS in 14m 30s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb31bafa07774769ab937e699e4af199 : FAILURE in 13m 13s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/678156ecbce64b4bbb09e3ab02e0bf14 : FAILURE in 9m 54s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d8c852410c749fabee5ca3d84571b47 : FAILURE in 12m 49s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4e7b74b222845629e076a6bb3f6e78d : FAILURE in 8m 13s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b10cfd33e51d4602b33284e75950d17d : SUCCESS in 58m 25s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/78020ee537194e3cbac661d568e2de1a : SUCCESS in 1h 16m 23s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d7c06d999f9450ca099ef3f614e768c : SUCCESS in 33m 39s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ecaeacbba114f2aa9f172fc0064b719 : SUCCESS in 14m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fcad4788a73e405389f0fcdac38729ff : SUCCESS in 20m 02s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ef10fa8bccf3379ad595256e0129a67bbe25854e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-12 15:40:13.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"ec31b51c7c6b0ce6405bcd8750bf6865185765cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 17:01:43.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07ad128805104fd88346c5c6d59490a2\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a952cfad5ee45f4bc88d76c4b92df5a : SUCCESS in 3m 06s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e173f68790604db692086abe60f7c196 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/1dcd9b3c7c4640e0bb2162791cf3b611 : SUCCESS in 5m 41s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/56e6d70ad54649648447ef152dcc692e : SUCCESS in 14m 42s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2662b1f76fd4748936f40171134292c : SUCCESS in 9m 47s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/36d1768d86e445df8e33aad987827f25 : SUCCESS in 14m 33s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eaf7fc188ad34ebd95739a1989274a7f : SUCCESS in 8m 33s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29dc6454f3544b08aeb500cf8d8916d1 : SUCCESS in 1h 04m 07s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d89f5d04e52a4cf9a6117c43087e4b31 : SUCCESS in 1h 14m 18s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/a64041e10dcb4005be21f320635cfefb : SUCCESS in 30m 56s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7e2dbceab194881ac1e39fed823e5e8 : SUCCESS in 19m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2742bbcf50014f7191c3041d43249a1c : SUCCESS in 18m 25s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"cc68f20aa3c08053d31c9ea36335b09a3b9ecdc8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-07-12 18:04:13.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"296d7c54d7ce543255645399529336f4b617a166","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-09-14 06:24:45.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"110962e5b523fa394b0a98147fa4a7b12e645301","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-14 06:48:43.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"fb61a531dcf238ccb5e4da9ba1879fc337c579e7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-14 06:49:00.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"76857d5fddefdbc92b668e48ae8af939cc5fd866","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 07:56:30.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55c9730d8cd745ddbaf2c634b6da1360\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5afaf16114794747b37a46e800cba081 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/376d27f0d8e249bfa8a922c800a55c9d : SUCCESS in 3m 52s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa2e3c54342e46598e6cd4fcdc55242a : SUCCESS in 7m 45s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6eb94f6f0e44f78b0dd07d18a45bc0d : SUCCESS in 13m 27s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e86b404ae2d64f908bab2d5673f01f99 : SUCCESS in 9m 52s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8d583bed42b418c98861bce30729a92 : SUCCESS in 12m 40s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50717f95cd3e4ca0b46d746be3c74c36 : SUCCESS in 9m 50s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cd03e48033140fe8ccc494da6881422 : SUCCESS in 44m 20s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9e92094de894b4c86af418571554c15 : SUCCESS in 1h 06m 17s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6efa945199743cba97e682c24982046 : SUCCESS in 23m 10s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5620c5b167ff4d718e80f6813cba29d1 : SUCCESS in 16m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94c32cfa44b44c6da96d5f44d88827bd : SUCCESS in 19m 25s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"0ca93e2a8e5fdb259f70f7b2cc7c246349fd944a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-09-15 14:31:31.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"bb33fa0924804c81ea002a32034daf4d36f4bfcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-09-19 13:33:00.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"47075d0df2291ad8764d14d00d2233889e2141e4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-09-19 13:33:54.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"4880653441301cf95692c354c17ac8f328d22a86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 14:57:25.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f1757f2fc5748d59e67c61feb3e4c0c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ace77024213a47759a7b509f345dbe86 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8337316649744b7f9e0d58cb70c05841 : SUCCESS in 3m 21s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff16eefef2f34e6ea15117e28185635a : SUCCESS in 6m 21s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9d9e674a95f4e999cd700eb1f187e4a : SUCCESS in 13m 58s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b33b35485b9f4ec8969756bb67ca023e : SUCCESS in 10m 10s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee1de6b0074d49efbc42f269104a31c4 : SUCCESS in 14m 15s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb6b12584f7d45e294af1f2cfa610911 : SUCCESS in 11m 18s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1fc0ffbf3e9407ab9b66f1635662c37 : SUCCESS in 44m 33s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/de284d88fa064474961cf193d3093670 : SUCCESS in 1h 18m 49s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/3919de6f8bc5423c95a22c8892e35ddd : SUCCESS in 26m 16s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2af1c4e4e1924160856e4d9e76963897 : SUCCESS in 16m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bccdde0c12204aa4be21001cc9eb79ca : SUCCESS in 16m 13s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"25ab61849f32fbf7345e1d7de2df78a1c6a18de0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-09-20 15:49:28.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"6572e10e3fa221ea1f33ef8d0177a7661eee1a8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 15:50:23.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"e9037f3ba9bb465ca677c1f72e51683d1c3fafda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 17:01:40.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/179dfdf604b644259a4e208be0633c0f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c221d0925794086b50cc8b50bfac8b7 : SUCCESS in 2m 52s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/561d4174aa314a1cb033e82301055cfc : SUCCESS in 2m 47s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a55833c1b894ee68b1bbcda091a801a : SUCCESS in 13m 04s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77bc7274bfdc4b27b30ec51833d8d7d0 : SUCCESS in 10m 09s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/b086a65231c946418857a4bc4e5a39f6 : SUCCESS in 12m 40s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc148ee14bbc4543be2b6002a3973e79 : SUCCESS in 1h 07m 15s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/fada1e3eab994195a2f7649dc67fc170 : SUCCESS in 28m 23s","accounts_in_message":[],"_revision_number":12},{"id":"ee5cd87a3c72b7b2a698c2f96313c60b601d8395","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 17:01:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"0f7bdd733aa4c090a8c6b1db3b92ce9184616186","revisions":{"05b2c378e302da70283d2afee1ce5c931b1c3de8":{"kind":"REWORK","_number":1,"created":"2023-06-26 09:15:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/886970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/70/886970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/1"}}},"commit":{"parents":[{"commit":"0ac71f0aa45cf8e3f674a2bc1ec959eb4b03ea18","subject":"Merge \"Drop unused parameter definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0ac71f0aa45cf8e3f674a2bc1ec959eb4b03ea18"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-26 09:15:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-26 09:15:17.000000000","tz":540},"subject":"wip: replace validate_legacy","message":"wip: replace validate_legacy\n\nChange-Id: I920294342c9c2c0567796f345cbfa9e39bb1f1d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/05b2c378e302da70283d2afee1ce5c931b1c3de8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/05b2c378e302da70283d2afee1ce5c931b1c3de8"}]},"branch":"refs/heads/master"},"08ca071b31495fb49430c3b535dfd9401a0fdf29":{"kind":"REWORK","_number":2,"created":"2023-06-26 13:28:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/886970/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/70/886970/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/2"}}},"commit":{"parents":[{"commit":"0ac71f0aa45cf8e3f674a2bc1ec959eb4b03ea18","subject":"Merge \"Drop unused parameter definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0ac71f0aa45cf8e3f674a2bc1ec959eb4b03ea18"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-26 09:15:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-26 13:28:21.000000000","tz":540},"subject":"wip: replace validate_legacy","message":"wip: replace validate_legacy\n\nChange-Id: I920294342c9c2c0567796f345cbfa9e39bb1f1d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/08ca071b31495fb49430c3b535dfd9401a0fdf29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/08ca071b31495fb49430c3b535dfd9401a0fdf29"}]},"branch":"refs/heads/master"},"8d6da58e13698f60e78dbf835e4e7cd5d9881613":{"kind":"REWORK","_number":3,"created":"2023-06-27 06:12:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/886970/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/70/886970/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/3"}}},"commit":{"parents":[{"commit":"de05151b9c8f058d164294b6951d2e30fd27e7a4","subject":"Expose hosts_allow/deny for rsyncd configuration.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/de05151b9c8f058d164294b6951d2e30fd27e7a4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-26 09:15:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-27 06:12:50.000000000","tz":540},"subject":"wip: replace validate_legacy","message":"wip: replace validate_legacy\n\nChange-Id: I920294342c9c2c0567796f345cbfa9e39bb1f1d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8d6da58e13698f60e78dbf835e4e7cd5d9881613"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8d6da58e13698f60e78dbf835e4e7cd5d9881613"}]},"branch":"refs/heads/master"},"473db31dbc94aaaad77e2e6c4aed25fff5b55871":{"kind":"REWORK","_number":4,"created":"2023-06-27 06:50:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/886970/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/70/886970/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/4"}}},"commit":{"parents":[{"commit":"de05151b9c8f058d164294b6951d2e30fd27e7a4","subject":"Expose hosts_allow/deny for rsyncd configuration.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/de05151b9c8f058d164294b6951d2e30fd27e7a4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-26 09:15:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-27 06:50:14.000000000","tz":540},"subject":"wip: replace validate_legacy","message":"wip: replace validate_legacy\n\nChange-Id: I920294342c9c2c0567796f345cbfa9e39bb1f1d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/473db31dbc94aaaad77e2e6c4aed25fff5b55871"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/473db31dbc94aaaad77e2e6c4aed25fff5b55871"}]},"branch":"refs/heads/master"},"d4bd06efceef45bb0de7f3a60e7e9c2f4ed7b95c":{"kind":"REWORK","_number":5,"created":"2023-07-05 03:08:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/886970/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/70/886970/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/5"}}},"commit":{"parents":[{"commit":"de05151b9c8f058d164294b6951d2e30fd27e7a4","subject":"Expose hosts_allow/deny for rsyncd configuration.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/de05151b9c8f058d164294b6951d2e30fd27e7a4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-26 09:15:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-05 03:08:32.000000000","tz":540},"subject":"wip: replace validate_legacy","message":"wip: replace validate_legacy\n\nChange-Id: I920294342c9c2c0567796f345cbfa9e39bb1f1d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d4bd06efceef45bb0de7f3a60e7e9c2f4ed7b95c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d4bd06efceef45bb0de7f3a60e7e9c2f4ed7b95c"}]},"branch":"refs/heads/master"},"dce90d184a1d9b0b587a8511eaa23a31e3b016e2":{"kind":"REWORK","_number":6,"created":"2023-07-12 08:05:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/886970/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/70/886970/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/6"}}},"commit":{"parents":[{"commit":"de05151b9c8f058d164294b6951d2e30fd27e7a4","subject":"Expose hosts_allow/deny for rsyncd configuration.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/de05151b9c8f058d164294b6951d2e30fd27e7a4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-26 09:15:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-12 08:05:51.000000000","tz":540},"subject":"wip: replace validate_legacy","message":"wip: replace validate_legacy\n\nChange-Id: I920294342c9c2c0567796f345cbfa9e39bb1f1d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/dce90d184a1d9b0b587a8511eaa23a31e3b016e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/dce90d184a1d9b0b587a8511eaa23a31e3b016e2"}]},"branch":"refs/heads/master"},"8329836b16efff6ff3dc8727872693a0c75c40c0":{"kind":"REWORK","_number":7,"created":"2023-07-12 08:07:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/886970/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/70/886970/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/7"}}},"commit":{"parents":[{"commit":"de05151b9c8f058d164294b6951d2e30fd27e7a4","subject":"Expose hosts_allow/deny for rsyncd configuration.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/de05151b9c8f058d164294b6951d2e30fd27e7a4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-26 09:15:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-12 08:07:04.000000000","tz":540},"subject":"wip: replace validate_legacy","message":"wip: replace validate_legacy\n\nChange-Id: I920294342c9c2c0567796f345cbfa9e39bb1f1d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8329836b16efff6ff3dc8727872693a0c75c40c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8329836b16efff6ff3dc8727872693a0c75c40c0"}]},"branch":"refs/heads/master"},"9e77c31e6db66b6133b42537040abe768793ee66":{"kind":"REWORK","_number":8,"created":"2023-07-12 10:55:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/886970/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/70/886970/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/8"}}},"commit":{"parents":[{"commit":"de05151b9c8f058d164294b6951d2e30fd27e7a4","subject":"Expose hosts_allow/deny for rsyncd configuration.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/de05151b9c8f058d164294b6951d2e30fd27e7a4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-26 09:15:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-12 10:54:45.000000000","tz":540},"subject":"replace validate_legacy","message":"replace validate_legacy\n\nthe validate_legacy function is marked for deprecation in\nv9.0.0 from puppetlabs-stdlib.\n\nChange-Id: I920294342c9c2c0567796f345cbfa9e39bb1f1d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9e77c31e6db66b6133b42537040abe768793ee66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9e77c31e6db66b6133b42537040abe768793ee66"}]},"branch":"refs/heads/master"},"bda5dc4bf4e6ad223c833670e671bfbe7e4c61c8":{"kind":"REWORK","_number":9,"created":"2023-07-12 13:10:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/886970/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/70/886970/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/9"}}},"commit":{"parents":[{"commit":"de05151b9c8f058d164294b6951d2e30fd27e7a4","subject":"Expose hosts_allow/deny for rsyncd configuration.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/de05151b9c8f058d164294b6951d2e30fd27e7a4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-26 09:15:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-12 13:10:19.000000000","tz":540},"subject":"replace validate_legacy","message":"replace validate_legacy\n\nthe validate_legacy function is marked for deprecation in\nv9.0.0 from puppetlabs-stdlib.\n\nChange-Id: I920294342c9c2c0567796f345cbfa9e39bb1f1d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/bda5dc4bf4e6ad223c833670e671bfbe7e4c61c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/bda5dc4bf4e6ad223c833670e671bfbe7e4c61c8"}]},"branch":"refs/heads/master"},"3f802792c3d698c789190d745c7029be270e96c5":{"kind":"REWORK","_number":10,"created":"2023-07-12 15:40:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/886970/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/70/886970/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/10"}}},"commit":{"parents":[{"commit":"de05151b9c8f058d164294b6951d2e30fd27e7a4","subject":"Expose hosts_allow/deny for rsyncd configuration.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/de05151b9c8f058d164294b6951d2e30fd27e7a4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-26 09:15:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-12 15:40:04.000000000","tz":540},"subject":"replace validate_legacy","message":"replace validate_legacy\n\nthe validate_legacy function is marked for deprecation in\nv9.0.0 from puppetlabs-stdlib.\n\nChange-Id: I920294342c9c2c0567796f345cbfa9e39bb1f1d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3f802792c3d698c789190d745c7029be270e96c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3f802792c3d698c789190d745c7029be270e96c5"}]},"branch":"refs/heads/master"},"cc79abaecb1c0d50885ee2bb99634fa4d447e5c8":{"kind":"REWORK","_number":11,"created":"2023-09-14 06:48:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/886970/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/70/886970/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/11"}}},"commit":{"parents":[{"commit":"de05151b9c8f058d164294b6951d2e30fd27e7a4","subject":"Expose hosts_allow/deny for rsyncd configuration.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/de05151b9c8f058d164294b6951d2e30fd27e7a4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-26 09:15:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-14 06:48:43.000000000","tz":0},"subject":"replace validate_legacy","message":"replace validate_legacy\n\nthe validate_legacy function is marked for deprecation in\nv9.0.0 from puppetlabs-stdlib.\n\nChange-Id: I920294342c9c2c0567796f345cbfa9e39bb1f1d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cc79abaecb1c0d50885ee2bb99634fa4d447e5c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cc79abaecb1c0d50885ee2bb99634fa4d447e5c8"}]},"branch":"refs/heads/master"},"0f7bdd733aa4c090a8c6b1db3b92ce9184616186":{"kind":"REWORK","_number":12,"created":"2023-09-19 13:33:00.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/70/886970/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/70/886970/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/70/886970/12"}}},"commit":{"parents":[{"commit":"de05151b9c8f058d164294b6951d2e30fd27e7a4","subject":"Expose hosts_allow/deny for rsyncd configuration.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/de05151b9c8f058d164294b6951d2e30fd27e7a4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-26 09:15:17.000000000","tz":540},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2023-09-19 13:33:00.000000000","tz":0},"subject":"replace validate_legacy","message":"replace validate_legacy\n\nthe validate_legacy function is marked for deprecation in\nv9.0.0 from puppetlabs-stdlib.\n\nChange-Id: I920294342c9c2c0567796f345cbfa9e39bb1f1d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0f7bdd733aa4c090a8c6b1db3b92ce9184616186"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0f7bdd733aa4c090a8c6b1db3b92ce9184616186"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
