)]}'
{"id":"openstack%2Fpuppet-swift~901890","triplet_id":"openstack%2Fpuppet-swift~master~Ie6c0cf7956c228fead74dcb59e94111402303a56","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-12-05 15:19:01.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-12-05 13:44:15.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2023-12-04 23:02:04.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ie6c0cf7956c228fead74dcb59e94111402303a56","subject":"Add parameter type validations","status":"MERGED","created":"2023-11-26 11:28:40.000000000","updated":"2023-12-05 15:19:01.000000000","submitted":"2023-12-05 15:19:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901890","meta_rev_id":"aab98e821f5e30d81285b6c231612566969d9a01","_number":901890,"virtual_id_number":901890,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-05 15:19:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-12-04 23:02:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-12-05 13:44:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-12-05 13:44:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-26 11:56:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-27 15:29:09.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-12-03 19:33:29.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"adc345df82745cd127dd39b5ec9479e11c5453ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-26 11:28:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fb209a6fb9b62b96d04ef71e757fe0f1fe76430","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-26 11:35:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"df8ca0580ae5a9cab56a70ffa5bd1acba9afb5a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-26 11:37:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"014b6b83ca28926a29e4ff21925ea5ee74810892","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-26 11:56:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/86e817f3cc7f47feab4bbe1e5c5ad843\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7259ef8662e64b25affa91a79367b53d : FAILURE in 2m 07s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9c09ca3bc2d4571bdd124ce45289fd6 : FAILURE in 2m 18s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a8000edfe0a4301bab260f9535dea15 : FAILURE in 4m 22s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/48394ce504814816a8475378a210bc6c : FAILURE in 9m 39s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3f22f9a60e5439da84a2b37cb02466f : FAILURE in 10m 12s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab1fda39b64841e0a0c9b85bde83cd41 : FAILURE in 9m 59s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/edcb89341a154612aafc4be991ec6d1a : FAILURE in 10m 11s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/189dfc8554f644d09f12056b92ad3ee1 : FAILURE in 5m 58s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d77ac2436ed436e8b5f29b8fb907c78 : FAILURE in 8m 21s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/fded29fdea5d402aa0a3c04a5aad78d8 : FAILURE in 9m 39s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/770091bc9af649c5b28a139c8f90b647 : FAILURE in 7m 44s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0aa18c5be3e44e628bcc8d4e8c2f93fb : FAILURE in 5m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a61106b0e31b5c043f71efd981726edc60bca488","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-26 12:03:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d0f4f91df7eaee7637941d3c202c1de679d766e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-26 12:22:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/60bd9b688b4443a09b0c2c45d21dfa16\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53bfa167b208423e9c75162fec15cfec : FAILURE in 4m 47s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be50cea55d8c46d39eedea0c732dd586 : FAILURE in 3m 02s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b47ff55b5eb45d283410c78b66c270b : FAILURE in 5m 34s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/9498ce2787174e3fb1491ee2b3f95d02 : FAILURE in 12m 50s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16d6f463e7eb47449e9f1fa9364013e8 : FAILURE in 9m 42s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/5306b660c9a749fcaff457da069f41c6 : FAILURE in 13m 21s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df803d26ef5d40c49e73dff3095ae678 : FAILURE in 7m 48s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/106db656ecec4706b68c6648b56a790c : FAILURE in 6m 36s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8fa00df5240545a38689aaf87ac2c01b : FAILURE in 9m 16s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/347135ed2d844721bea61662631664cc : FAILURE in 8m 49s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf690af0e9a14d06a245b61edfefbfd3 : FAILURE in 5m 30s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac57353432e9418a8873da6a54dff06c : FAILURE in 5m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f8e5c3137ea969cd32914734eb5fea241ee281ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-26 12:40:09.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"17bf8323d9a2ce378492fe1cdfa68a05593d5215","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-26 12:56:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23399d4d69514088be712ec5c649c20a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d503dd01f4640889a4c333ea934f6b7 : FAILURE in 2m 16s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7ae7234e5ca4b6a8280ac9150dfe44c : FAILURE in 2m 15s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1caaefb3c5c4aed8008ea80d8e458eb : FAILURE in 5m 38s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/b71b7c9b2e2d4932bbb31fb03ba8bb07 : FAILURE in 10m 03s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15d9dec0235f45ac8f9c026bdae35afd : FAILURE in 7m 40s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/d28acfb80fac41a7bc08c24b36e97131 : FAILURE in 10m 04s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca6190c9769e410f84c944121b85702f : FAILURE in 7m 15s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b1dbf62f71e489e9f0613320ce1b35e : FAILURE in 5m 14s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b65a65922124144a131b1f3d455cc99 : FAILURE in 10m 47s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9e9fa9cd224447ebfbc7cdaf2b0394d : FAILURE in 9m 09s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbee21acece447ca92c9044b8cdd2bb4 : FAILURE in 6m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac99f8a16a5a42a4a2e43630334bff9b : FAILURE in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"49b7836102c6ceb206870fd0c3800a7a90206e9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-27 13:09:04.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"362a73ef95c1765ffb97410d1c820d586704e264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 14:30:17.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20f8cf6c09f84f0b9f427f8a0269630c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9fdd1caea3bf4b1fb40ea78e2733d985 : SUCCESS in 2m 39s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1eb430f814e54bbc9c51ca35a7452eae : SUCCESS in 2m 18s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/85922d9203fa4cb3bd42dbd2caa7ed09 : SUCCESS in 7m 41s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/c31b2795ebcd4b4b8fd6f94c204b5cea : SUCCESS in 13m 19s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61b4f803c4bd47919ab041ce0f594064 : SUCCESS in 8m 26s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5d8a4973bbf40d9855f75de9e81b4f5 : SUCCESS in 14m 07s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb4bd6d8d0814cdaa12e9b34cdfab655 : SUCCESS in 9m 43s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f715f75eca741d58f01270feb429202 : SUCCESS in 53m 37s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/90c4677107874e8dba12674ff3412893 : SUCCESS in 1h 15m 46s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/e52dae0832b84299a754f60ed044bc09 : SUCCESS in 26m 49s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f63074c9da6c4bbba191e024d244de72 : SUCCESS in 16m 24s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/359affc1346a4030b7f785a6b1fdf6a4 : SUCCESS in 16m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bc59e20f1208490c4b2c4614a619d34fdd970fc5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-11-27 15:29:09.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"45a29fc732722bb2dabee266fae3a60018a2b160","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-12-03 19:33:29.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"5907f29d5ece6a2640abb0926458591850b55324","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-03 20:49:53.000000000","message":"Patch Set 6: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/puppet-swift for 901890,6","accounts_in_message":[],"_revision_number":6},{"id":"8f8d69d425db79151ed55710e374ef8c233e7d59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-04 01:22:32.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":7},{"id":"e166bb5a18165cd5c846decc9a62a9d1798c4c91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 02:45:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f653fba240864b2c9b9583cabfe839e6\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2010d1f7baf148a5857121f17db144f9 : SUCCESS in 2m 21s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96436883f0d3473c8678c7ded59e51c0 : SUCCESS in 2m 10s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d622f6613a74bebba219907c423b4b1 : SUCCESS in 4m 34s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb69ad003c3a44c6a3aece627af465a0 : SUCCESS in 9m 41s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9c9cdc4cf3948ee9c2d901b435615c9 : SUCCESS in 10m 03s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/29327c70ba994c33923f0ff925e5017f : SUCCESS in 12m 49s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/401ab271b00d4a14bbd82e294e2ebc25 : SUCCESS in 8m 35s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f306d0f4bf364347a654054f73eaa81f : SUCCESS in 1h 08m 08s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/96dafd239b794e5d9ac986f7925463bf : SUCCESS in 1h 17m 16s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/656345a9857c470082ea377952e14a30 : SUCCESS in 26m 01s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90b721da08414f3684bd218452d6a0a3 : SUCCESS in 15m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33f9e4de33a444d58151ace2bb71b2b0 : SUCCESS in 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3a2a4eaa91e1799f1aa26cf0ec438d1c0632e3f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-04 10:46:47.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"7db4fd830b5af6fc4a56d1257181b41b46a79c22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 11:51:43.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55afe4daa5054d0c95a342721600e1c0\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89d41333285a4cbb87549fc9a8717e36 : SUCCESS in 2m 44s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c917904d81a74bdc87ff1a075ed592fb : SUCCESS in 2m 32s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/819b51267c5e4882abb5691355927843 : SUCCESS in 6m 12s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/9df52158abbd4a29add17f8cd6fb138b : SUCCESS in 13m 35s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70acd551367c49c0ab08a775c6ba07e3 : SUCCESS in 10m 30s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/515641338f8c4e548d0d7166bf18c88d : SUCCESS in 14m 33s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d9660eae4f544c4a7c5108d93b32a9a : SUCCESS in 7m 11s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afcd341cd3634bf7b6896b51036f7b30 : FAILURE in 54m 37s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bda74125f61a457491497a1428acd731 : SUCCESS in 1h 02m 28s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce1a34b96f58447a9f6227fc937cecff : SUCCESS in 18m 44s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/019be3f7e0394366b3a1837be6fd401e : SUCCESS in 14m 14s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdd784d8e754496ebfc7826d46e1b708 : SUCCESS in 18m 36s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"25b356683e3c95fb0edef7f241255161d4b4433b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-04 12:49:31.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"9046cdd68017a50fba71ef20374f434069f3739f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 14:07:38.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9419e9fc13b4ab6899726c63c8596be\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79a208fb665d43bda1f08ded45c46f04 : FAILURE in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77e27fc6600b4fad82041e4aa9d2ca75 : SUCCESS in 2m 26s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/28b68df4a08b490da3032ecd74c4f596 : SUCCESS in 5m 16s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e723ff491b745e2a341513bd3992c55 : FAILURE in 14m 25s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4cf25bf0fe9409fbfe699d1ac0edeb9 : FAILURE in 10m 02s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/e37af2ea88564295968fdc2173ad9798 : FAILURE in 13m 39s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76e64b6390374f93808fca93d1818946 : FAILURE in 10m 00s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a763b5deee24421af201f96c4ffda84 : FAILURE in 1h 05m 28s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3da60cc5f74145a780f29ff3147eaabf : SUCCESS in 1h 11m 50s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/75d4acd47e03417d975cb3c71a24312f : SUCCESS in 24m 44s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eac76a04157647b7abbb0e1ea934c8c5 : SUCCESS in 15m 56s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/350bfc552f7448e9861152a10b294dc8 : SUCCESS in 17m 55s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"644a9976cf5f756782813892765b9eb62bbadb01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-04 14:28:21.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"0ac85ba544ffd0314bc87730504536609034ac41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 15:43:54.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c5d8fefe05949b0904c4ae1703e0131\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48db5555b0b94eda9f0b9a9b089d5c54 : SUCCESS in 2m 26s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6dca870cb2d2452a9f0da63fc78b500f : SUCCESS in 2m 42s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/091afe679431487991ecfee6ba38154e : SUCCESS in 5m 26s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/a12173cd567044ffb86d47ac6e8a49a4 : SUCCESS in 11m 24s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e9e451750ca42338c1d1e613aac5e38 : SUCCESS in 9m 26s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/e812bb172d1648c78e88a161f66f3fb6 : SUCCESS in 12m 03s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2141726b1b6742bfbd044fcaaf8827e7 : SUCCESS in 9m 31s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb0d519b3f6240f0b9d3eeb246ddb5e0 : SUCCESS in 54m 41s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f47889dd6d94455ea5884bbc3e986525 : SUCCESS in 1h 13m 55s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/93cc4f0208134552959745d673fc726a : SUCCESS in 24m 00s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11d72c143bce4064a9865125ee7b2a1c : SUCCESS in 11m 45s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9e77d6cb9274e2e8094f520a5b0e40c : SUCCESS in 15m 53s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"2da3eae922af42a8303ee2338ce876730164e465","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-12-04 23:02:04.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"f4d82dee99fff6e70f218bc6039daa2197f925b3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-12-05 13:44:15.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"9f590bfc606ac332643309f889f72226a907012a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 13:45:21.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"6cf04b31dd62d5e09b0416a04295d4d428c90199","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 15:19:01.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d265ba4e7dd4e64a654260750c8e0c2\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0c4b7c350b248319c4cb073b00cf30a : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/429d948d9f0f4d17891e50c361462145 : SUCCESS in 2m 39s\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/0430e428963c44c8aef2ac053be45aa5 : SUCCESS in 13m 57s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b7ee12928854575acf2c966912dda9d : SUCCESS in 9m 23s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a6223afcc3249fd831c5e767a8cdc59 : SUCCESS in 11m 57s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/59f898e62b64496dbcb553f0de2c6bae : SUCCESS in 1h 27m 03s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/aed4c0d2f03e482b8c5a391068b95950 : SUCCESS in 30m 07s","accounts_in_message":[],"_revision_number":10},{"id":"aab98e821f5e30d81285b6c231612566969d9a01","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 15:19:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"f772e96e12500a8312c84275094010b51c886ee2","revisions":{"ad176b8ae0f96cfb63c30070ca70edf92bbf1c8a":{"kind":"REWORK","_number":1,"created":"2023-11-26 11:28:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/901890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/90/901890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/1"}}},"commit":{"parents":[{"commit":"0f7f6bdee8fda718cc8bee16fa0d3511cb727491","subject":"Merge \"Move out swift client from service packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0f7f6bdee8fda718cc8bee16fa0d3511cb727491"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 11:27:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 11:27:54.000000000","tz":540},"subject":"objectexpirer: Validate parameter types","message":"objectexpirer: Validate parameter types\n\n- manage_service and enabled should be booleans\n\n- pipeline should be an array with valid elements.\n\nChange-Id: Ie6c0cf7956c228fead74dcb59e94111402303a56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ad176b8ae0f96cfb63c30070ca70edf92bbf1c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ad176b8ae0f96cfb63c30070ca70edf92bbf1c8a"}]},"branch":"refs/heads/master"},"150f5df19892251040d9d2c7a5da4eeeb8a48352":{"kind":"REWORK","_number":2,"created":"2023-11-26 11:35:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/901890/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/90/901890/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/2"}}},"commit":{"parents":[{"commit":"0f7f6bdee8fda718cc8bee16fa0d3511cb727491","subject":"Merge \"Move out swift client from service packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0f7f6bdee8fda718cc8bee16fa0d3511cb727491"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 11:27:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 11:35:35.000000000","tz":540},"subject":"Add parameter type validations","message":"Add parameter type validations\n\nChange-Id: Ie6c0cf7956c228fead74dcb59e94111402303a56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/150f5df19892251040d9d2c7a5da4eeeb8a48352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/150f5df19892251040d9d2c7a5da4eeeb8a48352"}]},"branch":"refs/heads/master"},"48364c423aff5fd18c71cb9c41991a5aeccaf246":{"kind":"REWORK","_number":3,"created":"2023-11-26 11:37:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/901890/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/90/901890/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/3"}}},"commit":{"parents":[{"commit":"0f7f6bdee8fda718cc8bee16fa0d3511cb727491","subject":"Merge \"Move out swift client from service packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0f7f6bdee8fda718cc8bee16fa0d3511cb727491"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 11:27:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 11:37:38.000000000","tz":540},"subject":"Add parameter type validations","message":"Add parameter type validations\n\nChange-Id: Ie6c0cf7956c228fead74dcb59e94111402303a56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/48364c423aff5fd18c71cb9c41991a5aeccaf246"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/48364c423aff5fd18c71cb9c41991a5aeccaf246"}]},"branch":"refs/heads/master"},"efbe9ef77d4b1dbf2c72335079dc35cee726ff88":{"kind":"REWORK","_number":4,"created":"2023-11-26 12:03:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/901890/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/90/901890/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/4"}}},"commit":{"parents":[{"commit":"0f7f6bdee8fda718cc8bee16fa0d3511cb727491","subject":"Merge \"Move out swift client from service packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0f7f6bdee8fda718cc8bee16fa0d3511cb727491"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 11:27:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 12:03:12.000000000","tz":540},"subject":"Add parameter type validations","message":"Add parameter type validations\n\nChange-Id: Ie6c0cf7956c228fead74dcb59e94111402303a56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/efbe9ef77d4b1dbf2c72335079dc35cee726ff88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/efbe9ef77d4b1dbf2c72335079dc35cee726ff88"}]},"branch":"refs/heads/master"},"92889daad2b188ec0cd0c56e98ec536b947c373a":{"kind":"REWORK","_number":5,"created":"2023-11-26 12:40:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/901890/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/90/901890/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/5"}}},"commit":{"parents":[{"commit":"0f7f6bdee8fda718cc8bee16fa0d3511cb727491","subject":"Merge \"Move out swift client from service packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0f7f6bdee8fda718cc8bee16fa0d3511cb727491"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 11:27:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 12:40:01.000000000","tz":540},"subject":"Add parameter type validations","message":"Add parameter type validations\n\nChange-Id: Ie6c0cf7956c228fead74dcb59e94111402303a56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/92889daad2b188ec0cd0c56e98ec536b947c373a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/92889daad2b188ec0cd0c56e98ec536b947c373a"}]},"branch":"refs/heads/master"},"91e968de584a7ce370144bab9c31e60fa672f0f9":{"kind":"REWORK","_number":6,"created":"2023-11-27 13:09:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/901890/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/90/901890/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/6"}}},"commit":{"parents":[{"commit":"0f7f6bdee8fda718cc8bee16fa0d3511cb727491","subject":"Merge \"Move out swift client from service packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0f7f6bdee8fda718cc8bee16fa0d3511cb727491"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 11:27:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-27 13:08:55.000000000","tz":540},"subject":"Add parameter type validations","message":"Add parameter type validations\n\nChange-Id: Ie6c0cf7956c228fead74dcb59e94111402303a56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/91e968de584a7ce370144bab9c31e60fa672f0f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/91e968de584a7ce370144bab9c31e60fa672f0f9"}]},"branch":"refs/heads/master"},"a71874239cfc590f583abfb861d7e7832c3d530a":{"kind":"REWORK","_number":7,"created":"2023-12-04 01:22:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/901890/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/90/901890/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/7"}}},"commit":{"parents":[{"commit":"f0a77d2cc2fe0939afc13f0db59fff7988d72df4","subject":"Remove redundant hardcoded default of log_headers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f0a77d2cc2fe0939afc13f0db59fff7988d72df4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 11:27:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-04 01:21:58.000000000","tz":540},"subject":"Add parameter type validations","message":"Add parameter type validations\n\nSome parameters accept only certain parameter types/values. This adds\nvalidations in parameter input to return sensible errors to users.\n\nChange-Id: Ie6c0cf7956c228fead74dcb59e94111402303a56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a71874239cfc590f583abfb861d7e7832c3d530a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a71874239cfc590f583abfb861d7e7832c3d530a"}]},"branch":"refs/heads/master"},"e6c496809a8dd4f63d156cb8d46c4c4d77a43661":{"kind":"REWORK","_number":8,"created":"2023-12-04 10:46:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/901890/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/90/901890/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/8"}}},"commit":{"parents":[{"commit":"f0a77d2cc2fe0939afc13f0db59fff7988d72df4","subject":"Remove redundant hardcoded default of log_headers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f0a77d2cc2fe0939afc13f0db59fff7988d72df4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 11:27:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-04 10:46:35.000000000","tz":540},"subject":"Add parameter type validations","message":"Add parameter type validations\n\nSome parameters accept only certain parameter types/values. This adds\nvalidations in parameter input to return sensible errors to users.\n\nChange-Id: Ie6c0cf7956c228fead74dcb59e94111402303a56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e6c496809a8dd4f63d156cb8d46c4c4d77a43661"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e6c496809a8dd4f63d156cb8d46c4c4d77a43661"}]},"branch":"refs/heads/master"},"ffff86a5fb27d52494faf976acf0d76ac3ff0157":{"kind":"REWORK","_number":9,"created":"2023-12-04 12:49:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/901890/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/90/901890/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/9"}}},"commit":{"parents":[{"commit":"9cc6fe410bdf49fc34666898b62c94e062e695c1","subject":"container-reconciler: Support more tunables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9cc6fe410bdf49fc34666898b62c94e062e695c1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 11:27:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-04 12:49:13.000000000","tz":540},"subject":"Add parameter type validations","message":"Add parameter type validations\n\nSome parameters accept only certain parameter types/values. This adds\nvalidations in parameter input to return sensible errors to users.\n\nChange-Id: Ie6c0cf7956c228fead74dcb59e94111402303a56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ffff86a5fb27d52494faf976acf0d76ac3ff0157"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ffff86a5fb27d52494faf976acf0d76ac3ff0157"}]},"branch":"refs/heads/master"},"f772e96e12500a8312c84275094010b51c886ee2":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-12-04 14:28:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/901890/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/90/901890/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/90/901890/10"}}},"commit":{"parents":[{"commit":"53198ba27499ad31f308cb94625b11648c3f12a2","subject":"container-reconciler: Support more tunables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/53198ba27499ad31f308cb94625b11648c3f12a2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 11:27:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-04 14:28:21.000000000","tz":0},"subject":"Add parameter type validations","message":"Add parameter type validations\n\nSome parameters accept only certain parameter types/values. This adds\nvalidations in parameter input to return sensible errors to users.\n\nChange-Id: Ie6c0cf7956c228fead74dcb59e94111402303a56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f772e96e12500a8312c84275094010b51c886ee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f772e96e12500a8312c84275094010b51c886ee2"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
