)]}'
{"id":"openstack%2Fpuppet-swift~909004","triplet_id":"openstack%2Fpuppet-swift~master~I1d4b6ef73f7f29ed53cda3cf89ed5b522e4652c2","project":"openstack/puppet-swift","branch":"master","topic":"dfc","attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-05-02 17:19:50.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2024-02-28 15:05:30.000000000","reason":"removed on reply"},"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2025-05-02 17:19:52.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I1d4b6ef73f7f29ed53cda3cf89ed5b522e4652c2","subject":"drive-full-checker","status":"NEW","created":"2024-02-14 16:21:48.000000000","updated":"2025-05-02 17:19:52.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e1c376bf1b79e57c1580a983575847637f319026","_number":909004,"virtual_id_number":909004,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-26 16:18:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-02-28 15:05:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-14 17:24:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-23 16:38:10.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-05-02 17:19:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-05-02 17:19:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"096da21bc5315ff4cd554078f2e1dc1df9e78e72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-14 16:21:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d71bd3ff3b8e3e8fdd4fa5def6a459a20ba7187","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 17:24:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/68f418c3e6a44620b6e1305f91342dcf\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfd88221f20f4034a8c63fc3ee9aa8f0 : SUCCESS in 2m 10s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da23cacbab0a446d9061eb21d75c83d2 : SUCCESS in 2m 22s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4db73eb876c4313a0bf5e10753978c7 : SUCCESS in 6m 08s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/baea799421eb4cbcb009c958dffedb2d : FAILURE in 11m 21s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf826f7be27643beaf660b76353e6700 : FAILURE in 10m 12s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/644c87924c94416589ef74842539f105 : FAILURE in 13m 39s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c10387dfd8f4fdbb7770f1180477364 : FAILURE in 10m 54s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b90d6a86ac84537b910c0e17fd88a98 : FAILURE in 54m 49s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a4782c427394ee1acdf5d981850028c : FAILURE in 57m 53s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/a794b6a1b011412b92d2cef8edb54653 : FAILURE in 59m 05s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26e7f3bb806647f28314184f1b1e119e : SUCCESS in 13m 44s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe6508efaedc49e899ee54ea164e2e24 : SUCCESS in 17m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e8463c4d5d5cc4dce2454730abd792697a7154b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-14 20:11:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5a4193cb328078034af82a26eb571de8e8d9b5e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 21:10:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d503263173764872895d8cd651820e53\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c7582f726824f7ea76f4e953861f823 : SUCCESS in 2m 23s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36b4cca7997146f189322c6853a8f272 : SUCCESS in 2m 33s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b3d3dbcc20d40228bb70cbae1f7694f : SUCCESS in 5m 11s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/96355bbef9bd4aec872afb22a5efd521 : SUCCESS in 10m 44s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78ccea3b87bf4e5793a0b75af16d00a4 : SUCCESS in 8m 07s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/e06b4e6a038e4c35b9b53b210e5ea1c3 : SUCCESS in 10m 32s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8008448b6b442a48443a14cd3f746fa : SUCCESS in 8m 03s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eff0b9bae55d496a8dc1e143290ec55e : SUCCESS in 52m 27s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac3922d9bacf4ecfaa8f5e6a5812066f : FAILURE in 55m 40s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/290c6f9b588b41a89b1135ab963bdd35 : FAILURE in 57m 57s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52a51e3ab8854e5aa27ab04d1ad3fc20 : SUCCESS in 13m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad5c053a9e3f48c1bf7f99b61a6ab888 : SUCCESS in 13m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d0fc10c693cd184a2ff1ef69a3ff42fffe10c141","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-14 21:19:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a82df5f3da0ba6fbb54a4d46a0df499e69f6442d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 22:29:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/63f75df2f3124a029888e06502518a26\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38cfde0ebde54ab7b59433a3638a693c : SUCCESS in 2m 15s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/198ff14821d24b02aba9b084f10d3ac4 : SUCCESS in 2m 13s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3df83f0450e416cae2eebe55f2d6fec : SUCCESS in 6m 06s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/056b9923d3364be6a9448f6caa1b7ee8 : SUCCESS in 10m 36s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a550372ee5614fd09a876e8d32a9face : SUCCESS in 9m 26s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/89275c74751b437b8ed18cc7a3a08b0f : SUCCESS in 11m 54s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82e40ce047ef4a0c93a1f14ff8d141b8 : SUCCESS in 8m 35s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/927278d5e5444b4cbcb9b8d0b63f9022 : SUCCESS in 59m 42s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6ded5e4ab2f4cb8a1845bf04a909f51 : FAILURE in 59m 54s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ed6565bf99b4f44aa3df5781020b8db : FAILURE in 1h 03m 24s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1504c9f415774f41a07539fe51554e82 : SUCCESS in 14m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55749c3488574d159c2e01d8de0c70a0 : SUCCESS in 15m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2f2456e21f8405b27f74e6f00941ddd515977ec7","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-16 13:11:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8d4a4b31f43deff518943c0e58c40649413cde8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 14:22:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0de5c083194d4af58b8cb2a0eb01ca22\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d007ad47ed344110afd715798eb9a5b1 : SUCCESS in 2m 52s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2659cd8b1486462995e50d813e57e237 : SUCCESS in 2m 36s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b210b0d4c7a047b58c3b61ce0cea76e8 : SUCCESS in 5m 35s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d2a6d1089b24d00ae2e531ef810e59f : SUCCESS in 11m 27s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffc83af9ba494632a85d240d06c9446f : SUCCESS in 9m 02s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e475e1350974ddb92ae87e1b0df4f7f : SUCCESS in 12m 18s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f4df83c6a6b47b3bee47079c2bdf431 : SUCCESS in 8m 30s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ce80fd477194105850cd16a89c51058 : FAILURE in 52m 46s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5dc561e3d2814ffa87c5b1a93ce1714d : SUCCESS in 1h 09m 05s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/8aca84791726439fbcca0e9ce658aaba : SUCCESS in 21m 45s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1cc91ab10fba4b33923a369b2b6a498b : SUCCESS in 15m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a06d7ba43811480a82947302698ec6e4 : SUCCESS in 16m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2bc0e6c716a535321eb0a16def47f3778704c71f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-21 16:02:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d95c8789f3081bdd99d1d89c28d3b931cacde863","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 16:28:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc1da2639fb24607bf44f6dfc526793e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10cd509b09b7445599e07afbc2e3dd3a : FAILURE in 2m 55s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acb8a278f61246fea8102ac54f6a348a : FAILURE in 2m 38s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/bec881157c644457b43376b85d5cf258 : FAILURE in 5m 46s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4ecc5358f5445629f041822dafa2387 : FAILURE in 13m 29s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e7196dcd2ee4efead79df5add8d3d79 : FAILURE in 10m 20s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/00ac50edd16d4b199c28340469d72d8e : FAILURE in 14m 39s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3abc889fef048bab985531a0222f42b : FAILURE in 9m 00s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/071432576bed44ff92af9a53adf8eb75 : FAILURE in 7m 11s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d770581cd77f42dcb24c749b1fd4bc08 : FAILURE in 13m 26s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5fe5e084c264008909c06871b93859d : FAILURE in 10m 35s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7229f94b27848e9b190b6aea1acdeb6 : FAILURE in 5m 20s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a9313c609194557881d427d75ce330c : FAILURE in 7m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b673505d9a153e965522dc37acf826759cdeea41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-22 07:33:25.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"564a95c71f69d840c78478b2264333ff4e941018","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-22 08:00:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"344ef7a6d4fe7ee9acf206e314fb5ffa8b598718","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-22 09:09:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f0282b45fbc116d03a8b497afce450bd61492dba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 10:35:33.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38e22a278292456d866aa4c8543b04c7\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d15be82c59a458b8614efb8944f9c9f : FAILURE in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70254c79d09a4537ba009b0629192087 : SUCCESS in 2m 55s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/0733ed86b90a4018b13b576dd7c28414 : SUCCESS in 5m 18s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/e44a161b07974bc9853b3cc1b46d316b : SUCCESS in 14m 32s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f500adec3d7469f8fe62c42310b213b : SUCCESS in 11m 34s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/58447a4f3b814254a2fca1bba4ca44e9 : SUCCESS in 19m 09s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73e51abda3204bdebabdfef0fe30f7fb : SUCCESS in 10m 45s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c90341c1b1345e5937b98ec27c098e4 : SUCCESS in 1h 17m 21s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b029bdc91fe046899d76ac423ce44ce7 : SUCCESS in 1h 24m 17s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/20f5bbb04652400992388d0a10e9fafe : SUCCESS in 27m 41s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/990beef62cae420ebb82f64bb9fc8bd4 : SUCCESS in 16m 31s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95b95c23f5ad4cff97b1947d56394608 : SUCCESS in 16m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"eb05a02f5c415f1f1716340b4bd3d54f31935727","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-22 17:20:35.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"fe5ad68bec02e135b8fd53c2142f3346f3be2d77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 18:45:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9682a49f6b4450399c793f9e93f5eea\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/657be00d8ced42ddb86b16d411f0119f : SUCCESS in 2m 56s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/337f2e174aed4726844b24936889d606 : SUCCESS in 2m 57s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/0edd6a4d604743c1a916e66c8aae41c9 : SUCCESS in 6m 02s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d281fadd9ad48a8947d7269b372ede5 : SUCCESS in 12m 54s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b77743eb38fd4f2580473a7ea2ef57d6 : SUCCESS in 8m 50s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/6322da6d74bb43499efdf884c3948150 : SUCCESS in 12m 08s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b0996da943c40a8bfafac3a759a82bc : SUCCESS in 10m 16s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/369ef99af6214c08af5cb382f70c14c1 : FAILURE in 56m 51s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd5a91abc77047d0a780fcfc6767f5af : SUCCESS in 1h 19m 10s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/13ccac61a90048d0872618775dd540ab : SUCCESS in 30m 49s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04f915ce8b574fa5a36dc65589720ae6 : SUCCESS in 20m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/016aac4657504d7cbe3d986b955de873 : SUCCESS in 15m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af0e2ab8ccaa44028681783262033508 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":7},{"id":"8558d59eef387a766ee40e1a3d31a5bccb8c8bf4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-02-23 16:38:10.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"978d9069e681800425f12a818c7db4a1113b526b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-26 14:55:24.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"84e9f6b60118f7063dee7c55bef7cc1bc4c58737","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 16:18:43.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fea29b25e8a4f14a9daf181c55dfa85\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c036eff19064c859055834b5518c9ae : SUCCESS in 3m 08s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/948f0955730a4305a4ffa7b73acbc826 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4999749c04e4b1db80b53ee89d65591 : SUCCESS in 6m 03s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/e310ce16da524318b28d3ae92750b17f : SUCCESS in 13m 58s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/389cd16f7a544cdcb6f1135c0dcf2d18 : SUCCESS in 9m 51s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/721cc96923c84635b75cfd9a04745bdf : SUCCESS in 13m 36s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08f1f7ad09704fb58f941159c5fd680d : SUCCESS in 9m 50s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6716645e5cdb480e88f1cfa228973bd1 : SUCCESS in 1h 01m 16s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bdb135d6ffc54cd597e822476b7f7fea : SUCCESS in 1h 18m 04s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/c278fae84fa54a9c8b5f40764d76ace9 : SUCCESS in 27m 32s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37fb01de83af48baa56728aa5003d1cd : SUCCESS in 17m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ceec0c90ff6a4898b1b84c792e84bd86 : SUCCESS in 17m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/565303ebfdd446f6a8bcf17e2094f706 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":8},{"id":"017f68c37d88db1f6a066f9e19dc5beda2683363","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-02-28 15:05:30.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"99b475abf4f48a5c76a2f90f45969f756af5b5d9","revisions":{"14e42ed483ba0d655b45da61583379e533d644d4":{"kind":"REWORK","_number":1,"created":"2024-02-14 16:21:48.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/04/909004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/04/909004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/1"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-14 16:20:22.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-14 16:20:22.000000000","tz":60},"subject":"WIP: do not merge. drive-full-checker: implements dfc","message":"WIP: do not merge. drive-full-checker: implements dfc\n\nThis implements swift_drive_full_checker_config only for the moment.\nI intend to write a swift::storage::drive_full_checker class in this\npatch if the config file edition is validated.\n\nChange-Id: I1d4b6ef73f7f29ed53cda3cf89ed5b522e4652c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/14e42ed483ba0d655b45da61583379e533d644d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/14e42ed483ba0d655b45da61583379e533d644d4"}]},"branch":"refs/heads/master"},"73d15b1f7ce4c9ad85f78b9133dd5355b5a12972":{"kind":"REWORK","_number":2,"created":"2024-02-14 20:11:11.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/04/909004/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/04/909004/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/2"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-14 16:20:22.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-14 20:10:54.000000000","tz":60},"subject":"WIP: do not merge. drive-full-checker: implements dfc","message":"WIP: do not merge. drive-full-checker: implements dfc\n\nThis implements swift_drive_full_checker_config only for the moment.\nI intend to write a swift::storage::drive_full_checker class in this\npatch if the config file edition is validated.\n\nChange-Id: I1d4b6ef73f7f29ed53cda3cf89ed5b522e4652c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/73d15b1f7ce4c9ad85f78b9133dd5355b5a12972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/73d15b1f7ce4c9ad85f78b9133dd5355b5a12972"}]},"branch":"refs/heads/master"},"cd2538f6ed328ab670a22ab3f89054494fb347c2":{"kind":"REWORK","_number":3,"created":"2024-02-21 16:02:28.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/04/909004/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/04/909004/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/3"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-14 16:20:22.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-21 16:01:11.000000000","tz":60},"subject":"drive-full-checker","message":"drive-full-checker\n\nThis implements swift_drive_full_checker_config to configure the\ndrive-full-checker.conf from swift when this patch is merged:\nhttps://review.opendev.org/c/openstack/swift/+/907523\n\nThis also implements a new swift::storage::drive_full_checker\nclass to set values in drive-full-checker.conf. This class also\nsetups the cron job for drive-full-checker.\n\nFinally, a new option use_drive_full_checker is added to\nswift::storage, so that it writes to /etc/swift/rsyncd.conf\ninstead of directly in /etc/rsyncd.conf, so that it lets\nthe swift-drive-full-checker do the modification before\nwriting to /etc/rsyncd.conf.\n\nNote that we need a new release of the puppetlabs-rsync module\nthat includes the new conf_file parameter to write in\n/etc/swift/rsyncd.conf instead of /etc/rsyncd.conf directly.\n\nChange-Id: I1d4b6ef73f7f29ed53cda3cf89ed5b522e4652c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cd2538f6ed328ab670a22ab3f89054494fb347c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cd2538f6ed328ab670a22ab3f89054494fb347c2"}]},"branch":"refs/heads/master"},"aed500b6b4b05a6e27c038fe092ab67680f49a4b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-02-22 07:33:25.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/04/909004/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/04/909004/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/4"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-14 16:20:22.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-22 07:30:40.000000000","tz":60},"subject":"drive-full-checker","message":"drive-full-checker\n\nThis implements swift_drive_full_checker_config to configure the\ndrive-full-checker.conf from swift when this patch is merged:\nhttps://review.opendev.org/c/openstack/swift/+/907523\n\nThis also implements a new swift::storage::drive_full_checker\nclass to set values in drive-full-checker.conf. This class also\nsetups the cron job for drive-full-checker.\n\nFinally, a new option use_drive_full_checker is added to\nswift::storage, so that it writes to /etc/swift/rsyncd.conf\ninstead of directly in /etc/rsyncd.conf, so that it lets\nthe swift-drive-full-checker do the modification before\nwriting to /etc/rsyncd.conf.\n\nNote that we need a new release of the puppetlabs-rsync module\nthat includes the new conf_file parameter to write in\n/etc/swift/rsyncd.conf instead of /etc/rsyncd.conf directly.\n\nDepends-On: I9828f5652c65eceae13825d6c46ca4a96949c1b5\nChange-Id: I1d4b6ef73f7f29ed53cda3cf89ed5b522e4652c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/aed500b6b4b05a6e27c038fe092ab67680f49a4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/aed500b6b4b05a6e27c038fe092ab67680f49a4b"}]},"branch":"refs/heads/master"},"7150e17d4b17536bb853cdfe8fdfdb5610ed946e":{"kind":"REWORK","_number":5,"created":"2024-02-22 08:00:55.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/04/909004/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/04/909004/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/5"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-14 16:20:22.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-22 08:00:43.000000000","tz":60},"subject":"drive-full-checker","message":"drive-full-checker\n\nThis implements swift_drive_full_checker_config to configure the\ndrive-full-checker.conf from swift when this patch is merged:\nhttps://review.opendev.org/c/openstack/swift/+/907523\n\nThis also implements a new swift::storage::drive_full_checker\nclass to set values in drive-full-checker.conf. This class also\nsetups the cron job for drive-full-checker.\n\nFinally, a new option use_drive_full_checker is added to\nswift::storage, so that it writes to /etc/swift/rsyncd.conf\ninstead of directly in /etc/rsyncd.conf, so that it lets\nthe swift-drive-full-checker do the modification before\nwriting to /etc/rsyncd.conf.\n\nNote that we need a new release of the puppetlabs-rsync module\nthat includes the new conf_file parameter to write in\n/etc/swift/rsyncd.conf instead of /etc/rsyncd.conf directly.\n\nDepends-On: I9828f5652c65eceae13825d6c46ca4a96949c1b5\nChange-Id: I1d4b6ef73f7f29ed53cda3cf89ed5b522e4652c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7150e17d4b17536bb853cdfe8fdfdb5610ed946e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7150e17d4b17536bb853cdfe8fdfdb5610ed946e"}]},"branch":"refs/heads/master"},"96399d56132c65a02ee4cf33b3880470d5821bd8":{"kind":"REWORK","_number":6,"created":"2024-02-22 09:09:36.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/04/909004/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/04/909004/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/6"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-14 16:20:22.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-22 09:09:20.000000000","tz":60},"subject":"drive-full-checker","message":"drive-full-checker\n\nThis implements swift_drive_full_checker_config to configure the\ndrive-full-checker.conf from swift when this patch is merged:\nhttps://review.opendev.org/c/openstack/swift/+/907523\n\nThis also implements a new swift::storage::drive_full_checker\nclass to set values in drive-full-checker.conf. This class also\nsetups the cron job for drive-full-checker.\n\nFinally, a new option use_drive_full_checker is added to\nswift::storage, so that it writes to /etc/swift/rsyncd.conf\ninstead of directly in /etc/rsyncd.conf, so that it lets\nthe swift-drive-full-checker do the modification before\nwriting to /etc/rsyncd.conf.\n\nNote that we need a new release of the puppetlabs-rsync module\nthat includes the new conf_file parameter to write in\n/etc/swift/rsyncd.conf instead of /etc/rsyncd.conf directly.\n\nDepends-On: I9828f5652c65eceae13825d6c46ca4a96949c1b5\nChange-Id: I1d4b6ef73f7f29ed53cda3cf89ed5b522e4652c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/96399d56132c65a02ee4cf33b3880470d5821bd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/96399d56132c65a02ee4cf33b3880470d5821bd8"}]},"branch":"refs/heads/master"},"ab4f64b7e29cc5bb0a13c936d0d0c06c49385591":{"kind":"REWORK","_number":7,"created":"2024-02-22 17:20:35.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/04/909004/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/04/909004/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/7"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-14 16:20:22.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-22 17:19:35.000000000","tz":60},"subject":"drive-full-checker","message":"drive-full-checker\n\nThis implements swift_drive_full_checker_config to configure the\ndrive-full-checker.conf from swift when this patch is merged:\nhttps://review.opendev.org/c/openstack/swift/+/907523\n\nThis also implements a new swift::storage::drive_full_checker\nclass to set values in drive-full-checker.conf. This class also\nsetups the cron job for drive-full-checker.\n\nFinally, a new option use_drive_full_checker is added to\nswift::storage, so that it writes to /etc/swift/rsyncd.conf\ninstead of directly in /etc/rsyncd.conf, so that it lets\nthe swift-drive-full-checker do the modification before\nwriting to /etc/rsyncd.conf.\n\nNote that we need a new release of the puppetlabs-rsync module\nthat includes the new conf_file parameter to write in\n/etc/swift/rsyncd.conf instead of /etc/rsyncd.conf directly.\n\nDepends-On: I9828f5652c65eceae13825d6c46ca4a96949c1b5\nChange-Id: I1d4b6ef73f7f29ed53cda3cf89ed5b522e4652c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ab4f64b7e29cc5bb0a13c936d0d0c06c49385591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ab4f64b7e29cc5bb0a13c936d0d0c06c49385591"}]},"branch":"refs/heads/master"},"99b475abf4f48a5c76a2f90f45969f756af5b5d9":{"kind":"REWORK","_number":8,"created":"2024-02-26 14:55:24.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/04/909004/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/04/909004/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/04/909004/8"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-14 16:20:22.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-26 14:55:11.000000000","tz":60},"subject":"drive-full-checker","message":"drive-full-checker\n\nThis implements swift_drive_full_checker_config to configure the\ndrive-full-checker.conf from swift when this patch is merged:\nhttps://review.opendev.org/c/openstack/swift/+/907523\n\nThis also implements a new swift::storage::drive_full_checker\nclass to set values in drive-full-checker.conf. This class also\nsetups the cron job for drive-full-checker.\n\nFinally, a new option use_drive_full_checker is added to\nswift::storage, so that it writes to /etc/swift/rsyncd.conf\ninstead of directly in /etc/rsyncd.conf, so that it lets\nthe swift-drive-full-checker do the modification before\nwriting to /etc/rsyncd.conf.\n\nNote that we need a new release of the puppetlabs-rsync module\nthat includes the new conf_file parameter to write in\n/etc/swift/rsyncd.conf instead of /etc/rsyncd.conf directly.\n\nDepends-On: I9828f5652c65eceae13825d6c46ca4a96949c1b5\nChange-Id: I1d4b6ef73f7f29ed53cda3cf89ed5b522e4652c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/99b475abf4f48a5c76a2f90f45969f756af5b5d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/99b475abf4f48a5c76a2f90f45969f756af5b5d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
