)]}'
{"id":"openstack%2Fpuppet-swift~909114","triplet_id":"openstack%2Fpuppet-swift~master~I27386f6ac00b2efcfa5fc4f5ed3b86efc49c17fc","project":"openstack/puppet-swift","branch":"master","topic":"separate-log-name","attention_set":{},"removed_from_attention_set":{"3153":{"account":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"last_update":"2024-02-21 18:42:58.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-21 18:42:58.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2024-02-21 17:11:22.000000000","reason":"removed on reply"},"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2024-02-21 18:42:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I27386f6ac00b2efcfa5fc4f5ed3b86efc49c17fc","subject":"Per daemon log name","status":"MERGED","created":"2024-02-15 13:38:39.000000000","updated":"2024-02-21 18:43:57.000000000","submitted":"2024-02-21 18:42:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":29,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909114-separate-log-name","meta_rev_id":"1fe6f30733edb2332052a7afd7221af5fbf2c550","_number":909114,"virtual_id_number":909114,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-21 18:42:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-02-20 09:18:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-02-21 17:11:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-02-21 17:11:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"}],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-15 14:00:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-16 14:55:33.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-02-20 12:28:07.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-02-20 12:28:07.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"}],"messages":[{"id":"c44df34f682f560afedd0de3b776acc1b05a2a00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-15 13:38:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b25b3e34beac03e5adbedf016d32dcaca351e93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 14:00:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a843b00e5d1475d9514eb4bb82e443c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9574968cda2d43bf93fe2c3eabb92cf2 : SUCCESS in 2m 29s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f73aec7e43724209942d2cf445d0bc60 : FAILURE in 2m 29s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b18471c1ba3d40df8bb5559a184d6cb6 : FAILURE in 5m 15s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1ac27e1344c4738aad5f8986de58e8c : FAILURE in 11m 50s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc58d99ec72a4026857cdd2ba5ce4047 : FAILURE in 10m 52s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/13df458a73d649969ae073a329675747 : FAILURE in 12m 28s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6cb3e09ad581447e99e725e93b26b340 : FAILURE in 11m 59s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc18b122bb9040dc86152b065027f8ca : FAILURE in 7m 02s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb74f47d61af4e9697390f0f8f4b89db : FAILURE in 10m 37s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/7057ab5e84a241a3a600d3a426f8b8bd : FAILURE in 10m 11s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/547fffcc62e64f198e066c17634c1a5c : FAILURE in 5m 22s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e802b2b25a9b4fa7bf4860b2e289b00f : FAILURE in 6m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0a6abe1c305631a12e8cdc993bf3d7b0edd53fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-16 14:08:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8fdbf34040cb2eeaa62ba42edc7bb44a5fd86eec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 14:34:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3272d548b954a5891002a8519c799d8\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae955bf5a9104ebb94578f1d3afce086 : FAILURE in 3m 05s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/248a39117ed34b438654fcf8ef844613 : FAILURE in 2m 34s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a4a38b6724e410eba2eb10b78afadf4 : FAILURE in 5m 14s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/526cbd8fd57f442e93cf953f6c760eed : FAILURE in 13m 19s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7566f304786437891ac2e75e4159790 : FAILURE in 9m 04s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2e2e1a3b53844fa886bfce9ba4ca895 : FAILURE in 13m 53s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c00dd5f37dc34dd2bf35e4480e3b304b : FAILURE in 8m 39s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3c2675349304927ac517e4784b8ca76 : FAILURE in 6m 00s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aea2a070afa4455f969d9a1f88a3b1cf : FAILURE in 10m 37s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/6372073cb9854084ba200ead80039db0 : FAILURE in 9m 35s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da51d8851f724a9bae7cc5d76ca18bfb : FAILURE in 6m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80ef531333054fd39477aa765e2a6ff7 : FAILURE in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7ad361df251d38aea97c125a21d5937257c34209","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-16 14:55:33.000000000","message":"Patch Set 2: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b08251239872bcc0973beb980b5f61f4809591e4","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-19 12:02:16.000000000","message":"Patch Set 2:\n\n(8 comments)","accounts_in_message":[],"_revision_number":2},{"id":"21335d6fc8c8edb5b6089629010b4b47c2d171a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-19 12:02:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6bf45c1a815d1813a4c8fa5a6627cdccbcc0f374","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 12:21:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8455cd96633f4f9c9b2263a77c9cf535\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7438c90adbd444278d98971e7e42770a : FAILURE in 3m 36s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6418162852c4ba59ee8d568530e230c : SUCCESS in 3m 33s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b094cb5426e54ca992301f3223123eb4 : SUCCESS in 4m 46s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/28a05fda8c174c82bd0c5b004ea25394 : FAILURE in 13m 17s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1781799fe50248ec93a2f70dccae5ed3 : FAILURE in 9m 39s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/65ceb752c3fc4b549fdaaa1918525a88 : FAILURE in 12m 57s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7684df53122d4933bbcc0947b13ab7c6 : FAILURE in 10m 03s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac8fcf90f92945fda383bca5a36c43cb : FAILURE in 6m 55s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c01ae91b4dd64ad1bb939c783f70a4ea : FAILURE in 11m 01s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/d58ea87d280d4363a9dd4794c4e08de5 : FAILURE in 11m 56s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf0dc4b5899b4e848a049b05a584506c : FAILURE in 5m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87377128cf444e4ab9e22e3c6a120eff : FAILURE in 7m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"71d369e41e3bcda9e940cfc89783bc10a0bbb63e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-19 13:08:09.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"cc9db0d5eb7d8b93b1b0df8260dccb49543660bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-19 13:08:59.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent tree, and commit message as Patch Set 4.","accounts_in_message":[],"_revision_number":5},{"id":"c8c1efe29042724327fd469d41df7a86e1c6d85c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 14:25:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/539b38e3247641fc8a21917f94ecf619\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1515044b82c43d49dfa93bab348d2ec : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ea08fdceb2f425a83fb366785954909 : SUCCESS in 2m 29s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/6709ee1aba754dc3b6ee4fff0a6d14d9 : SUCCESS in 6m 15s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4a2cb75f3004840b67cc7fa75fe62e6 : SUCCESS in 13m 42s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/503dc8f3c95841b390fd1bc4373d4979 : SUCCESS in 10m 08s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/9208f200d4d8432689d3e113ae7d1616 : SUCCESS in 14m 01s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e3580d6556d4875a979aedf2d9fc8e3 : SUCCESS in 8m 59s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78de9e9e014d445eaf5a95b638b2fd6c : FAILURE in 56m 27s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0be4a74e3b3345a38b66ec5694c2f36b : SUCCESS in 1h 15m 38s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/db9ec41c32ec419d914a394bf53cb4ce : SUCCESS in 33m 41s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a82a5955b0b343babc3d607253cf74bf : SUCCESS in 17m 25s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3b3239c96a84a89b642e9d655c195dd : SUCCESS in 15m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0c099b609df38b89ae5a899347cf491d95f95a61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-19 14:55:30.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"8d313f7c58aea189f1108fb2f71610912b09262f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-19 15:08:17.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"bccd4b337fe8abb2d4cfd55ce0562ac26db9d293","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-19 15:13:06.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1391d2e69dbeaadd7d7b6a428eec54ce869c6bc0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-19 15:14:01.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"58f120e2f70a2359a14fe6eff1458698aab073b1","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-19 15:33:07.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"00d3261a99dfe4cacc793ab0626a672ea95899e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-19 15:33:17.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":8},{"id":"7d9dfb0aa7f7584eb568ef6c7a0e422c36a2d30d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 16:59:16.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7236ff1b58e044e8a447b0cfa0317e2b\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23822aab856042199dae5ec621b7aa66 : SUCCESS in 2m 52s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8af23c6b39e7484e8b5defb9660d365a : SUCCESS in 2m 57s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d23d6c9fbd194914b8cb9f34a42c6dc1 : SUCCESS in 7m 13s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/29dbb8fb20944c5295033789ade74520 : SUCCESS in 13m 48s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c94aed8a6b0465086c7b32ad8adce02 : SUCCESS in 9m 18s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/32e7ffbe093e4d40bb51cbcd7bd79f0c : SUCCESS in 13m 35s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fea1c08460a64849a67bd29992db907c : SUCCESS in 9m 32s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5915226524c941bc9e42fb9e42701ffc : SUCCESS in 56m 57s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8b7de6005084b7c97860de3e3769e18 : SUCCESS in 1h 18m 18s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1501bf403804c3bb185fc8530e81f7b : SUCCESS in 26m 22s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f3f1a8f34a047968bc3bf5418e7eed6 : SUCCESS in 14m 35s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f2246376bb54c9f82640f999bb84ec7 : SUCCESS in 16m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99ffe408b4e640a7a110c237479696a7 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":8},{"id":"a6fc519ac9de4476e1fb77cea780aa990f5e2bd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 00:48:31.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"4ff52b76feef98ca84e0866f9accaec6cadec582","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 00:48:36.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"9fff5801b461506cd6ab5af5e934f13143eb194f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 00:49:16.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"100626911f2500f328b5845ad8b10c9d13ff3fc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 01:55:28.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16407419d5784a519d4ca7957f7c1007\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1dfdd0817014c0b850b6884d70f9ab2 : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27613221173c4cfab8b4c243daebdead : SUCCESS in 2m 45s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/007f664ae57d4dfbb5bfdbd5cc26f37f : SUCCESS in 5m 16s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/751fe6390af14b1080f2a56a45fcf6bb : SUCCESS in 9m 42s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6023bfc3b4214a7197effe9979f6c235 : SUCCESS in 6m 59s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/89b1a4a789d64522ba8def2d2f018004 : SUCCESS in 9m 51s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a17f4e81f7b4783bbe886b9929bf94a : SUCCESS in 7m 19s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e684445927434be3afe25fcc8eca79e4 : SUCCESS in 52m 12s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea6b150ecb09411ab6ee7b3d4e96d675 : SUCCESS in 1h 05m 30s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/6df8e49344504a03a5dc79e987a01d0a : SUCCESS in 19m 47s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fb1ca8e27104ba5933b6f1e405b9e3a : SUCCESS in 11m 41s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98b3537f5dbe4afc86dbc243e6da8bd4 : SUCCESS in 11m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cdb320008304247ae6cfad3e0fe14d5 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":9},{"id":"6cd174d841da36e4edcb00fd1b9333b745afddf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 02:15:22.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"2e02c946c60813581ad0b23acc69521ec8288d90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 02:16:08.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"609d8702df4c07b1519305b341262b12eb34affe","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 02:17:55.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"874ab1668d1b7e4ab9a7b655ded28b1f14487e42","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 02:23:11.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"279beb6dc65d35cf3f35836bef476c2a2f0da818","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 02:25:30.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":12},{"id":"edc1d45347b5e54a4229644aaf6495cbaf27d92b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 02:25:44.000000000","message":"Patch Set 11: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"e368a81a92b2f910245a23a9e0cd6c42b265c849","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 03:06:12.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"93db7752fb37b95a6855b9d33745e5fad185eb54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 04:36:22.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00e9a2c9aa0d4058af32ee1f16e4ba8c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3959ff67466f45ee8b2a32f169318a8a : SUCCESS in 2m 29s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05aba80e95d74400a95ba9746d4bda64 : SUCCESS in 2m 52s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/f327766af3d643ad820fd45c92e7a70c : SUCCESS in 5m 31s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/9fff96c6417645fca8a8909b3781c3be : SUCCESS in 13m 27s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fa5f81645854fb59e8df23cf0ac6c17 : SUCCESS in 10m 04s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/bafc49aaa59844f0a5fa9e4b23ebbad5 : SUCCESS in 13m 57s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/873d435052a5479bb205b22756ce265c : SUCCESS in 10m 23s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59fbfe9345de4ce4938ce8304379140a : SUCCESS in 59m 02s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7875fd2126e04e97b5e2d40a6a02fc7f : SUCCESS in 1h 22m 24s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b153d77fbf14f748787b759071b03d9 : SUCCESS in 29m 49s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f5e5b6bfd4f488281da04b214155a7c : SUCCESS in 16m 02s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52e08fab627b46b4a1987e029858dfea : SUCCESS in 16m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44ead5df85234d56a87833c8d568727d : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":13},{"id":"707cf1f91634cdd9eadc41e32b0fd9d1a8d58651","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 09:18:14.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"cd211547ef22b1bfd1c6cc6f22b9e51a726bbf66","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-02-21 17:11:22.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"ee2dee5b33bf10b4be4b4a3cf3b47d8d36583391","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 17:12:32.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"861412bb1e7bb5c73070f6193e94200c649d33c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 18:42:58.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7affcc4e5c64f8cafdbf332247b2f09\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16df00fef70540e9ad787dfea6370ce3 : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d5d4703e3254b1a8c596985a21fbc0f : SUCCESS in 2m 47s\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/5894c3ef7d914056aa620a92901fad3f : SUCCESS in 12m 44s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa82282c0bc44479b0d8d49fdf1b2ac9 : SUCCESS in 8m 36s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2df712ffdb7847cca57153e122884370 : SUCCESS in 1h 29m 09s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/13e4c7b7a33c4ff6a59a37a955fe4023 : SUCCESS in 30m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4853a6e17144227ae4d4a1c7aa09e87 : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":13},{"id":"a1759acb14cf2eae3b2070021c827e6e83ff55c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 18:42:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"1fe6f30733edb2332052a7afd7221af5fbf2c550","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 18:43:57.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4d4acafbb0c400b84fd04aeef31f817\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5eb984cbc48748f7ab995e50c1eb062f : SUCCESS in 41s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"43035bc12e310c337061fcee96bb35e9eb06e501","revisions":{"c26f8ba7c1aa3dd33d818d37227df78c1aa19490":{"kind":"REWORK","_number":1,"created":"2024-02-15 13:38:39.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/909114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/14/909114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/1"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-15 13:36:54.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-15 13:36:54.000000000","tz":60},"subject":"Per daemon log name","message":"Per daemon log name\n\nCurrently, puppet-swift can only set log_name. However, it\u0027s possible\nwith swift to have each daemon log with a different name, making it\npossible to have logging for each daemon in a separate file, which is\noften a way more convenient for operators.\n\nThis patch sets a different log_name for each daemon if set to true.\nThe default behavior is left unchanged.\n\nChange-Id: I27386f6ac00b2efcfa5fc4f5ed3b86efc49c17fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c26f8ba7c1aa3dd33d818d37227df78c1aa19490"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c26f8ba7c1aa3dd33d818d37227df78c1aa19490"}]},"branch":"refs/heads/master"},"b66fdc7862f803c8e352f00e5556aacf488f1ba5":{"kind":"REWORK","_number":2,"created":"2024-02-16 14:08:28.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/909114/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/14/909114/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/2"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-15 13:36:54.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-16 14:08:18.000000000","tz":60},"subject":"Per daemon log name","message":"Per daemon log name\n\nCurrently, puppet-swift can only set log_name. However, it\u0027s possible\nwith swift to have each daemon log with a different name, making it\npossible to have logging for each daemon in a separate file, which is\noften a way more convenient for operators.\n\nThis patch sets a different log_name for each daemon if set to true.\nThe default behavior is left unchanged.\n\nChange-Id: I27386f6ac00b2efcfa5fc4f5ed3b86efc49c17fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b66fdc7862f803c8e352f00e5556aacf488f1ba5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b66fdc7862f803c8e352f00e5556aacf488f1ba5"}]},"branch":"refs/heads/master"},"f36613028f1433ade3e43cd98ad64fdd542c1947":{"kind":"REWORK","_number":3,"created":"2024-02-19 12:02:34.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/909114/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/14/909114/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/3"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-15 13:36:54.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-19 12:02:22.000000000","tz":60},"subject":"Per daemon log name","message":"Per daemon log name\n\nCurrently, puppet-swift can only set log_name. However, it\u0027s possible\nwith swift to have each daemon log with a different name, making it\npossible to have logging for each daemon in a separate file, which is\noften a way more convenient for operators.\n\nThis patch sets a different log_name for each daemon if set to true.\nThe default behavior is left unchanged.\n\nChange-Id: I27386f6ac00b2efcfa5fc4f5ed3b86efc49c17fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f36613028f1433ade3e43cd98ad64fdd542c1947"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f36613028f1433ade3e43cd98ad64fdd542c1947"}]},"branch":"refs/heads/master"},"32bbe33fc1f06f3da7bdc9df9c4b827bd271cc39":{"kind":"REWORK","_number":4,"created":"2024-02-19 13:08:09.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/909114/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/14/909114/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/4"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-15 13:36:54.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-19 13:07:59.000000000","tz":60},"subject":"Per daemon log name","message":"Per daemon log name\n\nCurrently, puppet-swift can only set log_name. However, it\u0027s possible\nwith swift to have each daemon log with a different name, making it\npossible to have logging for each daemon in a separate file, which is\noften a way more convenient for operators.\n\nThis patch sets a different log_name for each daemon if set to true.\nThe default behavior is left unchanged.\n\nChange-Id: I27386f6ac00b2efcfa5fc4f5ed3b86efc49c17fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/32bbe33fc1f06f3da7bdc9df9c4b827bd271cc39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/32bbe33fc1f06f3da7bdc9df9c4b827bd271cc39"}]},"branch":"refs/heads/master"},"cc0679c4b91aafa3a8a00703a19c95ebfd0598d3":{"kind":"NO_CHANGE","_number":5,"created":"2024-02-19 13:08:59.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/909114/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/14/909114/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/5"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-15 13:36:54.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-19 13:08:38.000000000","tz":60},"subject":"Per daemon log name","message":"Per daemon log name\n\nCurrently, puppet-swift can only set log_name. However, it\u0027s possible\nwith swift to have each daemon log with a different name, making it\npossible to have logging for each daemon in a separate file, which is\noften a way more convenient for operators.\n\nThis patch sets a different log_name for each daemon if set to true.\nThe default behavior is left unchanged.\n\nChange-Id: I27386f6ac00b2efcfa5fc4f5ed3b86efc49c17fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cc0679c4b91aafa3a8a00703a19c95ebfd0598d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cc0679c4b91aafa3a8a00703a19c95ebfd0598d3"}]},"branch":"refs/heads/master"},"30f485f4f3949909d4e4dd1bd909dff16b4e2d8c":{"kind":"REWORK","_number":6,"created":"2024-02-19 14:55:30.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/909114/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/14/909114/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/6"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-15 13:36:54.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-19 14:55:21.000000000","tz":60},"subject":"Per daemon log name","message":"Per daemon log name\n\nCurrently, puppet-swift can only set log_name. However, it\u0027s possible\nwith swift to have each daemon log with a different name, making it\npossible to have logging for each daemon in a separate file, which is\noften a way more convenient for operators.\n\nThis patch sets a different log_name for each daemon if set to true.\nThe default behavior is left unchanged.\n\nChange-Id: I27386f6ac00b2efcfa5fc4f5ed3b86efc49c17fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/30f485f4f3949909d4e4dd1bd909dff16b4e2d8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/30f485f4f3949909d4e4dd1bd909dff16b4e2d8c"}]},"branch":"refs/heads/master"},"da170a3f6494bc59d7ccd140d46492e11fd6969f":{"kind":"REWORK","_number":7,"created":"2024-02-19 15:13:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/909114/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/14/909114/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/7"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-15 13:36:54.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-19 15:12:56.000000000","tz":540},"subject":"Per daemon log name","message":"Per daemon log name\n\nCurrently, puppet-swift can only set log_name. However, it\u0027s possible\nwith swift to have each daemon log with a different name, making it\npossible to have logging for each daemon in a separate file, which is\noften a way more convenient for operators.\n\nThis patch sets a different log_name for each daemon if set to true.\nThe default behavior is left unchanged.\n\nChange-Id: I27386f6ac00b2efcfa5fc4f5ed3b86efc49c17fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/da170a3f6494bc59d7ccd140d46492e11fd6969f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/da170a3f6494bc59d7ccd140d46492e11fd6969f"}]},"branch":"refs/heads/master"},"d132ba16b1c9348467387c055edeea918b822cf2":{"kind":"REWORK","_number":8,"created":"2024-02-19 15:33:17.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/909114/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/14/909114/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/8"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-15 13:36:54.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-19 15:33:15.000000000","tz":60},"subject":"Per daemon log name","message":"Per daemon log name\n\nCurrently, puppet-swift can only set log_name. However, it\u0027s possible\nwith swift to have each daemon log with a different name, making it\npossible to have logging for each daemon in a separate file, which is\noften a way more convenient for operators.\n\nThis patch sets a different log_name for each daemon if set to true.\nThe default behavior is left unchanged.\n\nChange-Id: I27386f6ac00b2efcfa5fc4f5ed3b86efc49c17fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d132ba16b1c9348467387c055edeea918b822cf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d132ba16b1c9348467387c055edeea918b822cf2"}]},"branch":"refs/heads/master"},"4cf2acdf1b8485faf7ab0fc94ffdf16a5b39a0f7":{"kind":"REWORK","_number":9,"created":"2024-02-20 00:48:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/909114/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/14/909114/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/9"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-15 13:36:54.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-20 00:48:31.000000000","tz":0},"subject":"Per daemon log name","message":"Per daemon log name\n\nCurrently, puppet-swift can only set log_name. However, it\u0027s possible\nwith swift to have each daemon log with a different name, making it\npossible to have logging for each daemon in a separate file, which is\noften a way more convenient for operators.\n\nThis patch sets a different log_name for each daemon if set to true.\nThe default behavior is left unchanged.\n\nChange-Id: I27386f6ac00b2efcfa5fc4f5ed3b86efc49c17fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4cf2acdf1b8485faf7ab0fc94ffdf16a5b39a0f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4cf2acdf1b8485faf7ab0fc94ffdf16a5b39a0f7"}]},"branch":"refs/heads/master"},"980cd3ebf35ea8e5d2a8b1656c2e2101707cd1c5":{"kind":"REWORK","_number":10,"created":"2024-02-20 02:15:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/909114/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/14/909114/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/10"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-15 13:36:54.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-20 02:15:22.000000000","tz":0},"subject":"Per daemon log name","message":"Per daemon log name\n\nCurrently, puppet-swift can only set log_name. However, it\u0027s possible\nwith swift to have each daemon log with a different name, making it\npossible to have logging for each daemon in a separate file, which is\noften a way more convenient for operators.\n\nThis patch sets a different log_name for each daemon if set to true.\nThe default behavior is left unchanged.\n\nChange-Id: I27386f6ac00b2efcfa5fc4f5ed3b86efc49c17fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/980cd3ebf35ea8e5d2a8b1656c2e2101707cd1c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/980cd3ebf35ea8e5d2a8b1656c2e2101707cd1c5"}]},"branch":"refs/heads/master"},"0072859db888563f7ed8d042008d1c86512fc362":{"kind":"REWORK","_number":11,"created":"2024-02-20 02:16:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/909114/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/14/909114/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/11"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-15 13:36:54.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-20 02:16:07.000000000","tz":0},"subject":"Per daemon log name","message":"Per daemon log name\n\nCurrently, puppet-swift can only set log_name. However, it\u0027s possible\nwith swift to have each daemon log with a different name, making it\npossible to have logging for each daemon in a separate file, which is\noften a way more convenient for operators.\n\nThis patch sets a different log_name for each daemon if set to true.\nThe default behavior is left unchanged.\n\nChange-Id: I27386f6ac00b2efcfa5fc4f5ed3b86efc49c17fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0072859db888563f7ed8d042008d1c86512fc362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0072859db888563f7ed8d042008d1c86512fc362"}]},"branch":"refs/heads/master"},"ae204f54245ddb0d9d2a6c6926a58e61a4c50fee":{"kind":"REWORK","_number":12,"created":"2024-02-20 02:25:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/909114/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/14/909114/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/12"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-15 13:36:54.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-20 02:25:19.000000000","tz":540},"subject":"Per daemon log name","message":"Per daemon log name\n\nCurrently, puppet-swift can only set log_name. However, it\u0027s possible\nwith swift to have each daemon log with a different name, making it\npossible to have logging for each daemon in a separate file, which is\noften a way more convenient for operators.\n\nThis patch sets a different log_name for each daemon if set to true.\nThe default behavior is left unchanged.\n\nChange-Id: I27386f6ac00b2efcfa5fc4f5ed3b86efc49c17fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ae204f54245ddb0d9d2a6c6926a58e61a4c50fee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ae204f54245ddb0d9d2a6c6926a58e61a4c50fee"}]},"branch":"refs/heads/master"},"43035bc12e310c337061fcee96bb35e9eb06e501":{"kind":"REWORK","_number":13,"created":"2024-02-20 03:06:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/909114/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/14/909114/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/14/909114/13"}}},"commit":{"parents":[{"commit":"7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058","subject":"Merge \"storage: support timeout parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6179dad943e24bb4f9e6de2e81cf5bfd4bd058"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-15 13:36:54.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-20 03:06:01.000000000","tz":540},"subject":"Per daemon log name","message":"Per daemon log name\n\nCurrently, puppet-swift can only set log_name. However, it\u0027s possible\nwith swift to have each daemon log with a different name, making it\npossible to have logging for each daemon in a separate file, which is\noften a way more convenient for operators.\n\nThis patch sets a different log_name for each daemon if set to true.\nThe default behavior is left unchanged.\n\nChange-Id: I27386f6ac00b2efcfa5fc4f5ed3b86efc49c17fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/43035bc12e310c337061fcee96bb35e9eb06e501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/43035bc12e310c337061fcee96bb35e9eb06e501"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
