)]}'
{"id":"openstack%2Fpuppet-swift~909491","triplet_id":"openstack%2Fpuppet-swift~master~If5c287d8fd1325b08705133756eece6e41cfdc5f","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-22 14:24:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If5c287d8fd1325b08705133756eece6e41cfdc5f","subject":"Add more log options","status":"MERGED","created":"2024-02-20 02:34:31.000000000","updated":"2024-02-22 14:25:44.000000000","submitted":"2024-02-22 14:24:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909491","meta_rev_id":"f6b28e022c80db35728dff1b5bcf3bf3afb6f3df","_number":909491,"virtual_id_number":909491,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-22 14:24:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2024-02-21 17:11:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2024-02-22 12:54:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2024-02-22 12:54:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-20 04:13:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-21 17:11:33.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-02-22 12:54:20.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"8bdc19fa1dd651bd66f456ec7672a12c535191d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 02:34:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba29623e70c9867c7f96c054333b26e9b0afd908","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 04:13:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b40b36d4935d44afb00268c7f31f9fbf\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e704caa0a16548dfa324d27f5ba87ba9 : SUCCESS in 2m 58s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3aa3f1d718a43eea8324a9f37a8c50b : SUCCESS in 3m 32s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b66c0f71fdcf45a9979c6b91b5002bfe : SUCCESS in 5m 49s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3ad50e5768b431d85ea4d2f947cfd5b : FAILURE in 13m 29s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f661bf27800f49aca91ec0f75d58c0ca : FAILURE in 11m 12s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/0eb7323de5814744a47cb7e90dfd83b9 : FAILURE in 14m 32s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1ec4fa50c064d999a08580b407e5339 : FAILURE in 9m 50s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce9e7096814c4857a01ddf9f43709704 : SUCCESS in 1h 02m 57s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb93abb907d24e32b1db91ecb615fff2 : SUCCESS in 1h 23m 54s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/e79c88adf9eb4fa3af652e8e688b5e43 : SUCCESS in 29m 53s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08894d723c3345da85476002d800bac0 : SUCCESS in 19m 47s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ed2642a6363427c98771f7446da327d : SUCCESS in 16m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80d78df7e1f848fa9462be416e1ccf2a : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":1},{"id":"6ebbf200a5f49f435f3d0c44b454908f7f4053c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 04:53:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8e6402da3805f0f1bb3036264030332e198ff464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 06:20:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc1bee3396e249e6ad96b6b374bfbe33\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28b5c6f2929047e1834032955472aa4e : SUCCESS in 2m 56s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b108ce4513240b888417be1b0a8fe00 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/206781e78c8d43b8bb3dd927e2edb739 : SUCCESS in 5m 33s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a290f5f5dbc4c4c9fad00e7eb2235b8 : SUCCESS in 11m 00s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b5575bd044d4f9e8d85fc7f0fcfe96d : SUCCESS in 7m 06s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/fef11e20cf7f46b49c3b34add60042ec : SUCCESS in 13m 30s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e055a56449b943c5ade098cbe0e32209 : SUCCESS in 9m 40s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8744b0c7444248cda3547f3b9a9bece0 : FAILURE in 48m 15s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c98925ab69ee41158fe63a75eca6332c : SUCCESS in 1h 21m 00s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/aaab228008ac4b83a6b87c1bc70fe62c : SUCCESS in 22m 00s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/edadedafc8a0403fb13290c41fb5d5db : SUCCESS in 17m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a301b7182554636b4d2c99acb7a5d8c : SUCCESS in 15m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7382ff6ab0e4a0e86f81b5ce40acb24 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"d38b5c5bf6cd411a3b3ddf7785c5d3c7772240a7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-02-21 17:11:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a5c764b9c099483154cca46e8d532ee916e12449","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-02-22 12:54:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"64357546e3131dc3ede5cea35018e6fe148b0e2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 12:55:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"40c54c4a5e29914ca383f3b7c6a0a7cb1db54bed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 14:24:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ef3ccae54434364bc5ec965abf9e111\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d6e26a20e9b4e3d90bc7631f6b694e3 : SUCCESS in 5m 24s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4e139b9149440f3a3568feaa7a32595 : SUCCESS in 2m 46s\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/77a4cd5e5f2843e0806843b70e40b81f : SUCCESS in 13m 07s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54264c5998634889aa55ae386177ac35 : SUCCESS in 9m 52s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4d3a1c83a394152b1d4e818314f23d5 : SUCCESS in 1h 24m 45s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3890c73f32844c08bfae51b75db2d75 : SUCCESS in 25m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9a2cbefd3564f28a5af119a7c09e3ea : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"ed4bbfc3d3be3ffc9cf3eec9fe95cfcb9cdda07c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 14:24:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f6b28e022c80db35728dff1b5bcf3bf3afb6f3df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 14:25:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5540644c94c9417a86b2f76b8b02d014\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6288641b25b45f6a42dc51731834ad7 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dde528e8f5fb82d33947d4c96492a96326913b5c","revisions":{"7921c873af8ab41001f0707cc9aa34d6a32973bf":{"kind":"REWORK","_number":1,"created":"2024-02-20 02:34:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/909491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/91/909491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/91/909491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/91/909491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/91/909491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/91/909491/1"}}},"commit":{"parents":[{"commit":"0e3f254347f93ece93617f882ec4b7b6334968a2","subject":"Merge \"Replace deprecated `stub` from rspec-mocks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0e3f254347f93ece93617f882ec4b7b6334968a2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-20 02:32:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-20 02:34:16.000000000","tz":540},"subject":"Add more log options","message":"Add more log options\n\nThis adds a few new log options for object-expirer and\ncontainer-reconciler so that users can set consistent logging options.\n\nThe default log level of container-reconciler is effectively changed\nfrom local0 to local2, to use the consistent log facility in all\nservices.\n\nChange-Id: If5c287d8fd1325b08705133756eece6e41cfdc5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7921c873af8ab41001f0707cc9aa34d6a32973bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7921c873af8ab41001f0707cc9aa34d6a32973bf"}]},"branch":"refs/heads/master"},"dde528e8f5fb82d33947d4c96492a96326913b5c":{"kind":"REWORK","_number":2,"created":"2024-02-20 04:53:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/909491/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/91/909491/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/91/909491/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/91/909491/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/91/909491/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/91/909491/2"}}},"commit":{"parents":[{"commit":"0e3f254347f93ece93617f882ec4b7b6334968a2","subject":"Merge \"Replace deprecated `stub` from rspec-mocks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0e3f254347f93ece93617f882ec4b7b6334968a2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-20 02:32:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-20 04:53:41.000000000","tz":540},"subject":"Add more log options","message":"Add more log options\n\nThis adds a few new log options for object-expirer and\ncontainer-reconciler so that users can set consistent logging options.\n\nThe default log level of container-reconciler is effectively changed\nfrom local0 to local2, to use the consistent log facility in all\nservices.\n\nChange-Id: If5c287d8fd1325b08705133756eece6e41cfdc5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/dde528e8f5fb82d33947d4c96492a96326913b5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/dde528e8f5fb82d33947d4c96492a96326913b5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
