)]}'
{"id":"openstack%2Fpuppet-swift~909638","triplet_id":"openstack%2Fpuppet-swift~master~I708e57d2904499613c85e720f75919483e80aaee","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-29 16:55:56.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I708e57d2904499613c85e720f75919483e80aaee","subject":"ext4: Sync interface with xfs","status":"MERGED","created":"2024-02-21 07:26:18.000000000","updated":"2024-02-29 18:26:41.000000000","submitted":"2024-02-29 18:25:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909638","meta_rev_id":"878af1b4ef6da275314ed338b86503a20d2e73e8","_number":909638,"virtual_id_number":909638,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-29 18:25:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-02-27 00:19:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2024-02-27 15:42:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-02-29 16:55:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-21 09:00:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-21 17:12:03.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-02-27 15:42:36.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"3fd7e383b5d5e44fc99b8bd7dc144ac7a0f5a27c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-21 07:26:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"103bf608237ccdaf4565bf846785c24305d3f045","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 09:00:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/069b0f46f799487c8d1aef87400e4bf4\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b9cedb4ebda42eaacf4f8927e54a445 : FAILURE in 2m 30s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a70d7676208413c89503812418340a0 : SUCCESS in 3m 14s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/4528c4092ec745afb0bd5e4b131138fb : SUCCESS in 4m 46s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8056b3eb3b24906982e32bc4f9f8efd : FAILURE in 12m 42s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7732eda1442646b9bda0bf38f95efac7 : FAILURE in 10m 33s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/98f604dda80a44c2a8241c51677d465d : FAILURE in 12m 36s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddc55adff19d4049a3132abc0fac45fb : FAILURE in 7m 19s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0cc30c82bdb453891fda136e52cac9f : SUCCESS in 1h 02m 20s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/50bdd58a45da4c7fb3c53d69e082891c : SUCCESS in 1h 28m 40s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce8986327c504470a9c01e99afae3fca : SUCCESS in 25m 31s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0cb643839a6484e81714582af757214 : SUCCESS in 15m 45s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/814ff520f6614306abf267237cc31a87 : SUCCESS in 12m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a3e4f8bfadf4f819c0e3572c6e925e2 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"41febd067c8bc9db1b81d389750fe5b48d10d638","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-21 09:29:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5c769c7d01837f298f2a5b2d6ba773f000b9f999","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-21 10:23:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3ced909a575a0c1f0eb840c9319db74dc432719c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 11:39:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b76ffd1941284b148f5235a8cf165308\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/962014c04c9f44149dee3e153efdebf1 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47f2c623b0c0484a84b13ce59f804953 : SUCCESS in 2m 42s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/35ae31c47c57439eb713fbd8c89a201d : SUCCESS in 7m 27s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/f22c9b60c2bf497bb028b5d1eaa0d5a2 : SUCCESS in 12m 51s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22c8d459ba534e78aad6478ff71b9736 : SUCCESS in 9m 53s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/b44f498203144c15ad13a7a2ff4ef34b : SUCCESS in 14m 05s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55df9747506d4a5085a8b32f8ddfe383 : SUCCESS in 9m 52s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e4873ef455a40658254a5821df54633 : SUCCESS in 1h 02m 50s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e83247b104a64415bb43f72a0e10d135 : SUCCESS in 1h 09m 21s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a5bf5df7b00402b9f52dac0bf2f158d : SUCCESS in 28m 07s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd5705be6aee4a86837ae2cc69cf9fdb : SUCCESS in 16m 34s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b16525912d99496eb3857efa97d8c4a4 : SUCCESS in 16m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e951346df214aef8afe2384009e05f6 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":3},{"id":"9e52468eb549a57275fad34dcb3dd8b7ee4f47b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-21 11:43:36.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c1315081a0bca91124be8bb7337bab3359511624","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 13:00:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5e20cfc92de4304bd3686b50d627f08\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90fc72eca4db46d4861dee13d77d4e3f : SUCCESS in 2m 50s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d5876d954f24636a41410f26bf11324 : SUCCESS in 3m 15s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b0c48aeb3cc429481f32919f14a57cc : SUCCESS in 5m 16s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/790971f4472847cf827ab50b9f01c87a : SUCCESS in 12m 21s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60b82c979c1f441cb85deaf2047a1ec0 : SUCCESS in 8m 21s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/104366f5ec614e6193a6912ca3aa0e89 : SUCCESS in 12m 30s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56aee8dd900e4167b7f7c55fba3c21c9 : SUCCESS in 8m 51s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/637061978a2e43249001b38e9a261658 : SUCCESS in 59m 07s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d9420ccecee4053b9a8144cd481c2b3 : SUCCESS in 1h 14m 37s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc450a5e36b445dc8106ca405cb36d0d : SUCCESS in 28m 02s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34b5b73b3c92475789ea40b8ec8955b9 : SUCCESS in 17m 00s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66fdefbeebde44cabafeda0d09ca26fb : SUCCESS in 15m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61d48dade6874215be7f61d090ac1b88 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":4},{"id":"18309b83af46760b28c2239223c174311fce5bba","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-02-21 17:12:03.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3ac2993e17d5a2c1bbb4e578b9ed5a7f55d25004","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-27 00:19:42.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"016fa5adef213f8d671a9550852b7b154d84cf05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 01:49:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c975a2d196743f7a9910af27205b683\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ed46b26ccd94b60a3b4b82e2fae6a9e : SUCCESS in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7baa7ab25af4effa1ce1b3c3245a3c9 : SUCCESS in 2m 21s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b7894a8ecf04781838adf04aa92890a : SUCCESS in 6m 16s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9f9986c55324cb8b3bbe9e273cd9938 : SUCCESS in 13m 01s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f07d37e18644f61b8dab65c196a64eb : SUCCESS in 9m 47s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e94792d4f834699a3e74c9e04589a89 : SUCCESS in 13m 18s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a976fe05e314df580a2e6e58dffd047 : SUCCESS in 10m 36s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68d29ba0e6a848479eda9be7590bcf92 : SUCCESS in 1h 18m 35s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/26221eadfe1e41648fa4664e35f82cd2 : SUCCESS in 1h 24m 00s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/d403d6fa852240f08ba596e682b79274 : SUCCESS in 28m 21s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5976bf30638482598f580202cbd489b : SUCCESS in 13m 50s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db3462ad1d484feea547bd66248a6cde : SUCCESS in 15m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b75210540d34b73b0a7159bc4b3f3ef : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":5},{"id":"71adbfbac975782ed46899ed9ed8e83c9c8cbd4c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-02-27 15:42:36.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"8d05459b6491b7335b861cc12073067f15a560a2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-29 16:55:56.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8cb4c2b2eaf75c60d88391dbdac5fff3637cb259","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 16:56:47.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5cebf1d3286d01e945eb7c6236ee61e95b0bac70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 18:25:41.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5d59d3e6b5344cfa13a5ccc2a5f12b5\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62168651f1a34582a5a53bc14bab60db : SUCCESS in 3m 02s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bcf159fdc2d44e3e8d08558da124e0e2 : SUCCESS in 3m 00s\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/71110e2d58874255b59198ca8e19e668 : SUCCESS in 15m 22s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09e54a4916c94b509e0bb1a74004b870 : SUCCESS in 10m 50s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b4a4c6ca60f491084ac30c22ca90e04 : SUCCESS in 1h 22m 05s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/88c1363a519c43e48c2354945118a0aa : SUCCESS in 27m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9773b4364b1f49ebbf9f8be40f18f394 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":5},{"id":"19f3e41a3ffd172c87491aa1e11974da34cc797b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 18:25:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"878af1b4ef6da275314ed338b86503a20d2e73e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 18:26:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b92798a2f1c9481c89d9a0c48d2f650f\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94794d4a961d448cbe55d9e01985a41f : SUCCESS in 41s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5d955ed0031de60d8e668e6a2efe8a919f0219a2","revisions":{"3b7cb4fa51438100e145bae1cab6dda3bc2bf5b9":{"kind":"REWORK","_number":1,"created":"2024-02-21 07:26:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/909638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/38/909638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/1"}}},"commit":{"parents":[{"commit":"40614cfcd496bd7f3eec85a2733f027f6bf967d9","subject":"xfs: Support mount by label","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/40614cfcd496bd7f3eec85a2733f027f6bf967d9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-21 07:16:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-21 07:26:08.000000000","tz":540},"subject":"ext4: Sync interface with xfs","message":"ext4: Sync interface with xfs\n\nThis fixes the inconsistent interfaces between swift::storage::ext4 and\nswift::storage::xfs, so that users can use both implementations more\neasily without looking into actual interfaces in detail.\n\nChange-Id: I708e57d2904499613c85e720f75919483e80aaee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3b7cb4fa51438100e145bae1cab6dda3bc2bf5b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3b7cb4fa51438100e145bae1cab6dda3bc2bf5b9"}]},"branch":"refs/heads/master"},"81178b0301ce1c28e77ae16f4ec5dacd0c3b4d1c":{"kind":"REWORK","_number":2,"created":"2024-02-21 09:29:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/909638/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/38/909638/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/2"}}},"commit":{"parents":[{"commit":"40614cfcd496bd7f3eec85a2733f027f6bf967d9","subject":"xfs: Support mount by label","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/40614cfcd496bd7f3eec85a2733f027f6bf967d9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-21 07:16:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-21 09:28:50.000000000","tz":540},"subject":"ext4: Sync interface with xfs","message":"ext4: Sync interface with xfs\n\nThis fixes the inconsistent interfaces between swift::storage::ext4 and\nswift::storage::xfs, so that users can use both implementations more\neasily without looking into actual interfaces in detail.\n\nChange-Id: I708e57d2904499613c85e720f75919483e80aaee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/81178b0301ce1c28e77ae16f4ec5dacd0c3b4d1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/81178b0301ce1c28e77ae16f4ec5dacd0c3b4d1c"}]},"branch":"refs/heads/master"},"f7998166fecaadb9f9cffc2f978f4161e9205684":{"kind":"REWORK","_number":3,"created":"2024-02-21 10:23:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/909638/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/38/909638/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/3"}}},"commit":{"parents":[{"commit":"40614cfcd496bd7f3eec85a2733f027f6bf967d9","subject":"xfs: Support mount by label","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/40614cfcd496bd7f3eec85a2733f027f6bf967d9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-21 07:16:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-21 10:23:15.000000000","tz":540},"subject":"ext4: Sync interface with xfs","message":"ext4: Sync interface with xfs\n\nThis fixes the inconsistent interfaces between swift::storage::ext4 and\nswift::storage::xfs, so that users can use both implementations more\neasily without looking into actual interfaces in detail.\n\nChange-Id: I708e57d2904499613c85e720f75919483e80aaee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f7998166fecaadb9f9cffc2f978f4161e9205684"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f7998166fecaadb9f9cffc2f978f4161e9205684"}]},"branch":"refs/heads/master"},"d1f12fb2d8f913de628f0d5ff2877cc09a3b1746":{"kind":"REWORK","_number":4,"created":"2024-02-21 11:43:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/909638/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/38/909638/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/4"}}},"commit":{"parents":[{"commit":"40614cfcd496bd7f3eec85a2733f027f6bf967d9","subject":"xfs: Support mount by label","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/40614cfcd496bd7f3eec85a2733f027f6bf967d9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-21 07:16:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-21 11:43:23.000000000","tz":540},"subject":"ext4: Sync interface with xfs","message":"ext4: Sync interface with xfs\n\nThis fixes the inconsistent interfaces between swift::storage::ext4 and\nswift::storage::xfs, so that users can use both implementations more\neasily without looking into actual interfaces in detail.\n\nChange-Id: I708e57d2904499613c85e720f75919483e80aaee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d1f12fb2d8f913de628f0d5ff2877cc09a3b1746"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d1f12fb2d8f913de628f0d5ff2877cc09a3b1746"}]},"branch":"refs/heads/master"},"5d955ed0031de60d8e668e6a2efe8a919f0219a2":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-02-27 00:19:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/909638/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/38/909638/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/38/909638/5"}}},"commit":{"parents":[{"commit":"9efd303bcbc8b32e6156ef0fff0eb5c571123eee","subject":"xfs: Support mount by label","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9efd303bcbc8b32e6156ef0fff0eb5c571123eee"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-21 07:16:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-27 00:19:42.000000000","tz":0},"subject":"ext4: Sync interface with xfs","message":"ext4: Sync interface with xfs\n\nThis fixes the inconsistent interfaces between swift::storage::ext4 and\nswift::storage::xfs, so that users can use both implementations more\neasily without looking into actual interfaces in detail.\n\nChange-Id: I708e57d2904499613c85e720f75919483e80aaee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5d955ed0031de60d8e668e6a2efe8a919f0219a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5d955ed0031de60d8e668e6a2efe8a919f0219a2"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
