)]}'
{"id":"openstack%2Fpuppet-swift~909648","triplet_id":"openstack%2Fpuppet-swift~master~Ib60576fc593f36c36a0e0bd03611a568e21913e9","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-29 18:30:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib60576fc593f36c36a0e0bd03611a568e21913e9","subject":"disk: Expose more options","status":"MERGED","created":"2024-02-21 11:55:05.000000000","updated":"2024-02-29 18:31:44.000000000","submitted":"2024-02-29 18:30:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909648","meta_rev_id":"76fcb2cb1a909a1c37b2e95f4c64412131daab68","_number":909648,"virtual_id_number":909648,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-29 18:30:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-02-27 15:43:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-02-28 15:05:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-02-28 15:05:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-21 13:24:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-27 15:43:07.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-02-28 15:05:23.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"a8e7d360d5c9b5e2c0682bf9595692174bf4b17b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-21 11:55:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fe7d385c669ca9ca21090011b29e53a2b3e4387","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 13:24:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/45e701bf32844f14b21749b937b5eb94\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8eb3c7e1c4f49099f38d16ba1f0da36 : SUCCESS in 2m 34s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff1cf2255f924619a6f93e6567b9a165 : SUCCESS in 2m 36s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc15b2ec01d543c4b33b25c823edfba6 : SUCCESS in 5m 42s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/323ee0873c1042d39273356de501bd5d : FAILURE in 11m 18s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b2bc6c323b34683b34848e3160069b9 : FAILURE in 10m 37s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/0858b7706e874870b7e5906590feb55b : FAILURE in 12m 57s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fff60f3b77d6489dab9e4e396b42a3f2 : FAILURE in 8m 45s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0b6eff2dc1b41709906bda4ecb880b6 : SUCCESS in 57m 21s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c15e41bd8b845c6bd880e7f12eeef5d : SUCCESS in 1h 19m 46s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/eee7db3d97904582bbe35361af8b4fa5 : SUCCESS in 28m 17s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bdaba8742154add84acfbdf9079fbdf : SUCCESS in 14m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30a5413a289a45fc9fc4d6f5b790ed6d : SUCCESS in 15m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/379d911342b9405c92cbb1c6ab3d2ecc : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6b1edfb2c5b50855852c6e88b3ac2db339ab9a74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-21 15:26:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9357c5ee83848fb5bf94a9a59c3308aa424fe188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 16:57:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f5bd15510d84a3891c4ec4034c0055f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04638ead24be4c4fb558b4b58ca2bfcc : SUCCESS in 2m 32s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9455f43407a4e73997354c0b901d134 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/0449aac970024064a2e6a8098ea17c81 : SUCCESS in 5m 35s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f83599d4a5040729791e921f3ec0a72 : FAILURE in 13m 39s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bc4d572ff8746efbcc2b88bb97bda5c : FAILURE in 11m 50s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/71cfec54ef36434baf969a63619621d3 : FAILURE in 13m 59s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ea1aec9352f4a6cb86def02d3b8325b : FAILURE in 9m 41s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3693699e85ac43bb8b82bd8f58306a31 : SUCCESS in 1h 05m 08s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b8a04e4b855423ebf1ad71558f4f6a8 : SUCCESS in 1h 28m 46s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/a54cbb8a97fc44b094cd9a7795683ab6 : SUCCESS in 27m 52s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eeec615a7da1494480c8d7a2a8f2dfa4 : SUCCESS in 18m 41s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7abddfeb691c41f9adc346d68c02aeb0 : SUCCESS in 16m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/400b40976f3c4439a05b9d00c98ec380 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":2},{"id":"52ad30a7a8db8a5ffa8c211e21d785d8fce69184","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-22 01:43:13.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9e74c7bfa46457f7c5669406ad5221df150793ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 03:01:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b37c2116c174aa2b012ef25d93eb0cf\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6fe03780d69419fb9f896c46ff435ef : SUCCESS in 2m 21s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/738a96d573994e3d89cc1de2714a95cd : SUCCESS in 2m 23s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/4636bfaeb0cf47c9b2a9808360356612 : SUCCESS in 5m 23s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/b31faa6e598d4b0eb5cfd555c15dbac8 : FAILURE in 10m 53s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ffc720d154e49ca817ca2cdf218998d : FAILURE in 7m 31s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/1bf9b75db2de45dbb30ed8c5b18a0084 : FAILURE in 10m 20s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3bdabb967d04c849082d33286c40f72 : FAILURE in 10m 26s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4305c365a8ce4997899d2aaea38770c4 : SUCCESS in 58m 45s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/656a93f5b0d644a2a1b07f174fd43315 : SUCCESS in 1h 17m 38s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ef7a003bef6446eb222c57ed9462161 : SUCCESS in 27m 37s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/501004fbfc9d44cb9d06ff56beda9419 : SUCCESS in 15m 22s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09d35134f2764d609009ce2be79cd390 : SUCCESS in 13m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31fc985a2b6e4bb89cea1cf08379045a : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":3},{"id":"0c4a9a6b97484dfd5839c32be4cb147e6c30ac2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-27 00:19:48.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"726873058992822d55393fdfb0bb749823cdeee2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-27 00:24:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6909097e29f831e0e05440df6be78b42bc07cbbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 01:58:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62d814191d7d45a8a11ff3162bd58ca4\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebccd1d4d24f4656b43197425b19342f : SUCCESS in 2m 57s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a86aff964a5408c8d8d720f506f5187 : SUCCESS in 3m 15s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/032fcc90c2a44c7e92f442bae3cae979 : SUCCESS in 6m 23s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/b67926ad122d4118951069dc83dad8e8 : SUCCESS in 12m 40s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3cfc4498ccc428e962f4b67ab651048 : SUCCESS in 9m 52s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/c115635e6fae404f81c889b44e35adc2 : SUCCESS in 13m 14s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82682114f7cd4ee48f948d88489159bb : SUCCESS in 9m 38s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7056464349d446258614fc394b20ef14 : SUCCESS in 52m 29s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b7908638ca84dfbbf3c50d654b5ebd8 : SUCCESS in 1h 28m 11s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1e221183e5e455397c421e1352d55de : SUCCESS in 24m 24s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ab9587acaa444fcaf78a2833af520d0 : SUCCESS in 16m 25s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e39e4bf488647f8b9fbf7306d2db506 : SUCCESS in 12m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1fa273af3ff4050ac0acdaece1f6b7d : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":5},{"id":"992b73a4a30ecc1655df3952a0c84dff07551316","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-02-27 15:43:07.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"9cd48fca05877065770430ca70dc424234bd2d96","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-02-28 15:05:23.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1fb9a82bb172ed0035b91e09241f37369984b1d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 16:56:53.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"97ad778b2d329926bc1d828a7b0e263b7f510154","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 18:30:48.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2bd9f00ff0044baa5eee2e72ca811b4\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9144b60b206443784632f1376a84fc6 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21439513216e44389acfc12917fcb675 : SUCCESS in 3m 48s\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/959ad464199b4fd19bab50459a413b8a : SUCCESS in 13m 54s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b0c17c6e41345808477ecfaa5572190 : SUCCESS in 12m 28s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d27270603b54b66b3909b6cb358ce4a : SUCCESS in 1h 25m 42s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e10c2a1ac5846af8d07027a993c1031 : SUCCESS in 28m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e0da5e580364319be1c4d0c8f19640c : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":5},{"id":"6354871a196f5dd7e641cdc459b3a499f3983779","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 18:30:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"76fcb2cb1a909a1c37b2e95f4c64412131daab68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 18:31:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/018fa5b9c8d3463fb389892ac61cd0d6\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df262571b07c4f978de5b93a58d24a70 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e888e9c0d30c12d28ac176f46fdb6fcf69eed479","revisions":{"40748452eec2bdb75dc6b6f043d86c7f2c7179e1":{"kind":"REWORK","_number":1,"created":"2024-02-21 11:55:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/909648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/48/909648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/1"}}},"commit":{"parents":[{"commit":"d1f12fb2d8f913de628f0d5ff2877cc09a3b1746","subject":"ext4: Sync interface with xfs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d1f12fb2d8f913de628f0d5ff2877cc09a3b1746"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-21 11:52:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-21 11:54:36.000000000","tz":540},"subject":"disk: Expose more options","message":"disk: Expose more options\n\n... so that users can use this wrapper with more flexible customization\nabout filesystems and mounts. Also, this introduces the new filesystem\ntype parameter so that this common implementation can be used when ext4\nis used.\n\nChange-Id: Ib60576fc593f36c36a0e0bd03611a568e21913e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/40748452eec2bdb75dc6b6f043d86c7f2c7179e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/40748452eec2bdb75dc6b6f043d86c7f2c7179e1"}]},"branch":"refs/heads/master"},"215c9995168b71dd46ccd648b6a74b7e9c5e9827":{"kind":"REWORK","_number":2,"created":"2024-02-21 15:26:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/909648/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/48/909648/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/2"}}},"commit":{"parents":[{"commit":"d1f12fb2d8f913de628f0d5ff2877cc09a3b1746","subject":"ext4: Sync interface with xfs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d1f12fb2d8f913de628f0d5ff2877cc09a3b1746"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-21 11:52:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-21 15:26:27.000000000","tz":540},"subject":"disk: Expose more options","message":"disk: Expose more options\n\n... so that users can use this wrapper with more flexible customization\nabout filesystems and mounts. Also, this introduces the new filesystem\ntype parameter so that this common implementation can be used when ext4\nis used.\n\nChange-Id: Ib60576fc593f36c36a0e0bd03611a568e21913e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/215c9995168b71dd46ccd648b6a74b7e9c5e9827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/215c9995168b71dd46ccd648b6a74b7e9c5e9827"}]},"branch":"refs/heads/master"},"18be4c69676ec1ca7f70026e0057252655e493be":{"kind":"REWORK","_number":3,"created":"2024-02-22 01:43:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/909648/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/48/909648/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/3"}}},"commit":{"parents":[{"commit":"d1f12fb2d8f913de628f0d5ff2877cc09a3b1746","subject":"ext4: Sync interface with xfs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d1f12fb2d8f913de628f0d5ff2877cc09a3b1746"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-21 11:52:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-22 01:42:48.000000000","tz":540},"subject":"disk: Expose more options","message":"disk: Expose more options\n\n... so that users can use this wrapper with more flexible customization\nabout filesystems and mounts. Also, this introduces the new filesystem\ntype parameter so that this common implementation can be used when ext4\nis used.\n\nChange-Id: Ib60576fc593f36c36a0e0bd03611a568e21913e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/18be4c69676ec1ca7f70026e0057252655e493be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/18be4c69676ec1ca7f70026e0057252655e493be"}]},"branch":"refs/heads/master"},"45d30083c108d0918736e6f39c670bba9d60ba30":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-02-27 00:19:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/909648/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/48/909648/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/4"}}},"commit":{"parents":[{"commit":"5d955ed0031de60d8e668e6a2efe8a919f0219a2","subject":"ext4: Sync interface with xfs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5d955ed0031de60d8e668e6a2efe8a919f0219a2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-21 11:52:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-27 00:19:48.000000000","tz":0},"subject":"disk: Expose more options","message":"disk: Expose more options\n\n... so that users can use this wrapper with more flexible customization\nabout filesystems and mounts. Also, this introduces the new filesystem\ntype parameter so that this common implementation can be used when ext4\nis used.\n\nChange-Id: Ib60576fc593f36c36a0e0bd03611a568e21913e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/45d30083c108d0918736e6f39c670bba9d60ba30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/45d30083c108d0918736e6f39c670bba9d60ba30"}]},"branch":"refs/heads/master","description":"Rebase"},"e888e9c0d30c12d28ac176f46fdb6fcf69eed479":{"kind":"REWORK","_number":5,"created":"2024-02-27 00:24:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/909648/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/48/909648/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/48/909648/5"}}},"commit":{"parents":[{"commit":"5d955ed0031de60d8e668e6a2efe8a919f0219a2","subject":"ext4: Sync interface with xfs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5d955ed0031de60d8e668e6a2efe8a919f0219a2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-21 11:52:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-27 00:24:20.000000000","tz":540},"subject":"disk: Expose more options","message":"disk: Expose more options\n\n... so that users can use this wrapper with more flexible customization\nabout filesystems and mounts. Also, this introduces the new filesystem\ntype parameter so that this common implementation can be used when ext4\nis used.\n\nChange-Id: Ib60576fc593f36c36a0e0bd03611a568e21913e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e888e9c0d30c12d28ac176f46fdb6fcf69eed479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e888e9c0d30c12d28ac176f46fdb6fcf69eed479"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
