)]}'
{"id":"openstack%2Fpuppet-swift~916351","triplet_id":"openstack%2Fpuppet-swift~master~I32711cc8c3d354c395439cb8d83daa9b4393dd39","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-04-25 15:24:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I32711cc8c3d354c395439cb8d83daa9b4393dd39","subject":"proxy: Support keepalive_timeout","status":"MERGED","created":"2024-04-19 01:29:13.000000000","updated":"2024-04-25 15:25:52.000000000","submitted":"2024-04-25 15:24:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916351","meta_rev_id":"a87133ef64ed28fa20799f6ca6adf1ca44bcac17","_number":916351,"virtual_id_number":916351,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-25 15:24:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-04-19 06:31:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-04-25 14:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-04-25 14:00:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-19 02:50:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-19 06:31:42.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-04-25 14:00:15.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"feccf84f3a1904c1f9f4bc34beee505ce133cf3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 01:29:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"599aa1a1f603cdf229933d30520d99a2f0d98bc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 02:50:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f197625aca9a409eb80ea495976b4970\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9370d96b7394c198443bc251439b731 : SUCCESS in 2m 27s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/345429ef6ddd46c7befcbf0868641fd8 : SUCCESS in 2m 36s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/0331323f7ce742fa8cd916f7b7aa605a : SUCCESS in 4m 31s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/94b786b2cd6444919fd258215f965ac5 : SUCCESS in 12m 01s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f428505a414a4e148b73e0f79d5173fd : SUCCESS in 8m 48s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f61c86662f34b7fa01132f89c4bfd1e : SUCCESS in 12m 26s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81ed85aa50294083a850ca03a597d148 : SUCCESS in 10m 06s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a07434a8e7f949519820e993e9ac2ef5 : FAILURE in 39m 39s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc84825716794fe6a6fc69ca613e618c : SUCCESS in 1h 16m 09s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c8b39f19b9d44a6a0364ad63b209078 : SUCCESS in 24m 25s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dad62cfbf73f495f910806b07bc11cad : SUCCESS in 13m 31s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8fbe589d8e654951a9dceb6883356f1a : SUCCESS in 15m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7b7a19fcd564e4196d81e20750d3f0d : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":1},{"id":"297b1aec33cc2bab4a7ab95de69941c07ff8c9d1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-04-19 06:31:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3f6a9600cbbdb85c288dda8c46ba74ea6d76b6c3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-04-25 14:00:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9f0f040d458bfb3e6b1c683250cfb9190908bd76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 14:01:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"861d99b3ad017ec267d59fc71c44b93459c37476","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 15:24:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6285642b10a74c4c9bfb329307ba151d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f408ae3639644b789ec5c3135e828b4c : SUCCESS in 3m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8b8c98a5458421686c8728dffbee89f : SUCCESS in 2m 50s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/405f541cbeba4e20b20f5bf349dea6be : SUCCESS in 16m 16s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffae3dd92fbd421d9b3179f850073233 : SUCCESS in 11m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b2e5fa071194e929ecdecd2cac8e4cb : SUCCESS in 1h 05m 39s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/c58f7aeaeeb24df082ca60c79cae17c7 : SUCCESS in 23m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8159d03da64945e983ae2f93cfa5df5a : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9541335a78322dd406b26afa65bc3cfd932d6680","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 15:24:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a87133ef64ed28fa20799f6ca6adf1ca44bcac17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 15:25:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3841f000803f475daecd772108f17e0b\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba14656d74fb4846bb01b543775455db : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9f986eb0a6bf05aaeda007854f5f9572ecfe0be","revisions":{"d9f986eb0a6bf05aaeda007854f5f9572ecfe0be":{"kind":"REWORK","_number":1,"created":"2024-04-19 01:29:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/916351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/916351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/916351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/916351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/916351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/916351/1"}}},"commit":{"parents":[{"commit":"3af1a20fe010caa1af233a9c2572699d80ca8704","subject":"Prepare 2024.1 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3af1a20fe010caa1af233a9c2572699d80ca8704"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 01:26:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 01:27:49.000000000","tz":540},"subject":"proxy: Support keepalive_timeout","message":"proxy: Support keepalive_timeout\n\n... which was added in 2.32.0.\n\nDepends-on: https://review.opendev.org/873744\nChange-Id: I32711cc8c3d354c395439cb8d83daa9b4393dd39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d9f986eb0a6bf05aaeda007854f5f9572ecfe0be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d9f986eb0a6bf05aaeda007854f5f9572ecfe0be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
