)]}'
{"id":"openstack%2Fpuppet-swift~919527","triplet_id":"openstack%2Fpuppet-swift~master~I714fe28547ea8308d7e82b66687ea0f24f816cda","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-14 15:51:59.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2024-05-14 14:27:49.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I714fe28547ea8308d7e82b66687ea0f24f816cda","subject":"Ensure app/filter factory options exist","status":"MERGED","created":"2024-05-13 14:27:15.000000000","updated":"2024-05-14 15:51:59.000000000","submitted":"2024-05-14 15:51:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919527","meta_rev_id":"c9039c3a1937f3ddb6164c22155a8bb4f70f5fe9","_number":919527,"virtual_id_number":919527,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-14 15:51:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2024-05-14 14:27:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2024-05-14 09:13:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2024-05-14 14:27:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-13 15:00:14.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-05-13 15:15:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-14 09:13:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"3b576a848cc5d07b85732b8fdb2ea8d1a99b0868","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-13 14:27:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65c076b00d77d22db0319a9b617c7b9c0d3a823d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-05-13 15:00:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"288736fbed99974fecb61c5a1253a7823164ae65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 15:15:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4d305144502428683dd22759e8826cd\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53decef2c1e6443aa0db98f099365789 : SUCCESS in 3m 14s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12a159c1e71c472aa845d1fd98312b24 : SUCCESS in 2m 32s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b914f65033b4261abe2430a8d73951c : SUCCESS in 5m 46s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fcfc86ff6c44220bb31585da757d737 : SUCCESS in 14m 10s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4007ab0594b4d9fbaf45b89adbe7ceb : SUCCESS in 9m 01s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/04f94b46646143dbaf204c96da1efddc : SUCCESS in 15m 28s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/743eab07bc624f9db4a3cbea555dd215 : SUCCESS in 9m 29s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22ccfb95862a457daeb51a9f7a23e46c : FAILURE in 35m 23s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc840790b45444be86f6725913187ce9 : FAILURE in 46m 04s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/1bd80d62ad3f47678e1e58b6b2a433a8 : FAILURE in 21m 14s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/834b6135f6ad4bdb83eade0f1fe3d7a7 : FAILURE in 15m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d86e4cb6ddc54e018fb3f269fe201d9f : FAILURE in 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d1fa0d18dddb6e6892b77d1a90980f7a6a15242e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-14 00:24:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9e6c7cd24a8ba9b504a6e2e648fa2219fe33aca6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 01:49:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7ca8505244e47a0b7fac497e84e3259\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9e2d41431cb443498ae49d3b93a62ed : SUCCESS in 3m 10s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8644602291ed4d29b00b5b669e45d51a : SUCCESS in 2m 32s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc409819202945aeba28f35fcbf9e00a : SUCCESS in 6m 10s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/630abd78b0ec44c0938ab13a6a2b5e0b : SUCCESS in 12m 54s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad56ed86c0104562a6f6a01e88e0bec0 : SUCCESS in 8m 07s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6c8db98e6c8488892649539d31387ca : SUCCESS in 10m 32s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b210077da73943d78f3273a655485f70 : SUCCESS in 10m 37s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0747bd1661374b4dba35ab89b73c3025 : SUCCESS in 1h 06m 08s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/97e441ef8ef44b95acb5d6e83d025989 : SUCCESS in 1h 19m 40s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/b74ffb49a89c49b7a92e71c7c8cbae17 : SUCCESS in 20m 33s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb1dfcc9b9de430192eb7104d9fbb463 : SUCCESS in 15m 11s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db14f800d9af4ae3a4483a6c3d9c182b : SUCCESS in 15m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"40eaf3d21a0ab8332e5a2f508b844ce8df3f1522","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-14 02:15:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6f70623a4c6172cc77224cbcc26b32032a2a59ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-14 02:15:58.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"db288df2294c2b4c491fb41f816f6fc7897d6b44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 03:17:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b8e229f20524a60b5a3cc2ee0967a8c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/497b5a7095f4418487fd92c4e8b2d83c : SUCCESS in 4m 08s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abfa97f6808645c3bc2bb6fd5c49e37d : SUCCESS in 2m 50s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/bef54feb1d874ddabd52f82f0d4e9071 : SUCCESS in 5m 48s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/b51a686b9ac240819dd882f862127f18 : SUCCESS in 14m 25s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3ea491224a342d3af6815e27c301bda : SUCCESS in 10m 48s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/a875b18f6e2349e698a3c4c74b36c34b : SUCCESS in 15m 12s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ffa444a48804b758511a5554254d21f : SUCCESS in 10m 55s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/169e9e61e2f0448f90d6c595e8640d93 : FAILURE in 52m 08s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/87a62e8ba38c44ec9e29f820c5bbb26d : FAILURE in 53m 57s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/046d548a15874e9188fb1d66c8f5a9ce : FAILURE in 22m 01s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d63bd1ae72a84d99a74669ef4a4d4908 : FAILURE in 17m 56s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2410d1c6df7458287311b5fa07f656e : FAILURE in 17m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f380142ad7bd5230d12a7ecfcfe5893c8a0bd6ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-14 03:43:52.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a39ffb065def42402c828c99751f110d22818351","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 05:13:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56a9cb4dbf8d4e31aa82d139bc511efe\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03a8c1ec47cf43a0a52547fd1d4c81ba : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11cf35abbe5d4423bd1034cf6d5a40b0 : SUCCESS in 3m 12s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3068a81f33e479a9803bc38838c24a8 : SUCCESS in 6m 02s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/9cf2d91d5ed34b43a147386a4548a175 : SUCCESS in 13m 21s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8745458682cd40f9943500e37105bafc : SUCCESS in 8m 05s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f86a06dcdef4ac393b964e478d11884 : SUCCESS in 13m 30s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af655147e30f4ca294bf6f2191cca38b : SUCCESS in 7m 55s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76150bcdd3a34af2bd85eda4b2d011ee : SUCCESS in 54m 13s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/08b9786567704f229f31ca9cdaa5e823 : SUCCESS in 1h 12m 48s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/b41188538788496d9794f90d8e52dce9 : SUCCESS in 27m 15s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/636d18efd62d46c1b31967204df12a04 : SUCCESS in 15m 52s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98baeb6f94694f2b90d2f1c409a5090e : SUCCESS in 15m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"438a2c085e8d8b0513b8017b29e284acdb21e6d6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-05-14 09:13:23.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a83f1327037f09ad80326713598ac7dab104003d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-05-14 14:27:49.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1710312523669ba7da470f52216cb16578158562","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 14:28:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8bf190faecaf90cdb0242635187dfe806d00b60e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 15:51:59.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d900934726640ae9c1175959190e91c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3786907e88284eb2a9510295ba1d9b23 : SUCCESS in 2m 33s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/243dbabc2e934da8b3cd4486271c1eac : SUCCESS in 2m 52s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e978e38930e46c4ac82e35d98ce3494 : SUCCESS in 11m 30s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73d4b9c26f214584a97abe1ce094d1e2 : SUCCESS in 10m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/07091f69ec354d5081158e3720802067 : SUCCESS in 1h 18m 11s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/59865cf209854442900ad298c558a264 : SUCCESS in 29m 27s","accounts_in_message":[],"_revision_number":5},{"id":"c9039c3a1937f3ddb6164c22155a8bb4f70f5fe9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 15:51:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a123fd166bfdb161071d9869c69a6769ad8f5990","revisions":{"bf4d4967a2dc19b72ada2e7bec4e1975f398d99e":{"kind":"REWORK","_number":1,"created":"2024-05-13 14:27:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/919527/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/27/919527/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/1"}}},"commit":{"parents":[{"commit":"2655744dcf1c103a007050ededc07aa84d135946","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2655744dcf1c103a007050ededc07aa84d135946"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-13 14:24:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-13 14:25:51.000000000","tz":540},"subject":"Ensure filter factory option exists","message":"Ensure filter factory option exists\n\nAll middleware filters enabled in pipeline should have corresponding\nfactory option (use \u003d ... ) but the lines may not exist in case\nthe config file does not exist or is generated incompletely.\n\nChange-Id: I714fe28547ea8308d7e82b66687ea0f24f816cda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/bf4d4967a2dc19b72ada2e7bec4e1975f398d99e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/bf4d4967a2dc19b72ada2e7bec4e1975f398d99e"}]},"branch":"refs/heads/master"},"b9f0de92d2d2604628b02052b1bcf249fed3ccc0":{"kind":"REWORK","_number":2,"created":"2024-05-14 00:24:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/919527/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/27/919527/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/2"}}},"commit":{"parents":[{"commit":"2655744dcf1c103a007050ededc07aa84d135946","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2655744dcf1c103a007050ededc07aa84d135946"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-13 14:24:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-14 00:24:38.000000000","tz":540},"subject":"Ensure filter factory option exists","message":"Ensure filter factory option exists\n\nAll middleware filters enabled in pipeline should have corresponding\nfactory option (use \u003d ... ) but the lines may not exist in case\nthe config file does not exist or is generated incompletely.\n\nChange-Id: I714fe28547ea8308d7e82b66687ea0f24f816cda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b9f0de92d2d2604628b02052b1bcf249fed3ccc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b9f0de92d2d2604628b02052b1bcf249fed3ccc0"}]},"branch":"refs/heads/master"},"646793096824aba4efa40236dbf1e45203641533":{"kind":"REWORK","_number":3,"created":"2024-05-14 02:15:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/919527/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/27/919527/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/3"}}},"commit":{"parents":[{"commit":"2655744dcf1c103a007050ededc07aa84d135946","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2655744dcf1c103a007050ededc07aa84d135946"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-13 14:24:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-14 02:14:41.000000000","tz":540},"subject":"Ensure filter factory option exists","message":"Ensure filter factory option exists\n\nAll middleware filters enabled in pipeline should have corresponding\nfactory option (use \u003d ... ) but the lines may not exist in case\nthe config file does not exist or is generated incompletely.\n\nChange-Id: I714fe28547ea8308d7e82b66687ea0f24f816cda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/646793096824aba4efa40236dbf1e45203641533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/646793096824aba4efa40236dbf1e45203641533"}]},"branch":"refs/heads/master"},"79216540d5fe72c1aecf46e308f8b922afde9b6f":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-05-14 02:15:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/919527/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/27/919527/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/4"}}},"commit":{"parents":[{"commit":"2655744dcf1c103a007050ededc07aa84d135946","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2655744dcf1c103a007050ededc07aa84d135946"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-13 14:24:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-14 02:15:22.000000000","tz":540},"subject":"Ensure filter factory option exists","message":"Ensure filter factory option exists\n\nAll middleware filters enabled in pipeline should have corresponding\nfactory option (use \u003d ... ) but the lines may not exist in case\nthe config file does not exist or is generated incompletely.\n\nAlso fix the wrong factory definition for proxy-server, which should be\napp instead of filter.\n\nChange-Id: I714fe28547ea8308d7e82b66687ea0f24f816cda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/79216540d5fe72c1aecf46e308f8b922afde9b6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/79216540d5fe72c1aecf46e308f8b922afde9b6f"}]},"branch":"refs/heads/master"},"a123fd166bfdb161071d9869c69a6769ad8f5990":{"kind":"REWORK","_number":5,"created":"2024-05-14 03:43:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/919527/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/27/919527/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/27/919527/5"}}},"commit":{"parents":[{"commit":"2655744dcf1c103a007050ededc07aa84d135946","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2655744dcf1c103a007050ededc07aa84d135946"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-13 14:24:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-14 03:42:59.000000000","tz":540},"subject":"Ensure app/filter factory options exist","message":"Ensure app/filter factory options exist\n\nAll middleware filters and apps enabled in pipeline should have\nthe corresponding factory option (use \u003d ... ) but the lines may not\nexist in case the config file does not exist or is generated in advance\nincompletely.\n\nChange-Id: I714fe28547ea8308d7e82b66687ea0f24f816cda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a123fd166bfdb161071d9869c69a6769ad8f5990"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a123fd166bfdb161071d9869c69a6769ad8f5990"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
