)]}'
{"id":"openstack%2Fpuppet-swift~920707","triplet_id":"openstack%2Fpuppet-swift~master~Ic6c34c5381eb4f57642493b5acf127dba4f7834e","project":"openstack/puppet-swift","branch":"master","topic":"keymaster_middleware","attention_set":{},"removed_from_attention_set":{"8064":{"account":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"last_update":"2024-06-10 07:28:19.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-06-10 07:28:19.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2024-06-10 07:28:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic6c34c5381eb4f57642493b5acf127dba4f7834e","subject":"Add keymaster middleware","status":"MERGED","created":"2024-05-29 02:53:29.000000000","updated":"2024-06-10 07:29:13.000000000","submitted":"2024-06-10 07:28:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920707-keymaster_middleware","meta_rev_id":"83110d6dccdebb683878948b850e3b82ec9d43a3","_number":920707,"virtual_id_number":920707,"owner":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-10 07:28:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-06-09 17:33:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-06-10 06:21:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-06-10 06:21:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-29 04:04:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-30 14:32:06.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-05-31 01:17:57.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-06-10 06:21:18.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"13e6f1eef761140f2a7abf574e099d4823f56a18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2024-05-29 02:53:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b8352e10d41d96ed0c311759960db81f9ea94ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 04:04:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99341f1616834ac7bdf5eeffac12c475\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a23f4e88eda44e03aacffc6c21c1e915 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f98429e7e5d43fe8dba85ed06f47409 : SUCCESS in 3m 24s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/66cb98dd77a5427cbc51a2e53f491ebc : SUCCESS in 4m 52s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/f09bf2e46f6c44a18774b83fdcbdad79 : SUCCESS in 13m 11s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/162c5dfc4bff4bb59c5b300b5741a86b : SUCCESS in 7m 58s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee366dab068440e8a8b5db6062bdfadd : SUCCESS in 14m 16s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/008d49ef832e4c00a8d999aef64967db : SUCCESS in 10m 10s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79d5cfca433e486b8f35cdd144aec251 : FAILURE in 48m 41s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/39892cefd5fb487ab0164784306d5494 : SUCCESS in 59m 35s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb887fea050d46cfbcd7d576b5ccd772 : SUCCESS in 29m 33s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99e4da8d90cc4bc6b83bd7e23a659787 : SUCCESS in 15m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8531c469315941c18105cfd075a4eb40 : SUCCESS in 14m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed07f6ce8717416e8cc472b04b78249f : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"9f5931ea554e3e795ce2c5a3ee003e80df17f34d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-05-30 14:32:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d74b4d35edecfd0bda28e4ec6353c19d31dd085f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-31 01:17:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d5d8d16eb5affebe22ad727097a50e8ef046db7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2024-06-06 10:28:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f8a4050aa74f0a030b1a66d4397e77557540735b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 11:54:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c41e371097154eccaf255a82155eb2d2\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99b41823648648f4814549c1f91dd6b4 : FAILURE in 2m 22s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be94bee03f0d4c2a815808fb1c89e8d2 : SUCCESS in 2m 44s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1009848abab406d9fbc99c73196ae3d : SUCCESS in 5m 04s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/bea5d583fd59412c886240106f77eeec : SUCCESS in 12m 53s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1344093526e34020a7fdcf4e75bacd35 : SUCCESS in 8m 47s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/8cefe5697a054d7ba081b061ef94a66e : SUCCESS in 12m 31s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ae03700c56346099285e70507ad9fa2 : SUCCESS in 9m 01s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ecf5646789048dc98b5dd84b6654d32 : SUCCESS in 59m 50s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/512bb2b2a073405d88a68de6884c2e22 : SUCCESS in 1h 24m 16s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/2bce7d5cd0134f4ea541b1c3a9812518 : SUCCESS in 27m 23s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b58353c74155482593b742fca5150df5 : SUCCESS in 18m 20s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02c76f5f9c334bd2881a84a609f56b51 : SUCCESS in 18m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8653120ca6f9401daf884d860281ec03 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":2},{"id":"cb028a9032ba9fc2c81a028cfb451ce38f8da0f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2024-06-07 08:40:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4aa1679715e03d28b082c867f1897db3c049895a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 09:48:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bb43602f29e46f2b37076d2dfcb4aec\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea30680a5f49414098ae429776f913b6 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33eecc84a1a74123a9de6d36a555e95c : SUCCESS in 3m 00s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/f089f0bbb40f48889c98fc571168e574 : SUCCESS in 4m 24s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a654c6c232f4fc3aaae36bfdcb41497 : SUCCESS in 13m 31s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cc2fabd352444259a20a0f0faea2b1c : SUCCESS in 9m 33s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/18231a3a6c79456e92db94a02777b2db : SUCCESS in 10m 19s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09041f8df15e4354b5f7b6e4e39f0780 : SUCCESS in 9m 18s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d251b025b70045248466342012961eb8 : SUCCESS in 1h 00m 21s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c6effba0f2e4a808df68c476eebc685 : SUCCESS in 1h 01m 13s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c8627834a2e47e8acafc82371df4b6d : SUCCESS in 24m 02s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e082775e89244a41a25aaa2d2a0b4aa7 : SUCCESS in 15m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/525e2145618649af8abcf836926eb837 : SUCCESS in 14m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88cb2ad4bd8a4d7eb0b7079b286026df : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":3},{"id":"ee442dbd330e6158ca6221e5132499fd6a07d010","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-09 17:31:43.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"7a279455ebbf5ae1485005421e90464e7cd20c8b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-09 17:33:03.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d99ef7616927e871da888626c44e80d6b3b6daaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-09 18:56:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08197d0de33a4f31ae43f70887b05b53\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d36b2fda382747b1a94aa515897f5383 : SUCCESS in 2m 36s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c43f36ed57e4f7394abb6fd81161de6 : SUCCESS in 2m 31s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/22e8a17bca184cfdacc7efd66935fe01 : SUCCESS in 5m 42s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/739bcfc02e694b31b7f3d662018d2bc5 : SUCCESS in 13m 12s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5544482fbfc944be86cbfcf4414ab9ca : SUCCESS in 9m 42s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/9be4834b617342969a7927467de277b1 : SUCCESS in 12m 39s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f233c175b3c94708a17cf7537265b6ff : SUCCESS in 9m 41s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/055c33d5d7a445268549f96b28761e14 : SUCCESS in 57m 06s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/24ec8536c4054643a2fd6776e9a2bc9f : SUCCESS in 1h 07m 17s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc68aed7a7dd45838aba6d2afeee3002 : SUCCESS in 25m 00s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc3c8611470948a28065ad35b6243807 : SUCCESS in 16m 49s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61474e5710f344609da407fd6a47ccb1 : SUCCESS in 11m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcb0e0a3943846b98b5be728d0bd9c99 : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":4},{"id":"c63a6da068965c1e9967a2436d194c2e9c2a5e24","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-06-10 06:21:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"af74e45d4de26315b9f04fb825bd1477b6077d62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 06:22:09.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"746610b5827de102531fbcf76fccc78ec8baef1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 07:28:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63240adf327c4ca4b799a1618e2acfc0\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4e1e50285b1402e931ba866909b61d1 : SUCCESS in 2m 25s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/584775d5ee3d41a28af8353886e9096a : SUCCESS in 2m 46s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0f1cf81273f4624a5d0c0d98a860e0c : SUCCESS in 9m 35s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d9906d7de5b4d9ea4520708b35b200a : SUCCESS in 9m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a499f31de5c47fe924206c01648c347 : SUCCESS in 1h 05m 15s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/a33ad71d014645cb9879c1839afeaf30 : SUCCESS in 19m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/956204f3e38e4eb290e2f4c3b9e61cf6 : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":4},{"id":"4b3155f8342d42de45db089e5a0aedef75ff55ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 07:28:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"83110d6dccdebb683878948b850e3b82ec9d43a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 07:29:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26125f33d18a47d9928ed1cf69ad5386\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e319dfeef1e948f791ad4fd5a3ec87d9 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7f1a945550ccb1bc7f2e62071b22bae3937b794b","revisions":{"5d5f60823c5722065e99cd9629bf051beb1c84e7":{"kind":"REWORK","_number":1,"created":"2024-05-29 02:53:29.000000000","uploader":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"ref":"refs/changes/07/920707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/07/920707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/07/920707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/07/920707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/07/920707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/07/920707/1"}}},"commit":{"parents":[{"commit":"2a5ec303af64087c82cb78bb1110dce04e5a209d","subject":"Merge \"Ensure app/filter factory options exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2a5ec303af64087c82cb78bb1110dce04e5a209d"}]}],"author":{"name":"Jake Yip","email":"jake.yip@unimelb.edu.au","date":"2024-05-27 09:01:53.000000000","tz":600},"committer":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2024-05-29 02:54:04.000000000","tz":600},"subject":"Add keymaster middleware","message":"Add keymaster middleware\n\nAdd support for configuring swift-proxy keymaster middleware. The\nkeymaster middleware handles keys for swift encryption[1].\n\n[1] https://docs.openstack.org/swift/latest/overview_encryption.html\n\nChange-Id: Ic6c34c5381eb4f57642493b5acf127dba4f7834e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5d5f60823c5722065e99cd9629bf051beb1c84e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5d5f60823c5722065e99cd9629bf051beb1c84e7"}]},"branch":"refs/heads/master"},"8ef198de1843cf7e45074c090000dfac95f85bbd":{"kind":"REWORK","_number":2,"created":"2024-06-06 10:28:13.000000000","uploader":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"ref":"refs/changes/07/920707/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/07/920707/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/07/920707/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/07/920707/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/07/920707/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/07/920707/2"}}},"commit":{"parents":[{"commit":"140456e44a2d6fc921d0b16ce81eb04ca2eb7946","subject":"Bump upper version of puppet-memcached","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/140456e44a2d6fc921d0b16ce81eb04ca2eb7946"}]}],"author":{"name":"Jake Yip","email":"jake.yip@unimelb.edu.au","date":"2024-05-27 09:01:53.000000000","tz":600},"committer":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2024-06-06 10:25:56.000000000","tz":600},"subject":"Add keymaster middleware","message":"Add keymaster middleware\n\nAdd support for configuring swift-proxy keymaster middleware. The\nkeymaster middleware handles keys for swift encryption[1].\n\n[1] https://docs.openstack.org/swift/latest/overview_encryption.html\n\nChange-Id: Ic6c34c5381eb4f57642493b5acf127dba4f7834e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8ef198de1843cf7e45074c090000dfac95f85bbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8ef198de1843cf7e45074c090000dfac95f85bbd"}]},"branch":"refs/heads/master"},"1b641efc7c215f92f537767154c7e80049dd8d79":{"kind":"REWORK","_number":3,"created":"2024-06-07 08:40:07.000000000","uploader":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"ref":"refs/changes/07/920707/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/07/920707/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/07/920707/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/07/920707/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/07/920707/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/07/920707/3"}}},"commit":{"parents":[{"commit":"140456e44a2d6fc921d0b16ce81eb04ca2eb7946","subject":"Bump upper version of puppet-memcached","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/140456e44a2d6fc921d0b16ce81eb04ca2eb7946"}]}],"author":{"name":"Jake Yip","email":"jake.yip@unimelb.edu.au","date":"2024-05-27 09:01:53.000000000","tz":600},"committer":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2024-06-07 08:40:03.000000000","tz":600},"subject":"Add keymaster middleware","message":"Add keymaster middleware\n\nAdd support for configuring swift-proxy keymaster middleware. The\nkeymaster middleware handles keys for swift encryption[1].\n\n[1] https://docs.openstack.org/swift/latest/overview_encryption.html\n\nChange-Id: Ic6c34c5381eb4f57642493b5acf127dba4f7834e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1b641efc7c215f92f537767154c7e80049dd8d79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1b641efc7c215f92f537767154c7e80049dd8d79"}]},"branch":"refs/heads/master"},"7f1a945550ccb1bc7f2e62071b22bae3937b794b":{"kind":"REWORK","_number":4,"created":"2024-06-09 17:31:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/920707/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/07/920707/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/07/920707/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/07/920707/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/07/920707/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/07/920707/4"}}},"commit":{"parents":[{"commit":"140456e44a2d6fc921d0b16ce81eb04ca2eb7946","subject":"Bump upper version of puppet-memcached","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/140456e44a2d6fc921d0b16ce81eb04ca2eb7946"}]}],"author":{"name":"Jake Yip","email":"jake.yip@unimelb.edu.au","date":"2024-05-27 09:01:53.000000000","tz":600},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-09 17:31:21.000000000","tz":540},"subject":"Add keymaster middleware","message":"Add keymaster middleware\n\nAdd support for configuring swift-proxy keymaster middleware. The\nkeymaster middleware handles keys for swift encryption[1].\n\n[1] https://docs.openstack.org/swift/latest/overview_encryption.html\n\nChange-Id: Ic6c34c5381eb4f57642493b5acf127dba4f7834e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7f1a945550ccb1bc7f2e62071b22bae3937b794b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7f1a945550ccb1bc7f2e62071b22bae3937b794b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
