)]}'
{"id":"openstack%2Fpuppet-swift~921666","triplet_id":"openstack%2Fpuppet-swift~master~I8b11fbdb0ea0ca814b962dd72b5c1cb4cd201c24","project":"openstack/puppet-swift","branch":"master","topic":"config-perm","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-06-27 14:10:30.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2024-06-27 14:07:43.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I8b11fbdb0ea0ca814b962dd72b5c1cb4cd201c24","subject":"Set appropriate ownership/permission to config files","status":"MERGED","created":"2024-06-10 13:12:52.000000000","updated":"2024-06-27 16:00:21.000000000","submitted":"2024-06-27 16:00:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921666-config-perm","meta_rev_id":"e09fcb37e312d59ac30cc222d68669ea2d8ddb5f","_number":921666,"virtual_id_number":921666,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-27 16:00:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2024-06-27 14:07:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2024-06-26 07:55:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2024-06-27 14:07:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-10 14:51:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-06-10 16:06:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-11 15:06:50.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"},{"updated":"2024-06-26 07:55:10.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"6e64ec5c286ac1c996ab318b64071c0a856d7ddf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-10 13:12:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8fe67aac7b818f5b07d66101c82fcdc52ee3692f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-10 13:13:08.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ab5b679dabe3fc72f0130f9d7fcfcff0a59bc07f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-10 14:30:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"70b57a85c9e629fbc81241fb8865e933828a9ca7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-06-10 14:51:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3b77608c0b85d964431cc1a959ac9e5ce4b55dae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 16:06:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d54d4ee2e8254023b085811c5ff742e5\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b63bbfc14e154ca4842ded4174d505b9 : SUCCESS in 2m 59s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/537d28b588ea41eb8538684623a29993 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f6741e5e5c64828a97e1b8d13f3f18b : SUCCESS in 5m 13s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0037820900b46fda1a42efeed5365dd : FAILURE in 12m 04s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/107589b8322c4128b1c01bbf9fe94dff : FAILURE in 9m 14s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b6de7549bfc44c297a852a27b339c0a : FAILURE in 12m 47s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ed318c4f1594ab092118bb70524620e : FAILURE in 9m 59s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/895e8d03f4064237b36661f7f8c71293 : FAILURE in 52m 03s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a71bf4535bd407e8983d737cf5a7be0 : SUCCESS in 1h 21m 28s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/864425324ec2473ab696f8e9d5085997 : SUCCESS in 23m 38s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c989881d035f4900bb92f53b3741c5b5 : SUCCESS in 14m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83e5db7bc1b1429da4e9a25b7ebf78bb : SUCCESS in 16m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"de3e07af9d5b2c332d2476bf0401077de43a5c9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-11 01:27:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"42af2575a1b5198981a572e5f2be3b9f56dd3a9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 02:56:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21ff374c616240a2a066126e0ad1d12a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36b8f8899946424e896c0e6fa176e337 : SUCCESS in 3m 08s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a300afff2a51467e980cd496e6544ad3 : SUCCESS in 2m 35s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/32f5c1391b8a4ba6bcd09e2fffa25840 : SUCCESS in 4m 18s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6d1b518062f437a80173289026b6f41 : FAILURE in 10m 13s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf63dbe8972e4929945c273786185b8b : FAILURE in 9m 19s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/16895d2046724b01bd2716175340b19a : FAILURE in 9m 31s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcd499ce68f64475b5c4f62e1f64d93c : FAILURE in 9m 42s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c59e043ce684b3d9479e62e5d176113 : SUCCESS in 1h 05m 09s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/63922aabdaaa44bba3c84677658e00b3 : SUCCESS in 1h 24m 11s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f3a9c61bc024aa4804020a5f43977a9 : SUCCESS in 19m 33s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63038e183b3b4749b582132318eaa9b3 : SUCCESS in 13m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49da1c0ea7784f8db2f9b30a83b9e94d : SUCCESS in 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"68e5e2ed4daef7f5c0fc36f8313627656a86441e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-11 03:04:08.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2e95c3528f3adb9ed58561611cad703645b56fcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 04:14:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed78ded8575149a69b5a6fd503319b30\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d3059118d4741688aa5dc63d669fe98 : SUCCESS in 2m 50s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a7b38cb520d4b3db3fddc58c3c0ae98 : SUCCESS in 5m 20s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/56c9702b6c14450c85909349ef40746b : SUCCESS in 6m 10s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/47d8680a46734ce0bc045a8ee4386c20 : SUCCESS in 11m 13s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30233c7e5aaa4333b98defc3bf868064 : SUCCESS in 11m 34s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/db351a5fa1e942bd853c9c9197281fcb : SUCCESS in 12m 41s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0440f8ea19ba4d8cb2a9e08b848c6960 : SUCCESS in 10m 09s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3c6b06b34fa46a4a19322ce3bf18fc4 : SUCCESS in 58m 18s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/754fb90c9efc4da8b7bfb060873d922e : SUCCESS in 1h 08m 40s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/90e81de001c24b5195fcc07b7aa16cc5 : SUCCESS in 25m 29s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c69baca94f746e5b4119cef2ccf2160 : SUCCESS in 15m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/157258a86bbb449b852cb602530b961c : SUCCESS in 19m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"32cc0a9a8c611d09b8b025c552532899ae2024ad","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-06-11 15:06:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ac0354c8b2f4c46adcf3c7a2962ecf25927e9ba0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-06-26 07:55:10.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"23459e5370280f5158bd4b28abda87875d6e7cb7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-06-27 14:07:43.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"90e2510b7e5242175298863a8a5d82020eba5a4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 14:08:32.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"42ed805edd181c8738c1bbd4ae54b0555c1bbaed","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-27 14:10:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"644cb734c9d7095b37d598ff82f6451f9f037101","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 16:00:21.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e7f8f63c9b14fe7ab12c17bfe8d5ec6\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb1b4f0517cd432eb3d7cad06ddcbef6 : SUCCESS in 3m 45s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5079c95d8c3c494b93b5fbed66e0cb3e : SUCCESS in 3m 35s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/a65f3832234d41af80985ce81a6b92a2 : SUCCESS in 14m 40s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11c5166cae9f4f7ab4f565208dbfaab5 : SUCCESS in 11m 48s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/835d29b4491a4ab895d9f083b87522a7 : SUCCESS in 1h 26m 46s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/992d2813b1c9475c96585064c1a6f51b : SUCCESS in 29m 21s","accounts_in_message":[],"_revision_number":5},{"id":"e09fcb37e312d59ac30cc222d68669ea2d8ddb5f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 16:00:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b0e59971b871c5432c93bbcddfb11edf60c15812","revisions":{"6d8f9d1739defb02afccb3aa52e2478cf2293808":{"kind":"REWORK","_number":1,"created":"2024-06-10 13:12:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/921666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/66/921666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/1"}}},"commit":{"parents":[{"commit":"7f1a945550ccb1bc7f2e62071b22bae3937b794b","subject":"Add keymaster middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7f1a945550ccb1bc7f2e62071b22bae3937b794b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-10 13:05:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-10 13:12:33.000000000","tz":540},"subject":"Set appropriate ownership/permission to config files","message":"Set appropriate ownership/permission to config files\n\n... to avoid making any config files world readable.\n\nAlso swift processes do not need write access to these files, thus\nwrite access can be blocked.\n\nCloses-Bug: 2068898\nChange-Id: I8b11fbdb0ea0ca814b962dd72b5c1cb4cd201c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/6d8f9d1739defb02afccb3aa52e2478cf2293808"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/6d8f9d1739defb02afccb3aa52e2478cf2293808"}]},"branch":"refs/heads/master"},"ff3195e43e4d8c55d6771129fbe4fdc271ac9bc3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-06-10 13:13:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/921666/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/66/921666/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/2"}}},"commit":{"parents":[{"commit":"7f1a945550ccb1bc7f2e62071b22bae3937b794b","subject":"Add keymaster middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7f1a945550ccb1bc7f2e62071b22bae3937b794b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-10 13:05:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-10 13:12:55.000000000","tz":540},"subject":"Set appropriate ownership/permission to config files","message":"Set appropriate ownership/permission to config files\n\n... to avoid making any config files world readable.\n\nAlso swift processes do not need write access to these files, thus\nwrite access can be blocked.\n\nCloses-Bug: #2068898\nChange-Id: I8b11fbdb0ea0ca814b962dd72b5c1cb4cd201c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ff3195e43e4d8c55d6771129fbe4fdc271ac9bc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ff3195e43e4d8c55d6771129fbe4fdc271ac9bc3"}]},"branch":"refs/heads/master"},"989bc579b0d58c27459f065e0f85396f6e4879df":{"kind":"REWORK","_number":3,"created":"2024-06-10 14:30:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/921666/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/66/921666/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/3"}}},"commit":{"parents":[{"commit":"7f1a945550ccb1bc7f2e62071b22bae3937b794b","subject":"Add keymaster middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7f1a945550ccb1bc7f2e62071b22bae3937b794b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-10 13:05:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-10 14:30:02.000000000","tz":540},"subject":"Set appropriate ownership/permission to config files","message":"Set appropriate ownership/permission to config files\n\n... to avoid making any config files world readable.\n\nAlso swift processes do not need write access to these files, thus\nwrite access can be blocked.\n\nCloses-Bug: #2068898\nChange-Id: I8b11fbdb0ea0ca814b962dd72b5c1cb4cd201c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/989bc579b0d58c27459f065e0f85396f6e4879df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/989bc579b0d58c27459f065e0f85396f6e4879df"}]},"branch":"refs/heads/master"},"dc3b57f94c871dd7d5f4a3ef7e572fba09b4a255":{"kind":"REWORK","_number":4,"created":"2024-06-11 01:27:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/921666/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/66/921666/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/4"}}},"commit":{"parents":[{"commit":"7f1a945550ccb1bc7f2e62071b22bae3937b794b","subject":"Add keymaster middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7f1a945550ccb1bc7f2e62071b22bae3937b794b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-10 13:05:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-11 01:26:57.000000000","tz":540},"subject":"Set appropriate ownership/permission to config files","message":"Set appropriate ownership/permission to config files\n\n... to avoid making any config files world readable.\n\nAlso swift processes do not need write access to these files, thus\nwrite access can be blocked.\n\nCloses-Bug: #2068898\nChange-Id: I8b11fbdb0ea0ca814b962dd72b5c1cb4cd201c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/dc3b57f94c871dd7d5f4a3ef7e572fba09b4a255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/dc3b57f94c871dd7d5f4a3ef7e572fba09b4a255"}]},"branch":"refs/heads/master"},"b0e59971b871c5432c93bbcddfb11edf60c15812":{"kind":"REWORK","_number":5,"created":"2024-06-11 03:04:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/921666/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/66/921666/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/66/921666/5"}}},"commit":{"parents":[{"commit":"7f1a945550ccb1bc7f2e62071b22bae3937b794b","subject":"Add keymaster middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7f1a945550ccb1bc7f2e62071b22bae3937b794b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-10 13:05:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-11 03:03:47.000000000","tz":540},"subject":"Set appropriate ownership/permission to config files","message":"Set appropriate ownership/permission to config files\n\n... to avoid making any config files world readable.\n\nAlso swift processes do not need write access to these files, thus\nwrite access can be blocked.\n\nCloses-Bug: #2068898\nChange-Id: I8b11fbdb0ea0ca814b962dd72b5c1cb4cd201c24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b0e59971b871c5432c93bbcddfb11edf60c15812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b0e59971b871c5432c93bbcddfb11edf60c15812"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
