)]}'
{"id":"openstack%2Fpuppet-swift~924628","triplet_id":"openstack%2Fpuppet-swift~master~I9589b420b8d1c6fcbd29d83945dbdcc2da2e22ef","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-30 16:28:38.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2024-07-30 15:10:08.000000000","reason":"removed on reply"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-07-30 10:59:48.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"I9589b420b8d1c6fcbd29d83945dbdcc2da2e22ef","subject":"Allow customizing sorting_method","status":"MERGED","created":"2024-07-22 13:51:11.000000000","updated":"2024-07-30 16:29:59.000000000","submitted":"2024-07-30 16:28:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924628","meta_rev_id":"321b42fa7b3746fa6118b05aaa2907f8fc032198","_number":924628,"virtual_id_number":924628,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-30 16:28:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-07-30 10:59:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-07-30 15:10:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-07-30 15:10:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-22 17:01:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-23 14:42:33.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-07-29 07:40:32.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"66412447a1063c47337787bc3b237b4b5d749bd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-22 13:51:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37ee62195859cfc537853d0fc806bd26969d7a0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-22 14:51:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a16b88e0af61a989c93138dc0dde350bc99ae006","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-22 15:30:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9a77fbfb5da5078a96f31673e4f46794eeda1de3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 17:01:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd5bd3fdf10f4c01963fe5f289f5a1bd\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60d1f0ba9dd149b784de2d7649530dc8 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c56c313e521d4cf99511d383b3e5a03e : SUCCESS in 3m 14s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2ac07d5a9b24157bd67cdb56d336cf7 : SUCCESS in 5m 27s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/26fc8159435044569916a18af053865f : SUCCESS in 13m 27s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2384fb78de2148449a5f1643904b103a : SUCCESS in 10m 07s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/75a59fe883244ec9ac9d6be5759f944c : SUCCESS in 13m 21s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0df6b217d2d148eeb348325ccc5d30f0 : SUCCESS in 9m 01s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/898ff37f2f2948b0a953e56379677593 : SUCCESS in 1h 04m 29s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf3a675395cb455e92559c5c74338067 : SUCCESS in 1h 11m 07s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd28733ba32641909c5310bc133cb023 : SUCCESS in 23m 43s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06c62a83987244aba362d2a366b634da : SUCCESS in 14m 37s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66bc9742156a4869ad05f5fb1b41dccd : SUCCESS in 16m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9fc2b662f294167af85d1f4bba09f10 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":3},{"id":"744c25a1cc85606c4cf7287bbec9ae2acd1383cf","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-07-23 14:42:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"715151e8a639964de9b6052a75e63883f45f4631","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-07-29 07:40:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3a169c6cddebe572467f39ac3607db3da30ef9c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 08:52:27.000000000","message":"Patch Set 3: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/puppet-swift for 924628,3","accounts_in_message":[],"_revision_number":3},{"id":"863d3f850b12c1d47a12941b81b5c656cce80ac8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-29 14:09:41.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2b2ce8f400f3ef04f55fa1fe7731cb0647557d2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 15:38:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/83f53787bff8481dbd74e5fac1152c58\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38e938c86ad44d2f9f2ddc1806d83005 : FAILURE in 3m 01s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/781953f635c84c3489903b1aa47b16f1 : SUCCESS in 2m 32s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d4adfe6e06a4210bcd2d70297f8261b : SUCCESS in 5m 28s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/566fd83e89bf4aa3b052c01737c89f6a : FAILURE in 14m 28s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2183b4337cd24458b6e8e06952760e6a : FAILURE in 9m 37s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd44d67ddfb94902a8b78d214f599ac8 : FAILURE in 12m 00s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcc825f1e253403387cd09fe468e952d : FAILURE in 10m 43s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/680ba1bb953f4405a4d17ba6aaf57dfc : SUCCESS in 1h 04m 58s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3fc0051628f45a9ba8f4a679045afb0 : SUCCESS in 1h 21m 08s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/721627cf88e845f7a7c6c99d88218ecf : SUCCESS in 27m 25s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc0ea74b28604f85890171f2b883bced : SUCCESS in 15m 42s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22f0521179ff4a9d94e855fba49c12f9 : SUCCESS in 15m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c1b03b1c58d4b07b548543e89999788 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":4},{"id":"0e5dcd300f6d96e8430434df08416a0bf3cc321d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-29 16:47:33.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"be20096d5196f43d4b611714ca81ca9fce764d73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 18:17:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6966a8f6f4d4ca599581ea258db61f9\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2db9f666118848e0bf713f600b651492 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a780698b9de4a0090fc314aec27fbf9 : SUCCESS in 2m 37s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/effe06b5f29440818fd4133cb673a8c6 : SUCCESS in 6m 18s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a010b2982ab4cf9a8539fb174b4980b : SUCCESS in 10m 26s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/810af6aeaa4b47ffad8eab4b85355e66 : SUCCESS in 10m 30s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/baf863df8e7d4a5ead8b9148882461f9 : SUCCESS in 12m 12s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e629f34a7cf42a386800e5c970cc4f3 : SUCCESS in 9m 43s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/150e5a8e6b9749c29f3a6368a0ef1d42 : FAILURE in 32m 24s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e01fa5aacbee4f1b94849ff7bae1686a : SUCCESS in 1h 21m 26s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/588bd2e1b1694a31a40bc2f418080e4e : SUCCESS in 24m 30s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0665c1bdbb9646c68cabdd45aef20653 : SUCCESS in 16m 45s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26fe0be24ba843eea90b8fe7cae14f27 : SUCCESS in 17m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7833902ddc24e04a8c50710f8ae6401 : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":5},{"id":"b1c4be7314772a5c5467a281edc3b066ffc799fa","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-07-30 10:59:48.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e80e845d58c6d52c3dc8472c4a8cc86a5b2bebbb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-07-30 15:10:08.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"af0f35ca9a709e42fe0e0d29c3882a4e2c489878","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 15:11:01.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3b924aaef6bed2bb42b80848f761bc140797f852","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 16:28:37.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cdf704e068c4411890f61f79683d6e9\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3abd32d671004611a9de3f7d9756fb94 : SUCCESS in 2m 50s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd7588e0b75049139079edd2bc0f0b8f : SUCCESS in 2m 43s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/9dda1cdce3af4bb296152793804f8020 : SUCCESS in 12m 17s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/358f06ed8c0e4a3d879ab9aa304446ab : SUCCESS in 10m 08s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d752957e8fb84731bde53852ede5dfab : SUCCESS in 1h 16m 53s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/1eee739013ae468f93512ca3ffd6b450 : SUCCESS in 22m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26d1cccae79a49888ea9fb34621f3b3f : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":5},{"id":"33ee0793b67c129508b813177046e044935d93a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 16:28:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"321b42fa7b3746fa6118b05aaa2907f8fc032198","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 16:29:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97ff8279c132449c9dd873bc8419aed1\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07cd314d16024c1181aeed5c0f3f3fee : SUCCESS in 50s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2f38c2923bd460563ea5645ac51312359029f1d7","revisions":{"54f059bfe18b3a95ebf5dfedd10c488bfa3e6e71":{"kind":"REWORK","_number":1,"created":"2024-07-22 13:51:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/924628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/28/924628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/1"}}},"commit":{"parents":[{"commit":"e0cb20b5061515b63484d2aad2abbbebab921c4a","subject":"Merge \"Set appropriate ownership/permission to config files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e0cb20b5061515b63484d2aad2abbbebab921c4a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 13:39:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 13:50:06.000000000","tz":540},"subject":"Allow customizing sorting_method","message":"Allow customizing sorting_method\n\nAdd a new option to customize sorting_method. This allows usage of\nthe \u0027timing\u0027 sorting method.\n\nChange-Id: I9589b420b8d1c6fcbd29d83945dbdcc2da2e22ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/54f059bfe18b3a95ebf5dfedd10c488bfa3e6e71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/54f059bfe18b3a95ebf5dfedd10c488bfa3e6e71"}]},"branch":"refs/heads/master"},"04cd4f5bfb93ca8c0390f18ad208ada98c30dfe2":{"kind":"REWORK","_number":2,"created":"2024-07-22 14:51:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/924628/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/28/924628/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/2"}}},"commit":{"parents":[{"commit":"e0cb20b5061515b63484d2aad2abbbebab921c4a","subject":"Merge \"Set appropriate ownership/permission to config files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e0cb20b5061515b63484d2aad2abbbebab921c4a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 13:39:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 14:51:13.000000000","tz":540},"subject":"Allow customizing sorting_method","message":"Allow customizing sorting_method\n\nAdd a new option to customize sorting_method. This allows usage of\nthe \u0027timing\u0027 sorting method.\n\nChange-Id: I9589b420b8d1c6fcbd29d83945dbdcc2da2e22ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/04cd4f5bfb93ca8c0390f18ad208ada98c30dfe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/04cd4f5bfb93ca8c0390f18ad208ada98c30dfe2"}]},"branch":"refs/heads/master"},"1c92c5f4bdda6c5549c83e46b1db94bdbed95927":{"kind":"REWORK","_number":3,"created":"2024-07-22 15:30:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/924628/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/28/924628/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/3"}}},"commit":{"parents":[{"commit":"e0cb20b5061515b63484d2aad2abbbebab921c4a","subject":"Merge \"Set appropriate ownership/permission to config files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e0cb20b5061515b63484d2aad2abbbebab921c4a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 13:39:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 15:30:42.000000000","tz":540},"subject":"Allow customizing sorting_method","message":"Allow customizing sorting_method\n\nAdd a new option to customize sorting_method. This allows usage of\nthe \u0027timing\u0027 sorting method.\n\nChange-Id: I9589b420b8d1c6fcbd29d83945dbdcc2da2e22ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1c92c5f4bdda6c5549c83e46b1db94bdbed95927"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1c92c5f4bdda6c5549c83e46b1db94bdbed95927"}]},"branch":"refs/heads/master"},"694fc0a42bda7e669587069f6f7dbbb1347e6bbf":{"kind":"REWORK","_number":4,"created":"2024-07-29 14:09:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/924628/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/28/924628/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/4"}}},"commit":{"parents":[{"commit":"d9eda0394f27618ffb1391288e21e9d6bf1d44f4","subject":"Merge \"Support per-method backend ratelimit options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d9eda0394f27618ffb1391288e21e9d6bf1d44f4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 13:39:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-29 14:08:53.000000000","tz":540},"subject":"Allow customizing sorting_method","message":"Allow customizing sorting_method\n\nAdd a new option to customize sorting_method. This allows usage of\nthe \u0027timing\u0027 sorting method.\n\nChange-Id: I9589b420b8d1c6fcbd29d83945dbdcc2da2e22ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/694fc0a42bda7e669587069f6f7dbbb1347e6bbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/694fc0a42bda7e669587069f6f7dbbb1347e6bbf"}]},"branch":"refs/heads/master"},"2f38c2923bd460563ea5645ac51312359029f1d7":{"kind":"REWORK","_number":5,"created":"2024-07-29 16:47:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/924628/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/28/924628/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/28/924628/5"}}},"commit":{"parents":[{"commit":"d9eda0394f27618ffb1391288e21e9d6bf1d44f4","subject":"Merge \"Support per-method backend ratelimit options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d9eda0394f27618ffb1391288e21e9d6bf1d44f4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 13:39:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-29 16:47:20.000000000","tz":540},"subject":"Allow customizing sorting_method","message":"Allow customizing sorting_method\n\nAdd a new option to customize sorting_method. This allows usage of\nthe \u0027timing\u0027 sorting method.\n\nChange-Id: I9589b420b8d1c6fcbd29d83945dbdcc2da2e22ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2f38c2923bd460563ea5645ac51312359029f1d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2f38c2923bd460563ea5645ac51312359029f1d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
