)]}'
{"id":"openstack%2Fpuppet-swift~924630","triplet_id":"openstack%2Fpuppet-swift~master~I87b2b6d936aa3d12453fd76df6394ad934b073cb","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-30 16:47:05.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-07-30 17:54:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I87b2b6d936aa3d12453fd76df6394ad934b073cb","subject":"Require write_affinity when write_affinity_node_count is customized","status":"MERGED","created":"2024-07-22 13:54:26.000000000","updated":"2024-07-30 17:54:54.000000000","submitted":"2024-07-30 17:54:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924630","meta_rev_id":"9f5082b60d729ae1316fc3cdedfc51ca9a25c5c9","_number":924630,"virtual_id_number":924630,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-30 17:54:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-07-29 16:47:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2024-07-29 16:47:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-07-30 16:47:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-22 14:13:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-23 14:42:38.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-07-29 07:40:44.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"9739025f1ffc26cbce779a426e68806eeee50644","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-22 13:54:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51739aa22a05de42d3d7c78166ce31982578e176","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 14:13:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/403c8bab25cc42758be2e75caf32c1e7\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2e6ed1d9c014e02972e7e738e885016 : SUCCESS in 2m 50s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba976d1356684e1bb4177fb06e643802 : SUCCESS in 2m 48s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/79ebed58facb4bedae6470f73b997d9a : SUCCESS in 4m 54s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/06d7399cf58c4c91879f9e24a2546e39 : FAILURE in 12m 34s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a907ade3cd844af8b69932b03549119 : FAILURE in 9m 21s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/4565102b22224bd48c85527286d11499 : FAILURE in 12m 09s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62e67156ad9e4a4199c6f13327ea6c4c : FAILURE in 9m 42s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47f463739b9148deb8cf622a64200aab : FAILURE in 7m 05s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/301ce692eb5146d0a105d691ddbcc1b2 : FAILURE in 10m 17s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/580d41728b314e3fbae28b392e114ed6 : FAILURE in 9m 21s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c01fd6d217d43058e118a0ab577b3bb : FAILURE in 6m 50s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b4536ddf20c4e72b1ea690a8f957431 : FAILURE in 6m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c98d65a5792b7b7cfc2b0452b890f9e255ec05a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-22 14:51:35.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"da378d77129769c13f70356c9ff2b0a2c5958c04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 15:11:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77291ec13b7a4ac48ec28948d480b8ec\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6faca9f5aa346669585f0206d4e3f4f : SUCCESS in 2m 56s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/929e3a9697d14def89f837428329605d : SUCCESS in 2m 45s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b116f251a9354f1380cadca635778858 : SUCCESS in 5m 27s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/b698df1ba9944a67b989dd12822a75f6 : FAILURE in 12m 18s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db7c63f38b374ece8e8c3c71378a0f9e : FAILURE in 10m 57s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ed4d29dfff44811b4d3393198cbe081 : FAILURE in 11m 26s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a133898f5d57487ead3b954454a8f330 : FAILURE in 10m 42s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/364c7fb8c07840b6a76bbc305e451bb1 : FAILURE in 6m 51s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0465d2efd0b4b1f86b74f846cd4603e : FAILURE in 11m 01s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/4cc8e1205ce54ef89972c686c1b646b6 : FAILURE in 12m 51s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0faef0c1cd16467e83bab2491dfb7f5f : FAILURE in 6m 35s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fffa1c6d21ef43d992dfa788f7b7f0a7 : FAILURE in 7m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"91f1374bf454aeda62e77d9c6852222133aec1b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-22 15:30:53.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"77aa516d31563b9bb0898bcbcacd7ae7e219be5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 16:05:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a800635b117e419eafbf0701a5a8abe2\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7613c7c58a2b4549afda4b325c66d40c : SUCCESS in 3m 00s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f54ae70f74cd4d50a576cdf274aa462f : SUCCESS in 3m 04s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b8f0e8d3b0b4a04afb7a7601f249211 : SUCCESS in 5m 37s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce03e810fa0f4090ad9bd5a989dec5a0 : FAILURE in 14m 05s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9fc4e78f7bb46f4aa19b4c7922d9742 : FAILURE in 10m 24s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f28a3500f4d4416b01e2a2600c5fe3b : FAILURE in 13m 23s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3df521dbc57041f4b219f2e134740c62 : FAILURE in 9m 01s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90d351304c314e2099f5366f7fb6aa14 : FAILURE in 6m 05s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/82a3d59a45ee4ad4971f5d989b7a7a28 : FAILURE in 11m 07s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/25a3cfd45d5248a397e59774d5fb8103 : FAILURE in 8m 39s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffcf67cfe058406aa24774e831ec997b : FAILURE in 6m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77b91942dbce4b988b9cbd2460df43eb : FAILURE in 6m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5f7548fc45c034e50c642d9bef960bec2a5bd361","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-23 01:22:45.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"cb762dcb37868801db30387307e69e9fdd586dc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 02:48:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b41c8b8434f143ea8c3d0a0c38295ee1\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a620ac9f74444bc08e40024dfdd51aa0 : SUCCESS in 2m 46s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a40c120ab1d429e9d99b4a925606886 : SUCCESS in 2m 36s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/12a77f2715fe4a90b706d7b3d3920120 : SUCCESS in 5m 24s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/661808f2c87645f0b951d99ff05fdfa3 : SUCCESS in 12m 24s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0eccf847be524936bb67220120956603 : SUCCESS in 9m 15s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf708d074180475b88d9cb5338e7510f : SUCCESS in 13m 29s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d92f27c409e4c60a1014e237d5ce968 : SUCCESS in 8m 46s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33819be557fd4110872a55359b621e7f : SUCCESS in 57m 57s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ed4125e92694dc1a7926ec6690f765d : SUCCESS in 1h 20m 02s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d409d09418844688b5c4514c2ec5356 : SUCCESS in 19m 17s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db19cf1739734176bc629101cf2a86a3 : SUCCESS in 14m 42s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/441ede2f082244e6a0a7819439e12005 : SUCCESS in 12m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f866f972fea74b0f76e320342b9cb3c65e23affe","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-07-23 14:42:38.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2a3f2da1e2a981c9dce3253977938912440871bf","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-07-29 07:40:44.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a428306cde4fa519b153479d39b153498a470827","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 08:52:28.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/924628 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"1db2980d55f6f5c235f08f2d5df91057e1d8a3ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-29 14:09:41.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"8cac6e4467f3ef6427c6cc657735f9937bcf83cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 15:34:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d508b2a09bc543018c6771105d7242e8\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d307b2091a4e43a29edfd86142c27c3b : FAILURE in 2m 35s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b64f027ec7f946cc827f6348d37ea96d : SUCCESS in 2m 28s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc218d7285c5468bab3d3e5348dabed6 : SUCCESS in 6m 02s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbab92f9d4f540ee8e02d68a47fdfe66 : FAILURE in 12m 12s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f136e76e93d24865a54048cd00a08723 : FAILURE in 9m 26s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd68e3b6c7544677be1ede9692679274 : FAILURE in 14m 41s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82b7a308b4404be780d23c39e74d8aaa : FAILURE in 9m 04s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/162469f60eab44b2ad769c52aabfba9d : SUCCESS in 1h 02m 37s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e5f8b19d76a4bb6951268599513094a : SUCCESS in 1h 18m 49s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3c5f57ac41948628c15baa96084418b : SUCCESS in 24m 39s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d6719a3d73744e2a2f894ea914e170d : SUCCESS in 15m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48dbc0bd5c01459bb8985f45fa597515 : SUCCESS in 17m 13s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4e73441c86d4a5a573c958161b0a61a502af166d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-29 16:47:33.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"f9506239eb13232e74e381a7b04083c1093861a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 18:20:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/920788569c1d429982d1638dc56d042f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d71bc74f7b354be787b1cebcc801a325 : SUCCESS in 2m 56s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb4bb7538562403bb729124606cebaf2 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/4cd54b84477e4f15951f39d6a6491472 : SUCCESS in 5m 30s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/a795c02ae6d249b383d3b6afd7efc62f : SUCCESS in 12m 56s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a4cf99b10d1400ead7626275f5d7bb8 : SUCCESS in 9m 43s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a2e39e08bcd4140a80eef5bc4635dc2 : SUCCESS in 12m 36s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84e64cd75a3f47bc97e09b97428b7f44 : SUCCESS in 9m 55s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bcefca76abcd4f1a950c00efbbdc242e : FAILURE in 54m 48s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ff0b830a67b4224bea1bb784febbb78 : SUCCESS in 1h 16m 03s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/46bb655d7ab24736a40bbb5dceafb8e5 : SUCCESS in 24m 50s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3db661aeb301440abb885651d159e61c : SUCCESS in 15m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2701ba77197447bbb1d7d9f427588d02 : SUCCESS in 14m 32s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e72fd2f1beb2d7c4786a9810d82d0e3d052f893c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-30 16:47:05.000000000","message":"Patch Set 6: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c141929a81cdf527bba844b08d87801ae884b437","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 16:47:56.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3414671fcfa634fbb1ef879629cbef1628b8cbe2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 17:54:54.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba3adcf264b94f5faa6280e7d6ae4fff\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c9f1103e27f4948891fabed575aa661 : SUCCESS in 3m 33s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d78a5c54015141ec8e4ff0d858cd366b : SUCCESS in 2m 37s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b4b4d16a8234c25b0a32d895408cfa9 : SUCCESS in 12m 55s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0988ac77873456f896eb8b29aa17212 : SUCCESS in 10m 58s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d1ce0a42a3f4ae28269a46eab9e6ffe : SUCCESS in 1h 06m 54s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8dfba64a895415cbf31b74dbfa8b76e : SUCCESS in 18m 47s","accounts_in_message":[],"_revision_number":6},{"id":"9f5082b60d729ae1316fc3cdedfc51ca9a25c5c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 17:54:54.000000000","message":"Change has been successfully merged\n\n4 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b97128abdbaa4f016af7dd3736a940752fa1768a","revisions":{"f85f7743bf8014db845e59b00792bbd492e21fda":{"kind":"REWORK","_number":1,"created":"2024-07-22 13:54:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/924630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/30/924630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/1"}}},"commit":{"parents":[{"commit":"54f059bfe18b3a95ebf5dfedd10c488bfa3e6e71","subject":"Allow customizing sorting_method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/54f059bfe18b3a95ebf5dfedd10c488bfa3e6e71"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 13:52:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 13:52:58.000000000","tz":540},"subject":"Require write_affinity when write_affinity_node_count is customized","message":"Require write_affinity when write_affinity_node_count is customized\n\nThis imports the logic in swift::proxy to swift::internal_client\nto apply the same validation to equivalent parameters.\n\nChange-Id: I87b2b6d936aa3d12453fd76df6394ad934b073cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f85f7743bf8014db845e59b00792bbd492e21fda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f85f7743bf8014db845e59b00792bbd492e21fda"}]},"branch":"refs/heads/master"},"7e752ca02274f14380a25270652c772c7fd220c0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-22 14:51:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/924630/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/30/924630/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/2"}}},"commit":{"parents":[{"commit":"04cd4f5bfb93ca8c0390f18ad208ada98c30dfe2","subject":"Allow customizing sorting_method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/04cd4f5bfb93ca8c0390f18ad208ada98c30dfe2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 13:52:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 14:51:26.000000000","tz":540},"subject":"Require write_affinity when write_affinity_node_count is customized","message":"Require write_affinity when write_affinity_node_count is customized\n\nThis imports the logic in swift::proxy to swift::internal_client\nto apply the same validation to equivalent parameters.\n\nChange-Id: I87b2b6d936aa3d12453fd76df6394ad934b073cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7e752ca02274f14380a25270652c772c7fd220c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7e752ca02274f14380a25270652c772c7fd220c0"}]},"branch":"refs/heads/master"},"9e4535afbf172ccb571864f1af0d1316f6a4ecd2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-07-22 15:30:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/924630/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/30/924630/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/3"}}},"commit":{"parents":[{"commit":"1c92c5f4bdda6c5549c83e46b1db94bdbed95927","subject":"Allow customizing sorting_method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1c92c5f4bdda6c5549c83e46b1db94bdbed95927"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 13:52:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 15:30:45.000000000","tz":540},"subject":"Require write_affinity when write_affinity_node_count is customized","message":"Require write_affinity when write_affinity_node_count is customized\n\nThis imports the logic in swift::proxy to swift::internal_client\nto apply the same validation to equivalent parameters.\n\nChange-Id: I87b2b6d936aa3d12453fd76df6394ad934b073cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9e4535afbf172ccb571864f1af0d1316f6a4ecd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9e4535afbf172ccb571864f1af0d1316f6a4ecd2"}]},"branch":"refs/heads/master"},"d3b0cf34020560ad8ceee8e07d548f359a220d38":{"kind":"REWORK","_number":4,"created":"2024-07-23 01:22:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/924630/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/30/924630/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/4"}}},"commit":{"parents":[{"commit":"1c92c5f4bdda6c5549c83e46b1db94bdbed95927","subject":"Allow customizing sorting_method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1c92c5f4bdda6c5549c83e46b1db94bdbed95927"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 13:52:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 01:22:31.000000000","tz":540},"subject":"Require write_affinity when write_affinity_node_count is customized","message":"Require write_affinity when write_affinity_node_count is customized\n\nThis imports the logic in swift::proxy to swift::internal_client\nto apply the same validation to equivalent parameters.\n\nChange-Id: I87b2b6d936aa3d12453fd76df6394ad934b073cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d3b0cf34020560ad8ceee8e07d548f359a220d38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d3b0cf34020560ad8ceee8e07d548f359a220d38"}]},"branch":"refs/heads/master"},"60ec911417f93b1675741cbf87137d9e7a53b708":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-07-29 14:09:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/924630/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/30/924630/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/5"}}},"commit":{"parents":[{"commit":"694fc0a42bda7e669587069f6f7dbbb1347e6bbf","subject":"Allow customizing sorting_method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/694fc0a42bda7e669587069f6f7dbbb1347e6bbf"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 13:52:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-29 14:08:55.000000000","tz":540},"subject":"Require write_affinity when write_affinity_node_count is customized","message":"Require write_affinity when write_affinity_node_count is customized\n\nThis imports the logic in swift::proxy to swift::internal_client\nto apply the same validation to equivalent parameters.\n\nChange-Id: I87b2b6d936aa3d12453fd76df6394ad934b073cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/60ec911417f93b1675741cbf87137d9e7a53b708"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/60ec911417f93b1675741cbf87137d9e7a53b708"}]},"branch":"refs/heads/master"},"b97128abdbaa4f016af7dd3736a940752fa1768a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-07-29 16:47:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/924630/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/30/924630/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/30/924630/6"}}},"commit":{"parents":[{"commit":"2f38c2923bd460563ea5645ac51312359029f1d7","subject":"Allow customizing sorting_method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2f38c2923bd460563ea5645ac51312359029f1d7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 13:52:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-29 16:47:24.000000000","tz":540},"subject":"Require write_affinity when write_affinity_node_count is customized","message":"Require write_affinity when write_affinity_node_count is customized\n\nThis imports the logic in swift::proxy to swift::internal_client\nto apply the same validation to equivalent parameters.\n\nChange-Id: I87b2b6d936aa3d12453fd76df6394ad934b073cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b97128abdbaa4f016af7dd3736a940752fa1768a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b97128abdbaa4f016af7dd3736a940752fa1768a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
