)]}'
{"id":"openstack%2Fpuppet-swift~924689","triplet_id":"openstack%2Fpuppet-swift~master~I47d111127cf37d46988c858cb553f765a303af70","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-30 18:03:25.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2024-07-30 15:10:15.000000000","reason":"removed on reply"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-07-30 11:00:00.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"I47d111127cf37d46988c858cb553f765a303af70","subject":"Support more options to control internal request behavior","status":"MERGED","created":"2024-07-23 01:59:13.000000000","updated":"2024-07-30 18:04:21.000000000","submitted":"2024-07-30 18:03:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924689","meta_rev_id":"f4ff4260f8de63b86977b0141a29cabb0700862e","_number":924689,"virtual_id_number":924689,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-30 18:03:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-07-30 11:00:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-07-30 15:10:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-07-30 15:10:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-23 03:43:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-23 14:42:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-07-29 07:40:59.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"e8541176974a9ab12df061980b4ca9f8fecb4eb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-23 01:59:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ae6b8ac48fd006e43c40bacef7df5732813653a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-23 02:10:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"514d0ff6d53dcccba032bc056a1e86eb2683bcf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 03:43:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/116d901fb59445b691033873d8a862e4\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80c0f85e00b6456fb1134a31a580f617 : SUCCESS in 3m 01s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0bb2c72e36fc4f1392f326d5200e7335 : SUCCESS in 3m 45s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ac129440e6e4a1c91aaa36344bb9581 : SUCCESS in 5m 48s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c7c4b8df1244b04b6a35bd51858529c : FAILURE in 15m 19s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e73aa72a1bd44bdad0d915c4001a896 : FAILURE in 11m 14s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/72e892cfef874838a0796e4cd8817724 : FAILURE in 13m 21s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88c66e7650da44808d5a8eb29bad7029 : FAILURE in 10m 58s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/499f626960e346e2a2c4679733009dd9 : SUCCESS in 57m 03s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b944d49b2d74a4aab91f39ce9f37b71 : SUCCESS in 1h 20m 12s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ae607d1a1e746c4bb99ddb93c9e0eb5 : SUCCESS in 29m 51s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5f42be5f1c44ecb9ac70e4118bd55e4 : SUCCESS in 16m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/689555a09d7f4de690298eaea632a526 : SUCCESS in 15m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b88791ff70194afea8ed58fa30844e8c : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":2},{"id":"f34deb44c229b829878a1d38097b3ea87908d20d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-23 07:22:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0d90721ca6f281d234f62321775a507a34492740","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 08:46:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc9874634c0d497f95c0b61d42c9130b\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be3a033d2e334068a1cd5d360e66fdff : SUCCESS in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5138cbb5c99c4cc7a85e30ce65c9624f : SUCCESS in 2m 41s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a850dbe1e304e708a3f2efe614da631 : SUCCESS in 4m 34s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/1712ceb486104469a3baeb8a8b8b6216 : SUCCESS in 11m 34s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88c82ee2ed484ddaa58f07c53aef142f : SUCCESS in 8m 11s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/a43e3c8f78fb4d4dae1e7a10e3582ddf : SUCCESS in 11m 14s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36972d1f9302427990fc9eadee273acd : SUCCESS in 7m 52s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e44e967546584f73b71c97ee97086c2a : SUCCESS in 59m 56s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9147b42310f4947b585057fc7eb2f5c : SUCCESS in 1h 16m 43s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/8843c6cb2da244c6be87df8f34057e82 : SUCCESS in 23m 27s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cda0c8828a004224a6a9834d3987f917 : SUCCESS in 15m 21s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03e2eebb6e8b40de877897ef919754c7 : SUCCESS in 17m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/accc53196a0f4af082e15e1ccdec9f55 : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":3},{"id":"fae3b38756c3906efea43fa584acdf7bfa171275","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-07-23 14:42:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fbf6753670904b8f525ad439abdcd94553c32708","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-07-29 07:40:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"06db52b8a4ac65ee2d6a2a2cb121d057579a7d2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 08:52:28.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/924630 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"6efa6bd7adf2f43b0972b1b6e9d701f4fae05f59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-29 14:09:41.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"bab1d01bd13098078774efd26fdd1e7f0068f435","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 15:30:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a22369b712b4577aa5047a793211c60\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efad993eb87c4ac1815da2a86989bf2c : FAILURE in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19b670aea235472a8aca3313f1788b34 : SUCCESS in 2m 37s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/5996ee5d729a4cc083d3dabf8bf217da : SUCCESS in 5m 40s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/10b9ce7bfe734954a44b4b7b7ed23184 : FAILURE in 11m 52s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b8cd03366be4f4cbe3204154da30110 : FAILURE in 8m 21s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a6737296475417990a3d04b2645234d : FAILURE in 11m 50s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b795419e0b7144fa94f0bfffbd10be83 : FAILURE in 9m 40s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39e1a2e8d0a440b495ca5169aee55c30 : SUCCESS in 1h 06m 25s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/93224f90762c4857815446934bdc7ad6 : SUCCESS in 1h 14m 13s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/484660b1439d41fa9e18ad121e3eec6c : SUCCESS in 25m 26s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d02a1fff3f144979bfc70c5f5cb428a0 : SUCCESS in 13m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0410fa1bdfb34498bca8e1f59db31e36 : SUCCESS in 16m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e40b4e911be84eb6b46fba34ffad0381 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":4},{"id":"98fcfffdd3390a798ee55011de141d8404deef71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-29 16:47:33.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"9882209f5d780837d127d313870c4af3aa86f16f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 17:56:29.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/229959e0dbfa446e85e8c80497d8222e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5633f47a5b3d4fbe853a73815b5d8284 : SUCCESS in 2m 34s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d79c60ba5c04347befc9106f3f947c2 : SUCCESS in 2m 41s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d5be5163e134f2c941645593688210c : SUCCESS in 6m 28s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb7c6fbaafa3475abb27dfc7c6ce3e71 : FAILURE in 12m 13s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86c8f7f81fbe4733bd5bdc5760732883 : FAILURE in 8m 51s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6e2982311514a06862e9f1a13707c5b : FAILURE in 12m 43s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db362a47496f4b0bb6d2e2efef6aafd1 : FAILURE in 8m 57s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df44363d2a56460c8877544f65384cc3 : SUCCESS in 53m 27s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/49134cd914aa45e6a5747b3b2b958700 : SUCCESS in 1h 07m 34s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/dcb83b1c570c466eb81c1261afe295eb : SUCCESS in 26m 40s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3bd0ab9d037a4369b70ac0f8ba3cc126 : SUCCESS in 13m 44s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc6530b45a9f471a8a736a18d3aae5e1 : SUCCESS in 14m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0445a90a9e744be83fe6d6a348945e1 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":5},{"id":"f99b61034459cb79af0a2d11a3539f42791f3d37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-29 23:51:55.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"66e8ec6bbdc8bc2474e75c0f84c9daad1b57304a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 01:07:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db2ead61b5d54bc0930276319fee2e1a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b60eeea9a8114435b770e277ca3a1bc7 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/131aeb40c8994bedbab4824cbdcec5d6 : SUCCESS in 3m 01s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/153f2e2d4ed543edaaa54bb33040a564 : SUCCESS in 6m 41s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/da1956328b6349069595ceaaaeb7dc5a : SUCCESS in 16m 00s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2c364668da54c248b6a1e3d9679ee01 : SUCCESS in 9m 42s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/aedd125c41054f4881ebe80317d2c400 : SUCCESS in 15m 12s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86253cfafc32413a88a4e724c4ed5ba6 : SUCCESS in 9m 57s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f634b21b28a4ad8a0d72dd795f2ce8b : SUCCESS in 47m 58s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/341850aede3c4bddb8259a41082111ba : SUCCESS in 1h 09m 21s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/7458c26818b540f6ba557ba57db5510d : SUCCESS in 18m 30s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd5665b44abb4439b0a9a0e021642157 : SUCCESS in 12m 34s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a32bc51d28e4eba95fb46575caeb49d : SUCCESS in 16m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d21198b80e8741cc903bcb813eb99ab5 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":6},{"id":"eb790e7a1b6c7973ca13464f3c0a0b9d20ba7d8e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-07-30 11:00:00.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"409edaaa15c21fe3c6c494942c272ff3672a9086","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-07-30 15:10:15.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"c1eed01cb42894232cba16a9cc3c5859d3780875","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 16:48:01.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f5b7e13994cf8a10e5b86bd6c0f46a7b1f4c5159","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 18:03:25.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8e0ffe97f414bb1b1a54ce17ef5534e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad001e4e5f334c0ba3c89a7dc7694c4c : SUCCESS in 3m 15s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30c3764bbea44d9c9c3bf853d7db656f : SUCCESS in 2m 28s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac95788f9f254646b9206810645a453a : SUCCESS in 11m 42s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49d25f2612d149658a8dcfaee4215080 : SUCCESS in 9m 24s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8783c2a3edcb4117bccb7e78bd320ed5 : SUCCESS in 1h 13m 40s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f2f0d7ae4c9480bbc1c6d2b27dd4b37 : SUCCESS in 19m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e69307736f946609f15b7b7d79f744c : SUCCESS in 2m 18s","accounts_in_message":[],"_revision_number":6},{"id":"d192923ffd0d1e83a0291690f5ea0f55fda8a34a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 18:03:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"f4ff4260f8de63b86977b0141a29cabb0700862e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 18:04:21.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a6faa16a68441be83227b4c4e1003c1\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17b6486b8e274e83a7f1312da025308a : SUCCESS in 37s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ea34ed8ee99315c32389d3c1a974348636c918f5","revisions":{"d40fa679aea34a4aa1ab65c73c30b4d51777cba3":{"kind":"REWORK","_number":1,"created":"2024-07-23 01:59:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/924689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/89/924689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/1"}}},"commit":{"parents":[{"commit":"d3b0cf34020560ad8ceee8e07d548f359a220d38","subject":"Require write_affinity when write_affinity_node_count is customized","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d3b0cf34020560ad8ceee8e07d548f359a220d38"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 01:55:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 01:55:59.000000000","tz":540},"subject":"Support more options to control internal request behavior","message":"Support more options to control internal request behavior\n\nAdd support for the following options to control internal requests\nwithin the cluster.\n - timing_expiry\n - request_node_count\n - write_affinity_handoff_delete_count\n\nChange-Id: I47d111127cf37d46988c858cb553f765a303af70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d40fa679aea34a4aa1ab65c73c30b4d51777cba3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d40fa679aea34a4aa1ab65c73c30b4d51777cba3"}]},"branch":"refs/heads/master"},"e68c6e4a85bebe1ef831868e52ed0aa8b15d897a":{"kind":"REWORK","_number":2,"created":"2024-07-23 02:10:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/924689/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/89/924689/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/2"}}},"commit":{"parents":[{"commit":"d3b0cf34020560ad8ceee8e07d548f359a220d38","subject":"Require write_affinity when write_affinity_node_count is customized","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d3b0cf34020560ad8ceee8e07d548f359a220d38"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 01:55:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 02:08:53.000000000","tz":540},"subject":"Support more options to control internal request behavior","message":"Support more options to control internal request behavior\n\nAdd support for the following options to control internal requests\nwithin the cluster.\n - timing_expiry\n - request_node_count\n - write_affinity_handoff_delete_count\n\nChange-Id: I47d111127cf37d46988c858cb553f765a303af70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e68c6e4a85bebe1ef831868e52ed0aa8b15d897a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e68c6e4a85bebe1ef831868e52ed0aa8b15d897a"}]},"branch":"refs/heads/master"},"7b6f04b49fe21f69f25c19952ef1c14c5800c9e6":{"kind":"REWORK","_number":3,"created":"2024-07-23 07:22:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/924689/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/89/924689/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/3"}}},"commit":{"parents":[{"commit":"d3b0cf34020560ad8ceee8e07d548f359a220d38","subject":"Require write_affinity when write_affinity_node_count is customized","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d3b0cf34020560ad8ceee8e07d548f359a220d38"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 01:55:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 07:22:03.000000000","tz":540},"subject":"Support more options to control internal request behavior","message":"Support more options to control internal request behavior\n\nAdd support for the following options to control internal requests\nwithin the cluster.\n - timing_expiry\n - request_node_count\n - write_affinity_handoff_delete_count\n\nChange-Id: I47d111127cf37d46988c858cb553f765a303af70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7b6f04b49fe21f69f25c19952ef1c14c5800c9e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7b6f04b49fe21f69f25c19952ef1c14c5800c9e6"}]},"branch":"refs/heads/master"},"ec4b2ced722d730cd672b06b4b0d305d9b810ce1":{"kind":"REWORK","_number":4,"created":"2024-07-29 14:09:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/924689/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/89/924689/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/4"}}},"commit":{"parents":[{"commit":"60ec911417f93b1675741cbf87137d9e7a53b708","subject":"Require write_affinity when write_affinity_node_count is customized","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/60ec911417f93b1675741cbf87137d9e7a53b708"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 01:55:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-29 14:09:30.000000000","tz":540},"subject":"Support more options to control internal request behavior","message":"Support more options to control internal request behavior\n\nAdd support for the following options to control internal requests\nwithin the cluster.\n - timing_expiry\n - request_node_count\n - write_affinity_handoff_delete_count\n\nChange-Id: I47d111127cf37d46988c858cb553f765a303af70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ec4b2ced722d730cd672b06b4b0d305d9b810ce1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ec4b2ced722d730cd672b06b4b0d305d9b810ce1"}]},"branch":"refs/heads/master"},"639271ff5e38173c75b90a936fac88389dbccc9d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-07-29 16:47:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/924689/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/89/924689/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/5"}}},"commit":{"parents":[{"commit":"b97128abdbaa4f016af7dd3736a940752fa1768a","subject":"Require write_affinity when write_affinity_node_count is customized","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b97128abdbaa4f016af7dd3736a940752fa1768a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 01:55:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-29 16:47:24.000000000","tz":540},"subject":"Support more options to control internal request behavior","message":"Support more options to control internal request behavior\n\nAdd support for the following options to control internal requests\nwithin the cluster.\n - timing_expiry\n - request_node_count\n - write_affinity_handoff_delete_count\n\nChange-Id: I47d111127cf37d46988c858cb553f765a303af70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/639271ff5e38173c75b90a936fac88389dbccc9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/639271ff5e38173c75b90a936fac88389dbccc9d"}]},"branch":"refs/heads/master"},"ea34ed8ee99315c32389d3c1a974348636c918f5":{"kind":"REWORK","_number":6,"created":"2024-07-29 23:51:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/924689/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/89/924689/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/89/924689/6"}}},"commit":{"parents":[{"commit":"b97128abdbaa4f016af7dd3736a940752fa1768a","subject":"Require write_affinity when write_affinity_node_count is customized","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b97128abdbaa4f016af7dd3736a940752fa1768a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 01:55:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-29 23:51:41.000000000","tz":540},"subject":"Support more options to control internal request behavior","message":"Support more options to control internal request behavior\n\nAdd support for the following options to control internal requests\nwithin the cluster.\n - timing_expiry\n - request_node_count\n - write_affinity_handoff_delete_count\n\nChange-Id: I47d111127cf37d46988c858cb553f765a303af70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ea34ed8ee99315c32389d3c1a974348636c918f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ea34ed8ee99315c32389d3c1a974348636c918f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
