)]}'
{"id":"openstack%2Fpuppet-swift~939270","triplet_id":"openstack%2Fpuppet-swift~master~I4567f6339f277c0cc2d0c3ee91671067af65a156","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-01-16 15:52:08.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-01-16 14:22:58.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"I4567f6339f277c0cc2d0c3ee91671067af65a156","subject":"Remove swauth support","status":"MERGED","created":"2025-01-14 17:51:44.000000000","updated":"2025-01-16 15:53:19.000000000","submitted":"2025-01-16 15:52:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939270","meta_rev_id":"14dd8d8c11d2eea1bab66ab4a8a80a631dd45c3f","_number":939270,"virtual_id_number":939270,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-16 15:52:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2025-01-15 17:04:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2025-01-16 14:22:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2025-01-16 14:22:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-14 19:17:21.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-01-14 19:36:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-15 17:04:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"f6d45211e9ee4e2ca2c3dfb7c5f1a50abdfaf37f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-14 17:51:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"012dd157a0721d511165e33b8904911f6e66e2b5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-01-14 19:17:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1af80c3a02f051a2917f4fbce8bbbe41fb2f1f99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 19:36:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff42e8a0ae934a23838bb41ce5d4bc2c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2aa0829a717345eabc199da0ac7e1931 : SUCCESS in 2m 26s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db8b55a498e44c62873b88fa3fff1e44 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d521b7d682334239aebe99922f80c9fd : SUCCESS in 4m 54s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/1eb7ab3d97d04311ad1fe9be1aaa372e : FAILURE in 12m 14s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f8f23ad9dc340d18659257d72b654b4 : FAILURE in 5m 57s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9b1eb25527f4fd89fa828628978f316 : FAILURE in 11m 39s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8dad5bb1a5d043ebae1e40f8806826cc : FAILURE in 8m 51s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48fd3bd0e1864d58904a002633002a65 : SUCCESS in 1h 01m 12s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2cb42dceb3c4ea388a20d74565298e5 : SUCCESS in 1h 17m 20s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/6def416574c243109a56850dcbd797ea : SUCCESS in 23m 38s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3180d634c4044dfa89696346b2737f2 : SUCCESS in 15m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/502944b1276f41afb9c608e56679d842 : SUCCESS in 13m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7ab87f84b9d4476954b07bf4befdfcd : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"ce5564b7087941d6dfd365981373c460ea8b7663","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-15 02:13:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2923dc99612497ee930177dd2c01035a66b59f27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-15 03:12:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4a5714155149ee62361a8be56035d28db17486f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-15 03:31:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8eccce8f83e5aa6845456b2b8418b32222cf2afb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 04:50:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d95851f3ec1b4f08bb5fe5d6c7ccf378\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f90e31f75b7e49e6af071aa426017912 : SUCCESS in 2m 32s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fc54a2eb7fe4c5b87f520a21970f9a2 : SUCCESS in 2m 35s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3ca69520f794a519fbae3fd941dd3d6 : SUCCESS in 5m 26s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/af550a3de7d1424fbfe9a1d9410f0873 : FAILURE in 11m 58s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb85e51bf93746f2b5310a0cdb645770 : FAILURE in 8m 46s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/93cb3e6b13584d22b4cd6d8d4db8c4a8 : FAILURE in 10m 00s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7fea378b77f49f59abdd3f68bb56a48 : FAILURE in 8m 23s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6210d7b7fa04637a9fced94a863302e : SUCCESS in 1h 12m 23s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3fc906921654ce1b52d7645967aa8dd : SUCCESS in 45m 01s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/86ed8736015d42bb88cb73da754af0f2 : SUCCESS in 21m 28s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16749c8d3d0942b58b23866fbedbe298 : SUCCESS in 11m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44628161e5e141afbe61c904e5d75949 : SUCCESS in 14m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2dc8946af12f43c797d5ff32531a97c5 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":4},{"id":"f5f46e57bcc61e36c4f04ef87ea02f5a895717a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-15 07:21:59.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2af96dba071fc5928c81f74729ed05ff532b9778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 08:50:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc7cc99f775342a7925ebd32c5962199\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91e96bc532044d5fb16a7ade19ec2dba : SUCCESS in 3m 11s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec93b4df36b94afd867c28d8f2c966f8 : SUCCESS in 3m 06s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d621fdf9330485ab465f59a7926aa6d : SUCCESS in 5m 34s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/3caa6444fa7148839f470a27259c6e81 : FAILURE in 10m 29s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/432aa62b804949fbadaab475037fe7d3 : FAILURE in 7m 42s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3bf393b57ade4608bd69037763694528 : FAILURE in 6m 21s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/962971c91a364efd9ac720c06cfc2e3c : FAILURE in 8m 36s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f07b22ccd6024d12921baadc99a4e63c : FAILURE in 1h 16m 14s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/15a83ada96ea4af2970427c9d07fe3da : SUCCESS in 1h 12m 55s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/914e65660a354a87b78b518bc2fe2252 : SUCCESS in 27m 35s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c588f0897f24eebbece3c21652dd7d1 : SUCCESS in 14m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de3152f121534e30a1283b24a9284dea : SUCCESS in 15m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73c2b7d885c845b5883e42563685c04e : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":5},{"id":"d93a785f5bed3f4bd701048b466ea8cc87b63219","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-15 10:17:46.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"7257c4b6bd86d33c1b6a45f269cc15aaa16ad1ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 11:03:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b12231a5af9043989feaf76a4e5a1631\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab1072f811884f3bad2f12d067bb6dcd : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37d57e60dfe24696b9d7c715f2577984 : SUCCESS in 2m 26s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c4fe980cbf94c819035691faf8cee13 : SUCCESS in 5m 52s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbee3d5b4cc949c795a9ed3771f2223e : SUCCESS in 11m 28s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3175231801a44385a78508f5bfe3318f : SUCCESS in 4m 41s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/8db259eba7c5438a9ff6095b4a6c63c8 : SUCCESS in 6m 10s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5bb8d47f5994b4593e5841041072bca : SUCCESS in 5m 23s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0af9912747c541f9b73dd01c782b7d3c : SUCCESS in 30m 04s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b5d6005107b452e834b1cd569665a62 : SUCCESS in 43m 42s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/3bf4f644b1924b8db67fa0742dd9d811 : SUCCESS in 12m 56s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9004e9739f84a27b625d6c7d80c35a2 : SUCCESS in 13m 06s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8db72c7603f74cf78c81a24ba302f0a8 : SUCCESS in 13m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb0d5e44cc13469193f9caa73a09a6e1 : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":6},{"id":"8296f17e56bcecdf22320f7c8309d30907c4f789","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-01-15 17:04:01.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"1a9e35be049286ab9c030a4f852548dbe603f4fd","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-01-16 14:22:58.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"810a895e3f5f977c6496927f357293c387cbce05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 14:24:00.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"58e1277e6055a4eb37a22c88cdff7bf4707163c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 15:52:08.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c21aa17d38324094b4d10c2d99923641\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a061d1f7129a4e16a3f9f3fe2e9515fe : SUCCESS in 3m 18s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30be517a05ff4266a4974337ab87edf5 : SUCCESS in 2m 45s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/c635d2adf05d4c94b767d10302ff025a : SUCCESS in 11m 43s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd6be67d12f44cbb990eddff9ab7f1a3 : SUCCESS in 9m 02s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b27d581db4546989d3f0d46cb57a09d : SUCCESS in 1h 22m 00s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/d39da46bb4fe414882baec30aec08758 : SUCCESS in 24m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc7ed4c58fb647c9a27e6fa5ec8822ab : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":6},{"id":"256cd08206a98c05957df3b0eb9f9774a8ac012a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 15:52:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"14dd8d8c11d2eea1bab66ab4a8a80a631dd45c3f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 15:53:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/941ae2bd054842f1b4c99a204044b5e2\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2af51bbc877d44c8891e1ec02ae3f0b8 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"cd0dd760149a708720e3dc0a9a8403a5b6ad0ab8","revisions":{"94c383e8567ea86e1a8ff65c6d8ec1dbc25b1ffb":{"kind":"REWORK","_number":1,"created":"2025-01-14 17:51:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/939270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/70/939270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/1"}}},"commit":{"parents":[{"commit":"275d6a2ffe2092a14e1402047835db60e978b674","subject":"Merge \"Use specific log name for independent daemons\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/275d6a2ffe2092a14e1402047835db60e978b674"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-14 17:51:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-14 17:51:06.000000000","tz":540},"subject":"Remove swauth support","message":"Remove swauth support\n\nIt was deprecated during zed cycle[1], because swauth was already\nretired at that time.\n\n[1] 0eda2a0a50e57d3183ef4d70ee6ce9d257d59bb3\n\nChange-Id: I4567f6339f277c0cc2d0c3ee91671067af65a156\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/94c383e8567ea86e1a8ff65c6d8ec1dbc25b1ffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/94c383e8567ea86e1a8ff65c6d8ec1dbc25b1ffb"}]},"branch":"refs/heads/master"},"a2c295140caf3aec59c63ee9b0050f71e370daa5":{"kind":"REWORK","_number":2,"created":"2025-01-15 02:13:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/939270/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/70/939270/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/2"}}},"commit":{"parents":[{"commit":"275d6a2ffe2092a14e1402047835db60e978b674","subject":"Merge \"Use specific log name for independent daemons\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/275d6a2ffe2092a14e1402047835db60e978b674"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-14 17:51:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-15 02:13:29.000000000","tz":540},"subject":"Remove swauth support","message":"Remove swauth support\n\nIt was deprecated during zed cycle[1], because swauth was already\nretired at that time.\n\n[1] 0eda2a0a50e57d3183ef4d70ee6ce9d257d59bb3\n\nChange-Id: I4567f6339f277c0cc2d0c3ee91671067af65a156\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a2c295140caf3aec59c63ee9b0050f71e370daa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a2c295140caf3aec59c63ee9b0050f71e370daa5"}]},"branch":"refs/heads/master"},"808b5da2671b778f58a0fe796e979d06830d75fe":{"kind":"REWORK","_number":3,"created":"2025-01-15 03:12:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/939270/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/70/939270/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/3"}}},"commit":{"parents":[{"commit":"275d6a2ffe2092a14e1402047835db60e978b674","subject":"Merge \"Use specific log name for independent daemons\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/275d6a2ffe2092a14e1402047835db60e978b674"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-14 17:51:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-15 03:11:50.000000000","tz":540},"subject":"Remove swauth support","message":"Remove swauth support\n\nIt was deprecated during zed cycle[1], because swauth was already\nretired at that time.\n\n[1] 0eda2a0a50e57d3183ef4d70ee6ce9d257d59bb3\n\nChange-Id: I4567f6339f277c0cc2d0c3ee91671067af65a156\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/808b5da2671b778f58a0fe796e979d06830d75fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/808b5da2671b778f58a0fe796e979d06830d75fe"}]},"branch":"refs/heads/master"},"cae7cb0f0141a091273125a38f24df2558de5329":{"kind":"REWORK","_number":4,"created":"2025-01-15 03:31:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/939270/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/70/939270/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/4"}}},"commit":{"parents":[{"commit":"feec56d0c8e9f9a4b2ad52fe649ade5e814bc13c","subject":"Fix undefined variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/feec56d0c8e9f9a4b2ad52fe649ade5e814bc13c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-14 17:51:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-15 03:31:12.000000000","tz":540},"subject":"Remove swauth support","message":"Remove swauth support\n\nIt was deprecated during zed cycle[1], because swauth was already\nretired at that time.\n\n[1] 0eda2a0a50e57d3183ef4d70ee6ce9d257d59bb3\n\nChange-Id: I4567f6339f277c0cc2d0c3ee91671067af65a156\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cae7cb0f0141a091273125a38f24df2558de5329"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cae7cb0f0141a091273125a38f24df2558de5329"}]},"branch":"refs/heads/master"},"db9c24a2849fbb839d3d2bdcd0765b7633d0b34b":{"kind":"REWORK","_number":5,"created":"2025-01-15 07:21:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/939270/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/70/939270/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/5"}}},"commit":{"parents":[{"commit":"feec56d0c8e9f9a4b2ad52fe649ade5e814bc13c","subject":"Fix undefined variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/feec56d0c8e9f9a4b2ad52fe649ade5e814bc13c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-14 17:51:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-15 07:21:31.000000000","tz":540},"subject":"Remove swauth support","message":"Remove swauth support\n\nIt was deprecated during zed cycle[1], because swauth was already\nretired at that time.\n\n[1] 0eda2a0a50e57d3183ef4d70ee6ce9d257d59bb3\n\nChange-Id: I4567f6339f277c0cc2d0c3ee91671067af65a156\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/db9c24a2849fbb839d3d2bdcd0765b7633d0b34b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/db9c24a2849fbb839d3d2bdcd0765b7633d0b34b"}]},"branch":"refs/heads/master"},"cd0dd760149a708720e3dc0a9a8403a5b6ad0ab8":{"kind":"REWORK","_number":6,"created":"2025-01-15 10:17:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/939270/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/70/939270/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/70/939270/6"}}},"commit":{"parents":[{"commit":"feec56d0c8e9f9a4b2ad52fe649ade5e814bc13c","subject":"Fix undefined variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/feec56d0c8e9f9a4b2ad52fe649ade5e814bc13c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-14 17:51:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-15 10:10:52.000000000","tz":540},"subject":"Remove swauth support","message":"Remove swauth support\n\nIt was deprecated during zed cycle[1], because swauth was already\nretired at that time.\n\n[1] 0eda2a0a50e57d3183ef4d70ee6ce9d257d59bb3\n\nChange-Id: I4567f6339f277c0cc2d0c3ee91671067af65a156\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cd0dd760149a708720e3dc0a9a8403a5b6ad0ab8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cd0dd760149a708720e3dc0a9a8403a5b6ad0ab8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
